)]}'
{"id":"openstack%2Fkolla~711917","triplet_id":"openstack%2Fkolla~master~I46c59250b1781aadcb486567a8fd0750a05e2f7e","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I46c59250b1781aadcb486567a8fd0750a05e2f7e","subject":"fix: Remove unused variables","status":"MERGED","created":"2020-03-09 13:02:54.000000000","updated":"2020-03-09 21:59:40.000000000","submitted":"2020-03-09 21:58:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711917-1583791083029-3b0434b8","meta_rev_id":"984ef12db2d0ce7e38cfd5bd3ac1f03a73adb735","_number":711917,"virtual_id_number":711917,"owner":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 21:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-09 13:32:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-03-09 13:56:44.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-09 20:32:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-03-09 20:32:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 13:32:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-09 13:56:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-09 20:32:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-09 21:58:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a3d7d5107eb117e467cb4a785cf52aa4881395c","author":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"date":"2020-03-09 13:02:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f637c2cda3ebabf06a5502628fe29c163ec1e4ca","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-09 13:32:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/635d19ae3cbe4eebb467e8698a2d12a5 : SUCCESS in 27m 45s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/17290e7494ce4cb580e7835cdfda606c : FAILURE in 13m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdddf5caa4bd8bdaf13f39f513b718206cb16ce0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:34:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a3ff28d8c06446afa972d19f2e9efc06 : FAILURE in 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c0df0d911cae6fc84b1c3039fc355cf51c36a21","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-09 13:56:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dd41da62a1e6e41ef35ac080c86c8f23286a0ddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 14:28:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b2e3bb1efd3b4ef4a4127a1b898d7595 : SUCCESS in 35m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/862d9bd2359f4ebaa35205ac8c38527a : SUCCESS in 40m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b6abda3f79954a9d9fb09d492a59979d : SUCCESS in 1h 24m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ec50f671db1641c8b4f4432615736577 : SUCCESS in 1h 16m 06s (non-voting)\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/41d490802c2b48928c2a947224bcd2a4 : FAILURE in 24m 14s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c36ea5162ad94112bc57cd88263b165d : SUCCESS in 21m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f9f5d6915ada4ce1af91d9559704a2db : SUCCESS in 22m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ab1e0dd6f26c449fa60b388fe56c7ae1 : SUCCESS in 1h 05m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7a6d7e377cc4424bf60359bb1aaf566 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b57d5d3fc7546eca4fe96edd5d4d5fa : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5a1390180f44c73be0454835dc0170a : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a7e5783382047bf9ef402316a842290 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23c20065fad64aabba9ef2fa3e150798 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e841e2c38c423da7f4bcd5b563decf : SUCCESS in 6m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0739ebd5b49f4754909ddf4963d454a0 : SUCCESS in 28m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ead71e8c9314feea3af0f0dfb9f63c4 : SUCCESS in 26m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6a1ec3c5dfd445b486c60343365a14d9 : SUCCESS in 1h 03m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cef0eb24167b483c8ec8c38a3f925fb5 : SUCCESS in 1h 04m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/57bd4e54c6b648a59891fb166f3fb5eb : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":1},{"id":"05ce5a6d0a79f5b9e99acb80ad8facabb5631678","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-09 20:32:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f0e550600b617fe6f9dcf2ae6a19d0c61da9dd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 20:32:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c563a3c99622a6046a6c190424f84fa42bdf71bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 21:58:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fa968a282eeb41958c458ef6d7a30617 : SUCCESS in 42m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f0343fb0a4e54bbbab7130edb06e43e0 : SUCCESS in 1h 12m 34s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/761e81c49aa8427ab0effed712715add : SUCCESS in 1h 20m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e0b90986e7646d0b0ee41ca0cc215de : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e121db86bf324867ac1590e5ff29c888 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9eee335388174ba69ada059af1e9796c : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/359060860c3d4f6fa00e86378c8777ef : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3132fd80ece54ebc9531e4943ccc5a75 : SUCCESS in 4m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4099dcf9cf8b4d4389181e549e6f39e1 : SUCCESS in 28m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/83cfc07117b04e04889d8165b14fcb0f : SUCCESS in 55m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/66ffa4d10cb04d6eb0f99987109820c1 : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1bf7e679a33244a0a89aac6615df04ee7694389a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 21:58:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"984ef12db2d0ce7e38cfd5bd3ac1f03a73adb735","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 21:59:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aa6559f44564245888f50aea8781fe4 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a82db69dd3b9b7b83f766244dfec5e18ef09603f","revisions":{"a82db69dd3b9b7b83f766244dfec5e18ef09603f":{"kind":"REWORK","_number":1,"created":"2020-03-09 13:02:54.000000000","uploader":{"_account_id":27809,"name":"hanxiang gao","email":"gaohanxiang@126.com","username":"gao.hanxiang"},"ref":"refs/changes/17/711917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/711917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/711917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/711917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/711917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/711917/1"}}},"commit":{"parents":[{"commit":"90e38f23d4e25c30c9807ab511ca4e101adcf688","subject":"Merge \"Deprecate Sensu images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90e38f23d4e25c30c9807ab511ca4e101adcf688"}]}],"author":{"name":"Hanxiang Gao","email":"gao.hanxiang@99cloud.net","date":"2020-03-09 12:51:25.000000000","tz":480},"committer":{"name":"Hanxiang Gao","email":"gao.hanxiang@99cloud.net","date":"2020-03-09 13:02:27.000000000","tz":480},"subject":"fix: Remove unused variables","message":"fix: Remove unused variables\n\n\u0027self.nocache\u0027 variable is not used, has replaced it with\n\u0027not self.conf.cache\u0027.\n\nChange-Id: I46c59250b1781aadcb486567a8fd0750a05e2f7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a82db69dd3b9b7b83f766244dfec5e18ef09603f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a82db69dd3b9b7b83f766244dfec5e18ef09603f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
