)]}'
{"id":"openstack%2Fkolla~712100","triplet_id":"openstack%2Fkolla~stable%2Frocky~Icca01cf4083a095916b56ff70d3575591ae3abcd","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"Icca01cf4083a095916b56ff70d3575591ae3abcd","subject":"Bump versions for rocky","status":"MERGED","created":"2020-03-10 14:43:09.000000000","updated":"2020-03-10 18:52:48.000000000","submitted":"2020-03-10 18:52:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712100-1583866368267-97979107","meta_rev_id":"33acb8c5559f8011e1f7b54ba2c19925b9c08a34","_number":712100,"virtual_id_number":712100,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 18:52:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-03-10 15:09:59.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-10 17:44:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-03-10 17:44:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-10 15:09:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-10 17:44:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-10 18:52:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48f3f3791bd05690500a1fd8d427b91ea1302a87","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-10 14:43:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb16d8ece39d351eb664c2e677bf9ba52b7ff178","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-10 15:09:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6049b9e0b6ce8fef6c8b37c405cfeb21dd623494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 16:38:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a8536e17154c4fc6ad5fa9a40c923dc9 : SUCCESS in 43m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/132d6c90ffd743a38ddcb35c2139bc21 : SUCCESS in 50m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/571a4e11514a482c94d0eed58acf8b1e : SUCCESS in 1h 36m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9183fef472a247d392a7c5f130ed2afd : SUCCESS in 1h 28m 14s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4cb5295108834bd0b02f2ecdcf27149b : SUCCESS in 47m 04s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d5c2a5dc4d634e64a2f217fc4e98c342 : SUCCESS in 1h 02m 14s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5982639b35b7463e9e134069b112837e : SUCCESS in 1h 19m 45s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/899bbdc9ed4649d9b98ff496c116423e : SUCCESS in 1h 04m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40994119f170460780a558c22d432a67 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900d959f92c147eba5c5d8b6d69c61ec : SUCCESS in 8m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/08eb91ff5acd4056bfc30fa20e5ca8ad : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/63c50aa2a0b8407d991f8806786cf794 : SUCCESS in 5m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/abfb3c588f5145bea1e3c177d43c0e25 : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/013a70b9ce1f48c89fa665ee9004bc50 : SUCCESS in 1h 21m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3fb276613ff74bbfbfd81254bf932e03 : SUCCESS in 38m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7fbd93bc64774a0d830da0ed0b64d8fa : SUCCESS in 1h 12m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc42ae2d7dce4a3b8726b2cde92b48b5 : SUCCESS in 1h 48m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6935a7cae9e2cccf3466912b2c25bcab6a7efe4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-10 17:44:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlooking good","accounts_in_message":[],"_revision_number":1},{"id":"c1fc4f80e5411e48be855ce9affc244e7b1803a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 17:44:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e483f79f85473097741f0ea9afdc82c40e8b194","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 18:52:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"33acb8c5559f8011e1f7b54ba2c19925b9c08a34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 18:52:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/12727d39674a47138904af9b858497c8 : SUCCESS in 51m 07s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d6273b0f053847e7aede270c83a7957a : SUCCESS in 57m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29247d91a2174614ad0d5a2fb5b429e6 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0724ef350b243a5bd562b5b8c1dfc83 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf98250c6321404c905bbbf950c6abe8 : SUCCESS in 5m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cef86850934345c5b9291d2e01f381a6 : SUCCESS in 6m 37s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d59f0af7fa704fd188f83b6dbeeaab65 : SUCCESS in 4m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9eedf79bbce04aafbb61389112ba990d : SUCCESS in 34m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9657ed249163d82dc501463cf07fb0158d88e859","revisions":{"9657ed249163d82dc501463cf07fb0158d88e859":{"kind":"REWORK","_number":1,"created":"2020-03-10 14:43:09.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/00/712100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/712100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/712100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/712100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/712100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/712100/1"}}},"commit":{"parents":[{"commit":"235f641737ccf0390ab14a01312962911366eba4","subject":"Fix renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/235f641737ccf0390ab14a01312962911366eba4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-10 14:42:39.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-10 14:42:39.000000000","tz":60},"subject":"Bump versions for rocky","message":"Bump versions for rocky\n\nChange-Id: Icca01cf4083a095916b56ff70d3575591ae3abcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9657ed249163d82dc501463cf07fb0158d88e859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9657ed249163d82dc501463cf07fb0158d88e859"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
