)]}'
{"id":"openstack%2Fkolla~712419","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I5d1bcb0f7d7902b4e122c12697483433bcf52ca2","project":"openstack/kolla","branch":"stable/train","topic":"add_systemd_source-stable/train","hashtags":[],"change_id":"I5d1bcb0f7d7902b4e122c12697483433bcf52ca2","subject":"Adds python3-systemd for ansible deploy interface","status":"MERGED","created":"2020-03-11 13:26:12.000000000","updated":"2020-03-16 10:37:58.000000000","submitted":"2020-03-16 10:31:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712419-1584354704919-72e87165","meta_rev_id":"1893776cae927ef1237ac6b1d7d90480c3210ac0","_number":712419,"virtual_id_number":712419,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-03-13 09:21:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-16 10:31:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":2,"date":"2020-03-16 08:30:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-03-11 14:06:30.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":1,"date":"2020-03-16 08:30:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-03-11 13:26:12.000000000","_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-11 13:26:12.000000000","updated_by":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"reviewer":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"state":"REVIEWER"},{"updated":"2020-03-11 14:06:30.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-16 08:30:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-16 10:31:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6da0dd1a87d9bfbaa772ae9ad62b63c2d94cbb03","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 13:26:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"feff2eae38c5a7552608638ad7e353b74d8f2fbd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-11 14:06:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44d790b6e03e41a6e928555c916706227a02aabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 15:35:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/da0e8e9402a147e5bbc730bac84e5f49 : SUCCESS in 49m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/48e4b87085ac4d49a7c45d955b76643c : SUCCESS in 50m 23s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e602ef6c9bae43d8b6ef2893f6cf25dc : SUCCESS in 27m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3712e7a014c4463281bdc38e6c296b62 : FAILURE in 31m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0424cc43fc9c4b3d9093853b74450b00 : SUCCESS in 1h 36m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fc0ec88744ac498eada44f2b404db6fd : FAILURE in 33m 11s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/69b3e7a3d6fd47529947323dff23816a : SUCCESS in 1h 14m 57s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ee1697932b624606b1ea98656756a49c : FAILURE in 1h 07m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/141185740ae7498e85f2b9b5f1156600 : SUCCESS in 1h 30m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/53969cef6b0c4732870e4dd3e033f667 : SUCCESS in 56m 30s\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/7ae6072461624faeaa8a1b3357172ff2 : SUCCESS in 1h 22m 56s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/08a1e9907ef64ad9a4033b9adf99f1fe : FAILURE in 27m 43s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4b78a726e021418489fef2c42774c6e4 : FAILURE in 24m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f90d1b9f27834d86bc72f33f1a4ccfe1 : SUCCESS in 23m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/388a05ab4dd24dfc87ed4cc714cd9d0d : FAILURE in 24m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c111f47c985455b9ae8077a0764b889 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4feeb77a52645fbbbcf6e9568e2720d : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2be796e6739a452f9930e47fa788a983 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c104605434f478a8c15ddb25264f6f9 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/739acdf7aa924e238bdd9b374912ac61 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6361390989f641a695ebfa598b9fc642 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f053a301e07a4a5ba4c0ff5c7208f272 : SUCCESS in 9m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3284a32e5b8a45e59ddc4117dd295ee1 : SUCCESS in 26m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4cbc3f4488fa4b759b4a99e9a807f325 : FAILURE in 37m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c25e059ab9f4489d93301a339f76bd91 : FAILURE in 23m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8158ec7c2bc54a959edb9c047a7b1bd9 : SUCCESS in 1h 01m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d36778c3a3b4e799cf230876348aa4e : FAILURE in 23m 48s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/efcd230dfa7f4a9b837e8d56b0362854 : FAILURE in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be5f9f8a40cc3cfd8899eb37a1a361d78909c15f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 09:21:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7754778467811ec150c3b1ae7717490273e8e132","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 11:11:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e550281f2d20467d80dfc0bcfd857fb7 : SUCCESS in 43m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a89e46d50610483791f0f4b148738160 : SUCCESS in 45m 12s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ebe93000799b4dfcbd3a22ca47813e51 : SUCCESS in 33m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b80a6961b1cc412aaca5e768e0bc0011 : SUCCESS in 40m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cba3b88c0bf245f997d158b0beb4302e : SUCCESS in 1h 22m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e6a5ffab12ee446097e3b96a475c7edb : SUCCESS in 1h 08m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/91d2c12c23d34ef1b8bc1006a3c08436 : SUCCESS in 1h 16m 35s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c5433178aa8d496c995c04e83e5bede5 : FAILURE in 1h 15m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8001837d494843e8867fb57676a43d05 : SUCCESS in 1h 41m 33s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/70898f082a474db7a4f3d14851e019f4 : SUCCESS in 48m 25s\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/20b952460be941ecb71a3cb90b0476ef : SUCCESS in 1h 27m 26s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/1e908b9a491b4cc0a81767a3d6c5bb60 : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b40149db7edb49f3b2393d44d3f39463 : SUCCESS in 26m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7c3c7af6c98a4010ac4be0a803b74c21 : SUCCESS in 18m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/80b386476ad74be182609feed65b52c8 : SUCCESS in 1h 17m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e543594463784d29a8b561d568a7c497 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50c8a93d36e4f4aaf08c3d1fc2e9b5b : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e51a85a5aa24a428a08e7aceba52286 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/686b59bf96c34d2eaa7cedfe3afac0f1 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/849752d3d38948b48123989dd667f521 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4771c42dc964f7598196ff07d802341 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7640e479986c45aaa03c4f4a99efaee0 : SUCCESS in 5m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/16727f0568b649d19223110e584b9a9f : SUCCESS in 23m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8b89c106d4b450e977c473b5ca4f03a : SUCCESS in 34m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/91b51a9b19cd4a40b0f4269f6b6e2d4c : SUCCESS in 57m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2dedef54de3e468d8b77e79c845e0994 : SUCCESS in 55m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/777892204d504aba851deb53d858a7a5 : SUCCESS in 1h 03m 12s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/47428e4f9dee4beaa8bc015c7d73eccc : SUCCESS in 1h 03m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a0c163714bd4a8085d8ca5af75a3687e9ba199a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-03-16 08:30:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c55aaa72ddd92bec0dad811c95051bd367d8a01e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 08:30:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a1974128b061daf34cfd093c7483dae0b101144","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 10:31:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"99f5f50433bfbffca3cf8bbe022d5332da40fb42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 10:31:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/026c3126934f4eb6be41643fabc8b19a : SUCCESS in 1h 03m 37s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/929998009a244abcb407308bdf96417d : SUCCESS in 43m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/989bc452bd9d436c8b048024b50cfcce : SUCCESS in 1h 39m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6b85b443f8734bfa8e863eb58cb7f3b5 : SUCCESS in 1h 11m 40s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/95164aac87eb40bd9ec1ef07ed778d2c : SUCCESS in 1h 49m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fe6e1bc39972412a8967ddc29f6856fa : SUCCESS in 1h 01m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38b84205d30b445dbfd3d44e9a6a759d : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a3ec2d5eb94820912b8245d14ddad6 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/604f0b16b4f6490d8c55743cb074c83f : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/910dd5b96c784c158d8fb5fca850f956 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dece15d766f243ff808d5e90c496f8dd : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b14c3cfd954f1e911008f1a8c2557a : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a0bc24844ff446b86cb8d736c4d8e3c : SUCCESS in 5m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/64c512474fba4fa8a7f00027c168aacf : SUCCESS in 30m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c941a17bb6e4f0da5ca6e1a53085127 : SUCCESS in 1h 38m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ded5c72d799d402cb574b3bca755b177 : SUCCESS in 1h 40m 12s","accounts_in_message":[],"_revision_number":1},{"id":"140a2ad0e7738c9f077acf4ea93cd4ef4a5ceb34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 10:37:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f6dcdf421b4482bd7a972cd9f4ca5a : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1b298dff5724437ab2ce498b9c8f660 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2a18a53423cde85e26c1611ae2b2a915b99bc3a","revisions":{"b2a18a53423cde85e26c1611ae2b2a915b99bc3a":{"kind":"REWORK","_number":1,"created":"2020-03-11 13:26:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/712419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/712419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/712419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/712419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/712419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/712419/1"}}},"commit":{"parents":[{"commit":"c421ed52ae7951e20c9d8caae60da5ae77440f86","subject":"Fix renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c421ed52ae7951e20c9d8caae60da5ae77440f86"}]}],"author":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2020-01-30 14:16:18.000000000","tz":-360},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-11 13:26:12.000000000","tz":0},"subject":"Adds python3-systemd for ansible deploy interface","message":"Adds python3-systemd for ansible deploy interface\n\nWhen doing a source install, ironic doesn\u0027t have python3-systemd as a\ndependency and the current ironic-conductor container doesn\u0027t install\nthe required dependencies which causes the ansible deploy interface to\nfail with the error, \"Systemd bindings do not exist\".  This adds the\nneeded dependencies when doing a source build.  This installs the needed\nfiles for both rpm and deb builds on python 2.x and 3.x.\n\nChange-Id: I5d1bcb0f7d7902b4e122c12697483433bcf52ca2\nCloses-Bug: #1861427\n(cherry picked from commit 4a67e58f7971641c55247f02985e13ef391ca7ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2a18a53423cde85e26c1611ae2b2a915b99bc3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2a18a53423cde85e26c1611ae2b2a915b99bc3a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
