)]}'
{"id":"openstack%2Fkolla~712729","triplet_id":"openstack%2Fkolla~stable%2Frocky~Ieda14018b75d39c83cdc21eb4f9dd588dd8d7040","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"Ieda14018b75d39c83cdc21eb4f9dd588dd8d7040","subject":"Bump manila version for Rocky","status":"MERGED","created":"2020-03-12 16:15:06.000000000","updated":"2020-03-17 18:47:15.000000000","submitted":"2020-03-17 18:47:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712729-1584470835449-f42c30c1","meta_rev_id":"e77ed8a518cfd49aca5985ec4b00da7c089027cf","_number":712729,"virtual_id_number":712729,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-17 18:47:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-03-17 18:27:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-03-12 16:44:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-03-17 17:41:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-17 17:41:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-17 18:27:15.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-17 18:47:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"23528155df6ad0f29709a953482a79b9597a3f4a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 16:15:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd169114423e8c2935667742faf77673ba4ef5b0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-12 16:44:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ab31a2bd7e35659a6ce89534ca3a7b2f2663bf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 18:21:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d5102fef15d740d893c9e74520ad5f7d : SUCCESS in 41m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ca37b4a8e6b64b8ca238580afbb141f8 : SUCCESS in 59m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3b8968ecebd54ec385672c5f165aad86 : SUCCESS in 1h 26m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8c03fca4ce204d8d8e14d356a6f3ed0f : SUCCESS in 1h 10m 33s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/69d20726d75943998663f56be5187ffb : SUCCESS in 44m 54s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/309b6a4e4756412cbdc3c191fba9865f : SUCCESS in 56m 09s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/62bf951b576c461f8e551dd5ed2bf6e5 : SUCCESS in 1h 26m 47s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/cae0c83b34244bc3885fdaa5a705d3b6 : SUCCESS in 1h 07m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/004d882098a746889db942902ffccf9b : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc99e1b0ebe4f85a68adb3c7ce6af1b : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f69bb4c2ab6419987359a93f724f61f : SUCCESS in 4m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9eb572dc93e044ee939fa984c036ab2f : SUCCESS in 4m 58s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/a7c1b253f11f48e5aef475ad086a97b0 : SUCCESS in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/927e18f8287142b58276871b68a13c2c : SUCCESS in 1h 30m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b017f0a040ae4156b123bad545d84d3f : SUCCESS in 36m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c1f61c9936f4f23a09d75d39e07608f : SUCCESS in 1h 02m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4cf620c5bad7426a86650729aae71dac : SUCCESS in 1h 57m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0a768a5859313e8efbc32ebadc5f0facfb90007","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-17 17:41:55.000000000","message":"Patch Set 1: Workflow+1\n\ngrace period","accounts_in_message":[],"_revision_number":1},{"id":"2262a65fda5ff5077dafe8916bce301ff08106f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 17:42:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0906a636976eea9a8aabe9e798e732d694efb152","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-17 18:27:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc432613d7a1a81c57cc3396ad9aa1b7602d4d8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 18:47:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/37db5eb4abc54a89bd8a48b136521687 : SUCCESS in 51m 35s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e7f47752e2234c4691e9b44bbdf05525 : SUCCESS in 50m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eda6bb7052e3410d9e7a1905feffe465 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533ca412a9664ffca61a9714f3e78da5 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ee3465f82c54af18722b98dd4973d1e : SUCCESS in 5m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5e402b4e097647f1a836bb8ad40ab4f1 : SUCCESS in 4m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e2115dc0245f41eda86cd392049d8929 : SUCCESS in 6m 24s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0d2feb011ba41718cd3382429ae2714 : SUCCESS in 36m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e77ed8a518cfd49aca5985ec4b00da7c089027cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 18:47:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c68f89753c647e8fa36e291a10348dde16d0f6d","revisions":{"4c68f89753c647e8fa36e291a10348dde16d0f6d":{"kind":"REWORK","_number":1,"created":"2020-03-12 16:15:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/712729/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/712729/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/712729/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/712729/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/712729/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/712729/1"}}},"commit":{"parents":[{"commit":"1e146739da9254527e4fbb467cd7a503688a0cad","subject":"Add ssh clients to Ironic Conductor container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e146739da9254527e4fbb467cd7a503688a0cad"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 16:14:58.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-12 16:14:58.000000000","tz":0},"subject":"Bump manila version for Rocky","message":"Bump manila version for Rocky\n\nChange-Id: Ieda14018b75d39c83cdc21eb4f9dd588dd8d7040\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c68f89753c647e8fa36e291a10348dde16d0f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c68f89753c647e8fa36e291a10348dde16d0f6d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
