)]}'
{"id":"openstack%2Fkolla~713054","triplet_id":"openstack%2Fkolla~master~Id67c58580d1eaad86cacc5094ee80a734ac1eb26","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"Id67c58580d1eaad86cacc5094ee80a734ac1eb26","subject":"CentOS 8: enable monasca-grafana image","status":"MERGED","created":"2020-03-13 19:42:34.000000000","updated":"2020-03-20 17:18:40.000000000","submitted":"2020-03-20 15:57:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713054-1584719827541-eaa21e4c","meta_rev_id":"490a7bc8b0024e41c5b8acd58a096a4d08bbb81d","_number":713054,"virtual_id_number":713054,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-03-20 15:39:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 15:57:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-13 20:16:31.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-03-20 14:26:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-14 09:25:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-03-20 14:26:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-03-19 17:05:15.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 20:16:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-14 09:25:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-19 17:05:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-20 14:26:47.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-20 15:57:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b905d49d3703fdd105ec911652849b884715065","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 19:42:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f2697f487e766bea2f5a893c627f14f4421d697","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 19:46:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8e865f0b41e4a91b309767225524c0f : FAILURE in 3m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de322b7311c472e5cc9e85eb972f392ce899400b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-13 20:16:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/0754cc096ba94c5689fd91d8bf2cd085 : SUCCESS in 26m 29s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/895ca4e1af1a4161ae5033355f50d659 : SUCCESS in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4f5148fccad80355e97bbc067649d053d12f967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 21:14:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/36c53ae7016b4351bba1489b4c593ab8 : SUCCESS in 23m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/55a944087ad148f8a372d817aeff8142 : SUCCESS in 29m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e529a4802a0f4e9591a8b14330e47a41 : SUCCESS in 1h 25m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bcd1aaf7fbcf4a4ba3bf4b82b452796f : FAILURE in 48m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/94556ea962a04aa6bb40e0c14df9a9e2 : SUCCESS in 1h 29m 55s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/33e68b618bab493ba35737c687d5d6cf : SUCCESS in 47m 12s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/61de2a4160f5415eb4062413b9c85365 : SUCCESS in 25m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f8c6b45e15c84e0c91d0c8c3ef549f8a : SUCCESS in 18m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fe92005ffa8347d6bdcd2311051ed7b8 : SUCCESS in 59m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96b6e193877e43a0be681c7939ed35b3 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcbc615b57e44ce9be20a3061b821187 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19d61ec7ef994e05b01b12b4cf97a47f : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/300f29bf976046fe85775d5e8bedf525 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21fc3dc88d464f18b63fd6e710f34c58 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac563578bf484607ac268580305cdef2 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/32fc77c466ea497a8a74648b4d59a86e : FAILURE in 21m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a09ce83db19d48438f1f8e6358d47abf : SUCCESS in 25m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d38a5b3c8ab40cd8b44a9a84060a22f : SUCCESS in 54m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/14ea0fdb92e942af9bf6205ec81d1ca1 : FAILURE in 20m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/95225e7fe8194b50b8d55d78bfb4ed4e : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0511f4056397edcf17076df9a86dc471344d1f88","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-14 09:25:42.000000000","message":"Patch Set 1: Code-Review+2\n\nlet the monasca-grafana break our gate again! (from time to time)","accounts_in_message":[],"_revision_number":1},{"id":"1e13a6488d5cf20c3bbe9293b9e430898c73e673","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-20 14:26:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlet it rot","accounts_in_message":[],"_revision_number":1},{"id":"78103b2f7f1b915d30293ad37d4fae45778dda35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 14:27:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9ad323fa9d46d3d7b449479bc08b9dd7e99dad6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-20 15:33:43.000000000","message":"Patch Set 1:\n\n\u003e let it rot\n\nIt\u0027s \"let it roll\" spell. \"Let it rot\" is -2 -1.","accounts_in_message":[],"_revision_number":1},{"id":"461d4bb180f2398da1800752546457a6644cbfa0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-20 15:39:21.000000000","message":"Patch Set 1:\n\n\u003e \u003e let it rot\n \u003e \n \u003e It\u0027s \"let it roll\" spell. \"Let it rot\" is -2 -1.\n\nI think he was implying that it will break at some point...","accounts_in_message":[],"_revision_number":1},{"id":"f1f6e7bd8367ff860625daf98d27eb20d214e005","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 15:57:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ae9f5167946a495b43633fe80d610bf6e0820234","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 15:57:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/09e07d3f80ea4615b6053186f2ff0bb1 : SUCCESS in 27m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4520399043e14dfcbe3a9852510b6b33 : SUCCESS in 1h 22m 43s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/aa25526154024acc86330d00f91b1cf2 : SUCCESS in 1h 24m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49d222f65fc24570a694a1748836a19b : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c36d6bf9272426fa18d0a28bdaadbd5 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18643ff12d4d4eccbbe1b56935a87de6 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/691f261b13f54e8d908731d9ea7eecda : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b0f589ffceb47d69d8cac3c38efe112 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0516a9b1ca72424ca8d1c3f12627e50a : SUCCESS in 23m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb2057d42a8946319f49d46ec365094c : SUCCESS in 59m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c63bd6580bec4f8e91326f923d1338d7 : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cc4391bc9b54cee2d4c2db60cdb2ab245247ba91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:05:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/252fe8ce4d95462bad49d1b1c4216f06 : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":1},{"id":"cc1af286edaece3f4680ffcbdfd8a87cc19d7a81","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-20 16:36:56.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 9ebbbbff171d36b9746bc3331154577de66b8b5c","accounts_in_message":[],"_revision_number":1},{"id":"04d03f24efa9b23a134227d2f82ec0c7a035c3cb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-20 17:18:40.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e let it rot\n \u003e \u003e\n \u003e \u003e It\u0027s \"let it roll\" spell. \"Let it rot\" is -2 -1.\n \u003e \n \u003e I think he was implying that it will break at some point...\n\nAh, that is true.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0c1647352dda3e785612c6c66e07b97290ae85d","revisions":{"b0c1647352dda3e785612c6c66e07b97290ae85d":{"kind":"REWORK","_number":1,"created":"2020-03-13 19:42:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/54/713054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/713054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/713054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/713054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/713054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/713054/1"}}},"commit":{"parents":[{"commit":"5498671b344e48d6db865887537ff15cf4028510","subject":"openstack-base: pin setuptools \u003c 46 to get horizon working","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5498671b344e48d6db865887537ff15cf4028510"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-13 19:28:35.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-13 19:28:35.000000000","tz":0},"subject":"CentOS 8: enable monasca-grafana image","message":"CentOS 8: enable monasca-grafana image\n\nChange-Id: Id67c58580d1eaad86cacc5094ee80a734ac1eb26\nPartially-Implements: blueprint centos-rhel-8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0c1647352dda3e785612c6c66e07b97290ae85d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0c1647352dda3e785612c6c66e07b97290ae85d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
