)]}'
{"id":"openstack%2Fkolla~713591","triplet_id":"openstack%2Fkolla~master~Ia7df926aebcbd63aa5703f326a81d3890ae82291","project":"openstack/kolla","branch":"master","topic":"dnm-no-binary-wheels","hashtags":[],"change_id":"Ia7df926aebcbd63aa5703f326a81d3890ae82291","subject":"DNM: No binary wheels","status":"ABANDONED","created":"2020-03-18 08:31:14.000000000","updated":"2020-06-15 17:29:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"51b469a0a88e7cdc1f8a4b4adb287935f8ece8d3","_number":713591,"virtual_id_number":713591,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-18 15:52:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-03-18 15:08:43.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 09:06:10.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-18 15:08:43.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-18 15:52:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0babae975a3c085680acf0ac2fe4137f9e738abd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-18 08:31:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"761ac8f0328d7fe33fd070e0289985b47b63e31f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 08:50:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6c45061994b4b72a1816bb573b5c45c : FAILURE in 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3f578c467ccf690cf7a5d9870a75cca1d1c77a3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-18 09:06:10.000000000","message":"Patch Set 1:\n\nWe run \u0027no binary\u0027 on aarch64.","accounts_in_message":[],"_revision_number":1},{"id":"616751a648cb36626419c4a936346ea3fd2a937c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-18 09:33:41.000000000","message":"Patch Set 1:\n\n\u003e We run \u0027no binary\u0027 on aarch64.\n\nHmm, though with a limited set of images. (Really still no binary?)","accounts_in_message":[],"_revision_number":1},{"id":"0758ce9865f64f6e958c617e3d90c1dba879d503","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-18 09:34:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/7f9997e0f4b447d5b604de0e829beb4d : FAILURE in 6m 08s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/dca4448bfb85485abef81b5191794fe2 : SUCCESS in 1h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a08301c9d1298c53bec653ac17495cdae4d1d5d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-18 09:51:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8baa88ea4ee9601ade27f1ce36f1d0c91f15b19f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-18 10:24:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/50603f3d393e407fbf57edd43e4302c6 : FAILURE in 6m 43s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/02a217054d7d452bb6ca2e8d9d1052e3 : SUCCESS in 28m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d60fc891882e17bf27708d68857d171a3b272fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 10:41:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/539853b0601c4bfe8c4fb8349ae060d8 : FAILURE in 45m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d88242d9eb3305c578310301d0c37ec2424f0ca2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 11:26:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/528df0b1cd43470592eb7df3906e3c63 : SUCCESS in 54m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e2ad4bd38daa4c589847ebfc3519cabc : FAILURE in 1h 01m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1fdc052a83b14d8f86f01b745449572a : FAILURE in 1h 21m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/599d4d5ccca949529b1f2c8f59faa36d : FAILURE in 57m 05s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ada444961f8e4ceca46fb559bf5da21e : FAILURE in 40m 48s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/0f35692f0f3a44e6b0e55321bf05e159 : SUCCESS in 42m 12s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d8014946122e478baca45f460cb0dd27 : FAILURE in 46m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/116a5c1c4a1a42d492020c852214144a : SUCCESS in 19m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/49c5a19b70f8444daae64f62bf098e81 : FAILURE in 41m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5b5e6ae75d04e8f849be244f50c2163 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b26ebb3bc094598a91027f94bca3726 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca3f444650aa4893992233952a3c025f : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e940c8fe18e54155a31d4032daa21d79 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf802b908690489085c44905314e2913 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae1ceacd8674ead89a7d155ac8e302d : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c8e4e03778984716919f11fdaa3c916d : SUCCESS in 23m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a657d6766ba3443fab4bd92425598a70 : FAILURE in 1h 04m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1bb372de4e6049e3b57a15f78286f3cf : FAILURE in 47m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c14e556cc3a447db726ee2561b65a54 : SUCCESS in 1h 09m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aab54bfefd894664a11341e5a333fa93 : FAILURE in 40m 31s","accounts_in_message":[],"_revision_number":2},{"id":"4da9ddf4d920b70e2decc32e0fc927e391d068af","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-18 14:37:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"10749f209a4730a41a10bb27870ff88ca0a1926e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-18 15:08:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/45c88e41b5914a9eaadbfa7462928e5a : FAILURE in 5m 49s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/076bfbd02ea94f65bb0636a5b040dd68 : SUCCESS in 30m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6dc7b85864fa137148783c215bd83b82b7f770cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 15:09:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"72e19320d61b48970e4ef92779cb6f8c7d20b871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 15:52:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ff4c7c6b397e47ed8a309d3a06bad99d : SUCCESS in 22m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/dfb40eb3945c476abf04f813ff2fce6f : FAILURE in 1h 01m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a8e135f7fc954b5f8cce9d666311531d : FAILURE in 50m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c0f90681ba0f46c6ae766b774ed28d19 : FAILURE in 57m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3f1998a3dd1b4ece97d3a7f35d01780a : FAILURE in 49m 33s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/be0dc9a1e7324a939987557b34834a60 : SUCCESS in 41m 06s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/1266119cd2b34e209355229bd782d741 : FAILURE in 38m 37s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/a8ad611338c6412798be088ddcb6eaf4 : FAILURE in 51m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2fddc92a790e46e7955366c757a33c1f : FAILURE in 51m 28s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/8b8f185b4825490bbf7a4857e999db55 : FAILURE in 52m 53s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/24c526df0e4244ed84f6f9301999eb6f : FAILURE in 56m 09s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/fa9d9163810947ff8482bb54799ef3b4 : FAILURE in 48m 49s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3936153c345645bd92d715fbbf20471d : FAILURE in 50m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/87354af581e94f99b4676742a6f4c13d : SUCCESS in 21m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3fc858a2a17c4f4a826c1cf99ac7b5f5 : FAILURE in 43m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52907695166a4e5aa0ac7af39bf60660 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/591c48af94fa4e82bcc03fd107028a28 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce543b9756544431aa7e3950e9a7c58a : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b83d235f47c4a6b91bd62caa5b85b73 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3363d343ee24412c955bc6fc78d9dcfc : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff967e87ded4d2e97a72e73a403af2c : SUCCESS in 6m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bdc7b0765945432aa165e663a107dc55 : SUCCESS in 22m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5786159919004b13b4bfdcb405724dc9 : FAILURE in 51m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/62234751f8ae415bbcfcea54071ecbcb : FAILURE in 56m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a41dd1bd3de448cb0e38e75f074a5bb : SUCCESS in 56m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/05f3997e8d1542569112d36cfc3fb89c : FAILURE in 49m 37s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/71146bfb6c08461a8736728cb35b3389 : FAILURE in 44m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ad9462afb8844ee09729196f0ce55c77 : FAILURE in 44m 07s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/34efd136138a43a4b6a8cc51ca0fab27 : FAILURE in 47m 19s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/4fb5a786fa924928b297d9803886359b : FAILURE in 49m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9dcb03461a43433a6f5e1e01a1883110e5085f9c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-15 17:29:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"434d006b02104ffd591c3c1a129f44ba43b547cb","revisions":{"cdfcf805a269000df5330119639b27cd0851d444":{"kind":"REWORK","_number":1,"created":"2020-03-18 08:31:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/91/713591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/713591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/713591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/713591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/713591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/713591/1"}}},"commit":{"parents":[{"commit":"3e4f3b18dd34543c9bc00a3e07a02a565c8ea536","subject":"Fix Horizon build after switch to pyscss2 which lacks binary wheel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e4f3b18dd34543c9bc00a3e07a02a565c8ea536"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-18 08:26:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-18 08:26:39.000000000","tz":60},"subject":"DNM: No binary wheels","message":"DNM: No binary wheels\n\nWe rely a lot on binary wheels being available nowadays, and most\ndep packages remain, in fact, unused.\nLet\u0027s see if they are at least enough when binary wheels are\ndisabled.\n\nChange-Id: Ia7df926aebcbd63aa5703f326a81d3890ae82291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdfcf805a269000df5330119639b27cd0851d444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdfcf805a269000df5330119639b27cd0851d444"}]},"branch":"refs/heads/master"},"defc69a1f88e6330dd96821b790d71adf2ec6b89":{"kind":"REWORK","_number":2,"created":"2020-03-18 09:51:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/91/713591/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/713591/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/713591/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/713591/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/713591/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/713591/2"}}},"commit":{"parents":[{"commit":"3e4f3b18dd34543c9bc00a3e07a02a565c8ea536","subject":"Fix Horizon build after switch to pyscss2 which lacks binary wheel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e4f3b18dd34543c9bc00a3e07a02a565c8ea536"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-18 08:26:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-18 09:42:53.000000000","tz":60},"subject":"DNM: No binary wheels","message":"DNM: No binary wheels\n\nWe rely a lot on binary wheels being available nowadays, and most\ndep packages remain, in fact, unused.\nLet\u0027s see if they are at least enough when binary wheels are\ndisabled.\n\nChange-Id: Ia7df926aebcbd63aa5703f326a81d3890ae82291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/defc69a1f88e6330dd96821b790d71adf2ec6b89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/defc69a1f88e6330dd96821b790d71adf2ec6b89"}]},"branch":"refs/heads/master"},"434d006b02104ffd591c3c1a129f44ba43b547cb":{"kind":"REWORK","_number":3,"created":"2020-03-18 14:37:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/91/713591/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/713591/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/713591/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/713591/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/713591/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/713591/3"}}},"commit":{"parents":[{"commit":"3e4f3b18dd34543c9bc00a3e07a02a565c8ea536","subject":"Fix Horizon build after switch to pyscss2 which lacks binary wheel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e4f3b18dd34543c9bc00a3e07a02a565c8ea536"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-18 08:26:39.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-18 14:37:07.000000000","tz":60},"subject":"DNM: No binary wheels","message":"DNM: No binary wheels\n\nWe rely a lot on binary wheels being available nowadays, and most\ndep packages remain, in fact, unused.\nLet\u0027s see if they are at least enough when binary wheels are\ndisabled.\n\nChange-Id: Ia7df926aebcbd63aa5703f326a81d3890ae82291\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/434d006b02104ffd591c3c1a129f44ba43b547cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/434d006b02104ffd591c3c1a129f44ba43b547cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
