)]}'
{"id":"openstack%2Fkolla~713851","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ibae2578b4bc18c0ab7fcc1840c39bb86659ab920","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"Ibae2578b4bc18c0ab7fcc1840c39bb86659ab920","subject":"CI: publish AArch64 Debian/source images","status":"MERGED","created":"2020-03-19 11:22:54.000000000","updated":"2020-03-19 21:41:32.000000000","submitted":"2020-03-19 21:38:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713851-1584653883371-140d9962","meta_rev_id":"f12b310324d90fcf38d10cb4af23feb528666080","_number":713851,"virtual_id_number":713851,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 21:38:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-03-19 18:07:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-19 18:24:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-03-19 18:24:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 18:07:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-19 18:24:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-19 21:38:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b74405ff365a5ae9cd0fb53e056cf0413c8513e4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-19 11:22:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5987746ddcb0f4eed008f9ccc3e7f3f7b362f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:53:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7685483a7a014607880b9da3f403968b : SUCCESS in 39m 46s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d57e3289d49646d891f006f5bc9ad769 : SUCCESS in 59m 04s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5896b26465e7415f99960c81d1ffc27c : SUCCESS in 44m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8bd075a916054cab9aee01b220f98b8a : SUCCESS in 31m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/eaff11dd21e049f9b5625436664a1788 : SUCCESS in 1h 27m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f3b61beb3dee4242887569d9afbae4ed : SUCCESS in 1h 09m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f2b599ec5da44c10995eecc4baa0c0e8 : SUCCESS in 1h 14m 08s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/70513707efdf432a91cc92e3bacdd023 : FAILURE in 50m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/03b56faccb9a44a091b05152a17c4cfd : SUCCESS in 1h 22m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2dab3d480f404f7e8b9ec375e6988c48 : SUCCESS in 35m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f4b6fdf2f691473cbd678093944d9622 : SUCCESS in 21m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/361a77f07e114c1f834b92c86a0d35ba : SUCCESS in 1h 01m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5894c2cc469645809bf2cb217ad1790c : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d4635b22644f8981f8aaa976762fb8 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c44b08043584c738bfe0e4040ad9105 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f6d922800444c5ab1123d3182b9ca34 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8d76004d9de4118b6fdbd89ac747436 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c0d13098cb483dae96e4cd338c9102 : SUCCESS in 7m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e17477627724de4819749dbb986b737 : SUCCESS in 43m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e1b7e31b68147f499ee51ce1b86f313 : SUCCESS in 31m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dde34855587e4829b0c7c79791b37fc3 : SUCCESS in 57m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9cd45c3b23944748b4fde08e6745d5bb : SUCCESS in 57m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d3a6ab1458504ed083d53f88730c85e5 : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6d80fc87d6e0e67085a615706672bdb07d488638","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-19 14:04:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3967e4b5d3c4bca8792fedb20db0eab189deca69","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-19 14:05:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"171a895961b401b74c121bd24e3dce9bd7c59381","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-19 16:14:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ae69c2e8194012366d9b87a70d987e990e020a61","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-19 16:15:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"64583da9f355e3333cd949aae45b90755c62f65b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 17:55:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ef99d43cea0b48a1aa0aba2a6728a209 : SUCCESS in 36m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7cd293bb6dbb4f92a116301670eb4e26 : SUCCESS in 56m 26s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/73faad5b9fa74858902127755f890f4a : SUCCESS in 28m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/29cea7dc4b634c78b46eed371b76e01b : SUCCESS in 38m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/11830d413e804ba5a5625f936083649d : SUCCESS in 1h 24m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4fb7299ae9e14eb7a56e68eea73e6450 : SUCCESS in 1h 16m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f5052b75c61e4409966704cae4d1ff0d : SUCCESS in 1h 28m 10s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/98edc50871444e2aa1e4d1bec23b1313 : FAILURE in 59m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/11b410faa1f64fc6b50eb1fa04b565e4 : FAILURE in 17m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6aceeb6a0d1944929277cf01a2b5af76 : SUCCESS in 30m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cdfc62f92afb455ebef1fb22f0f48bc8 : SUCCESS in 22m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a4a1bf3f84104baaa1bd59e6bf4b7144 : SUCCESS in 1h 05m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95b7326bee1c41928350fff516ca1117 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/138a7cdf4690422cbacef1671f444cf0 : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b41ac313a6943b89bdd71caf4e6df8d : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e564a533be3c43eaa914968baaf646ed : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7cb99479d8f46cb9a19b81bde7f27a1 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/707b2fbccbc948bbaea3016b721ac2fd : SUCCESS in 7m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca37608985574c9f8aa0720ced2aa033 : SUCCESS in 33m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/51686457cf814207873fe82bdada942a : SUCCESS in 35m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe41d6e53d35430a9983466e0a66238d : SUCCESS in 1h 01m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2beaac6191f745a5b57d07488d15a9fe : SUCCESS in 57m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/537db0b0c4144f78a3752c59bf82571c : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0b779a2cfe20eca5b5287f4d11da8ace8ce60d5c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-19 17:57:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"01474cbf5a7541f80aa7f81e1ee8fd96d55ef246","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 18:06:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cfb8b890b0fc2f12e157744c27e19732a8981870","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 18:07:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3d6747806c427e5863b67d6ec1dd3362208e9493","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 18:07:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c62a99f34c203d3230e3bc86e446720c67adc0c7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-19 18:24:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"deee707aa73cd271ac9c70e0358ce185b38f3197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:00:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3105aa61388f471fa8d4cb3bf26c2290 : SUCCESS in 44m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1bfcf11f89d24581a20a98504a00b6b6 : SUCCESS in 55m 06s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8881bb18c1d5424abebce7cc1fcde04b : SUCCESS in 24m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cf60baf0916b4552a3333501fc7bfdea : SUCCESS in 37m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/701cb9157bd248a5b1ca2147c535f259 : SUCCESS in 1h 44m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/25253c024dc54bf3ba2bbc37af035cd4 : SUCCESS in 1h 26m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/509d85dc769a4684be7bb7565899a87d : SUCCESS in 1h 19m 30s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a50bda2de0954530ad8dbe47f422264a : FAILURE in 58m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/39a4846b5fe24aa78147f4c9d594661f : SUCCESS in 1h 28m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/636607df06ee48799b4effffd547902c : SUCCESS in 26m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d4751a7a1c574194ab23549b39d7f561 : SUCCESS in 25m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fbab7d2b231944e7ab0b2f65914253d5 : SUCCESS in 57m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b7fc8c3bcbe4fd6a339fd9f6cde139b : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dc7d4eb93fe42f79e6f60e1e9dbfa39 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd93956182624487851b68e751a0e8e5 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0883d683a896472796fb4dc9fe75a9fe : SUCCESS in 7m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13a6c044bed949378fd50e5b59521fae : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af698dfad4541009a9db33c50a09783 : SUCCESS in 8m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/482a27624ea84c89b16d23cc7147a802 : SUCCESS in 34m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3278d9bff8284617a8a691fd853dd31e : SUCCESS in 37m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ded2ab0989a54f61a9b43375aa785947 : SUCCESS in 58m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/27a0ae2366704f43960b16871b0ddb16 : SUCCESS in 58m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3581d90b4caa45f89c9134efcf963451 : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":3},{"id":"4f2de2af4dd7576835319b5bd9a289f41fdf8dce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:01:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fb6c55282080327c34d4ffea6649ac09f9cf341c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:38:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8847285f79cc4042a007c5d2e8c63158 : SUCCESS in 48m 43s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9471f850c6224760b15493282b47041b : SUCCESS in 36m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ff665bd20abe481c86ab4830d2c7b1ae : SUCCESS in 1h 24m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/29db8c151e00464a94107bcdd1d24bc9 : SUCCESS in 1h 23m 28s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/556d0c3db9c6451b815ca46cee724db6 : SUCCESS in 1h 31m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95da99fe447b481a9cc54f47f110ca86 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e547ad80ff69412da4d069d376d9e205 : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67d38249979f4b96b2d8051a01bbb44e : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a36f0afe0544ad9abd3e487ee152766 : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da5b10adc1db4bdf832d30b76fea6119 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1387ddd5e663441b85c171109384571f : SUCCESS in 10m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c3eb833464bf45c590fe64104f411bda : SUCCESS in 25m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07f65a4b9a25495da51cc10103812dd0 : SUCCESS in 57m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2936c47a2bbd433997b681bc0e9dcc3f : SUCCESS in 1h 01m 25s","accounts_in_message":[],"_revision_number":3},{"id":"37ac048b4a76e0216ff731b0c230450a96761596","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:38:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f12b310324d90fcf38d10cb4af23feb528666080","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:41:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68f4257090c143ec923e1c19c28e52e6 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d43c8b78ac9e1165445e4748c4f5b9e757e5dd8d","revisions":{"3262b02d1e2d9c89752ec0e5e1e26260fb896534":{"kind":"REWORK","_number":1,"created":"2020-03-19 11:22:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/713851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/713851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/713851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/713851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/713851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/713851/1"}}},"commit":{"parents":[{"commit":"ac08a2d595e88e078967fb40aabb52d827bd0f12","subject":"Merge \"Fix Horizon build after switch to pyscss2 which lacks binary wheel\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac08a2d595e88e078967fb40aabb52d827bd0f12"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-19 11:12:28.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-19 11:22:36.000000000","tz":60},"subject":"CI: publish AArch64 Debian/source images","message":"CI: publish AArch64 Debian/source images\n\nWe want to test upgrades on aarch64. For this we need to publish images.\n\nAArch64 images with get \"-aarch64\" added to tag to not collide with\nx86-64 ones.\n\n(manually adapted from master change)\n\nChange-Id: Ibae2578b4bc18c0ab7fcc1840c39bb86659ab920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3262b02d1e2d9c89752ec0e5e1e26260fb896534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3262b02d1e2d9c89752ec0e5e1e26260fb896534"}]},"branch":"refs/heads/stable/train"},"8cb4579f46a1ee4c3ff3b587787c70992879a922":{"kind":"REWORK","_number":2,"created":"2020-03-19 16:14:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/713851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/713851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/713851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/713851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/713851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/713851/2"}}},"commit":{"parents":[{"commit":"ac08a2d595e88e078967fb40aabb52d827bd0f12","subject":"Merge \"Fix Horizon build after switch to pyscss2 which lacks binary wheel\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac08a2d595e88e078967fb40aabb52d827bd0f12"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-19 11:12:28.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-19 16:14:31.000000000","tz":60},"subject":"CI: publish AArch64 Debian/source images","message":"CI: publish AArch64 Debian/source images\n\nWe want to test upgrades on aarch64. For this we need to publish images.\n\nAArch64 images with get \"-aarch64\" added to tag to not collide with\nx86-64 ones.\n\n(manually adapted from master change)\n\nChange-Id: Ibae2578b4bc18c0ab7fcc1840c39bb86659ab920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cb4579f46a1ee4c3ff3b587787c70992879a922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cb4579f46a1ee4c3ff3b587787c70992879a922"}]},"branch":"refs/heads/stable/train"},"d43c8b78ac9e1165445e4748c4f5b9e757e5dd8d":{"kind":"REWORK","_number":3,"created":"2020-03-19 18:07:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/713851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/713851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/713851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/713851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/713851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/713851/3"}}},"commit":{"parents":[{"commit":"ac08a2d595e88e078967fb40aabb52d827bd0f12","subject":"Merge \"Fix Horizon build after switch to pyscss2 which lacks binary wheel\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac08a2d595e88e078967fb40aabb52d827bd0f12"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-19 11:12:28.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-19 18:07:04.000000000","tz":0},"subject":"CI: publish AArch64 Debian/source images","message":"CI: publish AArch64 Debian/source images\n\nWe want to test upgrades on aarch64. For this we need to publish images.\n\nAArch64 images with get \"-aarch64\" added to tag to not collide with\nx86-64 ones.\n\n(manually adapted from master change)\n\nChange-Id: Ibae2578b4bc18c0ab7fcc1840c39bb86659ab920\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d43c8b78ac9e1165445e4748c4f5b9e757e5dd8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d43c8b78ac9e1165445e4748c4f5b9e757e5dd8d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
