)]}'
{"id":"openstack%2Fkolla~713952","triplet_id":"openstack%2Fkolla~stable%2Frocky~I29028976acb884e799a035fcb76cb998e40e1171","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I29028976acb884e799a035fcb76cb998e40e1171","subject":"Bump glance for Rocky","status":"MERGED","created":"2020-03-19 17:54:35.000000000","updated":"2020-03-20 10:50:15.000000000","submitted":"2020-03-20 10:50:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713952-1584701415713-e3550527","meta_rev_id":"35664180ec7a20a69ebb54e7722dddbd84009dd6","_number":713952,"virtual_id_number":713952,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 10:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-19 18:52:05.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-03-20 09:55:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 18:52:05.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-20 10:50:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e02ed4122d98d6fa38162ec9971c42b600da40d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-19 17:54:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9a7d83ff3163754023cd4c32f8d477c25b1c995","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-19 18:52:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6399b6af5cb8310100fe77f7d968fca37dba963a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 20:01:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cd0b84802aad420bbb4b21f2620f8f04 : SUCCESS in 49m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8046ee326a7f4390abeb62ed3cb46bfd : SUCCESS in 50m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/62b18dcb38f34723acc54185a11ddda0 : SUCCESS in 1h 26m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cc418bf2e9044c9096d2759ea025c01d : SUCCESS in 1h 16m 02s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/5cf99e89ef1e482d923785ac71b5f437 : SUCCESS in 47m 09s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3a3ed7b203ce4d09b5dbc67250047492 : SUCCESS in 49m 15s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a0c56d40c44f41e6a216e422181ed47c : SUCCESS in 1h 12m 06s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/fffaf1bdb8da4bf5bdb9a8ddc032548e : SUCCESS in 1h 02m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da7599ae6eb84fc4942a4ef84542b0f0 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4254890331e949f180c01eb0b590e23c : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad83069dd41641f68c5f3e944acac280 : SUCCESS in 6m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a67e956736254a8cac657fe5b491ed44 : SUCCESS in 6m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/65f82ed112fc446081257829ebba9032 : SUCCESS in 6m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b92838e12ac7403cafac88bdc69a2664 : SUCCESS in 1h 20m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be3aa019e6cb49c494f4582a595fff79 : SUCCESS in 34m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/52b3a3eba55b47f7aa4069761f726599 : SUCCESS in 1h 02m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e153a1b4ea7b4803a60053a590a87495 : SUCCESS in 1h 50m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42aa1c0641aebce1e96831581ce2ca30b3121f9e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-20 09:55:22.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5ae4a78cdd5bc31dade5927c1c7f117fc5a08fba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 09:55:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9c20d1a82596b71dd7982d2d199c242a851635d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 10:50:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2c84d21b9e7f4b10920dce4ccf4a0230 : SUCCESS in 53m 24s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/bdbff39d4bf64737910df9fe8b479f74 : SUCCESS in 48m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79ca53603a744eb9a9f7ac09d8fb0d8e : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf65d953149d4da5abc3bf01cbf1d378 : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f81bf3a85a1940e9824a3ce1cd2a3cd2 : SUCCESS in 4m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f3b42f1dae664b76a7390e8749228fed : SUCCESS in 4m 42s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/39ba00d3a81e4af3b6a13f355cc11300 : SUCCESS in 4m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/33e3a74039d04a9295b62ad496599683 : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":1},{"id":"35664180ec7a20a69ebb54e7722dddbd84009dd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 10:50:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"703dedbb7f34c2b7923847943ba88e08b6b69809","revisions":{"703dedbb7f34c2b7923847943ba88e08b6b69809":{"kind":"REWORK","_number":1,"created":"2020-03-19 17:54:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/713952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/713952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/713952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/713952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/713952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/713952/1"}}},"commit":{"parents":[{"commit":"43d4c62fef5f349e78a6d55a4aa71c4a422b4bbb","subject":"Merge \"CI: Install tox\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43d4c62fef5f349e78a6d55a4aa71c4a422b4bbb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-19 17:54:19.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-19 17:54:19.000000000","tz":0},"subject":"Bump glance for Rocky","message":"Bump glance for Rocky\n\nChange-Id: I29028976acb884e799a035fcb76cb998e40e1171\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/703dedbb7f34c2b7923847943ba88e08b6b69809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/703dedbb7f34c2b7923847943ba88e08b6b69809"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
