)]}'
{"id":"openstack%2Fkolla~714391","triplet_id":"openstack%2Fkolla~master~I3aac1e6a7f7c5f839b036f15713af949fba206cc","project":"openstack/kolla","branch":"master","topic":"bug/1868574","hashtags":[],"change_id":"I3aac1e6a7f7c5f839b036f15713af949fba206cc","subject":"Add nfs dependencies to glance container","status":"MERGED","created":"2020-03-23 07:37:58.000000000","updated":"2020-04-04 13:55:52.000000000","submitted":"2020-04-04 13:54:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714391-1586008456981-fe09d30a","meta_rev_id":"1610ec30a0c65729a5f24c21968a2bf1d7020699","_number":714391,"virtual_id_number":714391,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-04 13:54:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-04 10:55:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-02 08:09:32.000000000","_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":2,"date":"2020-04-02 07:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-02 08:01:07.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-02 08:01:07.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 11:49:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-31 11:49:29.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2020-04-02 07:53:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-02 08:01:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-04 10:55:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-04 13:54:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26a2297c747bfd85fcccfa183ffe20ef89dd0559","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-03-23 07:37:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd54b626090940f66ab1376709febe5f0036acb3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-23 08:03:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f1757bffa8b842499e6df3a68654ffde : FAILURE in 6m 21s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/776bc32f9f994967a7b295fdf3c3b468 : SUCCESS in 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c51aecb4203df90dc9d1976e9665e6e54ae8d5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 08:22:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e138f826165e4e798093451e48542722 : SUCCESS in 43m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c104df90321296f105933c506cd2834a3869421","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-23 08:42:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a134ba53e876b84abdb162f1e9d93af7c454892d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-03-23 09:09:31.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nPlease find my reply inline and let me know if there are any further queries.","accounts_in_message":[],"_revision_number":1},{"id":"5626e8683b1b7bed94c3fe0b91b75200dd16b55d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-23 09:15:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"214a16ea6c552dd3789c32c276bb60ee04a14d93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 10:58:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b0b74db7d08746aeb272ef79d2f1f146 : SUCCESS in 26m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c546b87883bd441fb425d360040083aa : SUCCESS in 27m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f97fc74865a748769815fb4af40047ed : SUCCESS in 1h 13m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/212af69ffce441478c57a391133b7096 : FAILURE in 50m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/04b1169c318d4ef2949b6832bf933d87 : SUCCESS in 1h 24m 32s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/1177d0127a8943d9bc0cd27872475f1b : SUCCESS in 43m 55s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/39b4e038bfa945bcbc35c875b6c1a70e : SUCCESS in 1h 52m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d49d0541c94e4588bf44185cb03a5a01 : SUCCESS in 2h 05m 33s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/20c7f627933d4216a4df96404448a3dd : SUCCESS in 25m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6c6fa38eb49b411aa45bdf25eaf48ec1 : SUCCESS in 22m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3d9499f481874c10b2f175a833d7e4e3 : SUCCESS in 1h 03m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b463d55f39f4e5ca6541b267acb092c : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffebe8f71f14ca7ba22b9219bce7a26 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/328b5f5d72f44a12983744f2f142396d : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcae341ea6e944b1805e4f2752973352 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/064ad89a9b2f4142ad14d8a301bf0ab6 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d8f8269a614faf8eec7c4906b74730 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/037d77c1f9b042ee9edc593371e08817 : SUCCESS in 22m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/17d1b8bd31074fcfb5714314688d3243 : SUCCESS in 29m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2af6337a58154452b1ab9b9b2f3a1f19 : SUCCESS in 1h 00m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc405e1588a24a1898312687aef2a4d7 : SUCCESS in 53m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/86910a6b47504aeb909b6a91718e5419 : SUCCESS in 1h 10m 16s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ccb02ba344e84916af13be1a695c9b5e : SUCCESS in 1h 33m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e25685556c384c66bfcbd52c5180d8b6 : SUCCESS in 1h 50m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94f47edb7497722c0dde018c258241f826ec89c6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-03-23 15:06:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6de64650bb3a95b965b57aff47542f71b74663a5","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-03-23 15:09:26.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks for the review Radoslaw. please check my inline comments.","accounts_in_message":[],"_revision_number":2},{"id":"adac9b6cd979304ef512150cbfaa81ba449fcccd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-23 15:33:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a09818ad983045e287e394853b2e6c84 : FAILURE in 5m 48s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/6bc1823bdb0e4ba3a219cf26d7a85b98 : SUCCESS in 26m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a60483ba3d2519879c095c556b7bbfd6dd0a5d2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 15:55:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f635f5c541644f0969d81d6c0c4117d : SUCCESS in 47m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4084d9915181925ee05261e17f92a6d610c08ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 17:46:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3f758d1218994d75b4cde1f032b38ecf : SUCCESS in 38m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e37032709db845549a0d22f0c05a0ede : SUCCESS in 33m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cf10c5eb42874a8eb9807d9b8d00b62c : SUCCESS in 1h 21m 12s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/aadbb47a7c3e4309848abbc3e61c66ac : FAILURE in 1h 01m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fa944f26e0d745b8a8630903e64aa331 : SUCCESS in 1h 26m 06s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/9fab69e6879f427b9c0c822473e6c7bf : SUCCESS in 45m 35s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/875f458bd5e14f4aa9d0abfd96744ac5 : SUCCESS in 1h 49m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/33ff9a68f7764cfd9162abdafd83b5f4 : FAILURE in 1h 24m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3d0fa5daa2574fca9061a7ed4944954b : SUCCESS in 23m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a096577bfcd048ed85b09d86faca2774 : SUCCESS in 19m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b1776015d9f04177bd124bb5e0761616 : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/447d41441a9440948675d70bbdc9759e : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b108ebf55e354b4caf3972739445dc73 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d28d83f90084a0c8519acd0d9c21c7e : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/def60416c572477fa44f9181902d120f : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93837cab74154c90aef350b9b66864bf : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2bfeb27d89443a4ad319591026b055f : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eb7525116fd14871838465e7c1a3fc1b : SUCCESS in 22m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0c075cd25d0407b927aceca2bef04a1 : SUCCESS in 34m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bb0793ced014f66bb06a8f42a49d81c : SUCCESS in 1h 19m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f531abaee0964e8cb71ae90bf369ace8 : SUCCESS in 59m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d26f9d07d5a4c1198ff8a8c4b6ff795 : SUCCESS in 1h 26m 27s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5f1cf6118f1140419a3501140410a306 : SUCCESS in 1h 34m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f2e7fbe950d947c9ac1d0242818aef62 : SUCCESS in 2h 27m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"986f5f3542e37c3faed04479dcaefc26287ccf04","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-29 14:19:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"76d5964f4ef1d78d69e955b26bbabae761760f64","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-29 14:21:12.000000000","message":"Patch Set 3: Code-Review+2\n\nfixed desc and added reno","accounts_in_message":[],"_revision_number":3},{"id":"5d3d7a93d3873b44e25cba080d38bfa7d6804a6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 14:27:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"49105b8af4326932d24c4591458c839a999a704c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-29 14:53:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/fa1853747668441bb2e49c6373099e52 : SUCCESS in 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"91819b78d31cb64b8d82d8afa548af61644f30c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 17:18:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/75bb1bd78e984907a24b7add840673b7 : SUCCESS in 23m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/80c6fd4b0fc3429683642602810a3cfb : SUCCESS in 31m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/81ff5b1a1e9545c3a9aba0444ac3087f : SUCCESS in 1h 26m 43s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/70cd366d20894bc28f34370445790834 : SUCCESS in 1h 21m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1174200eedfc4bac9f0ff6e4eef716a8 : SUCCESS in 1h 20m 02s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/4d93de726afa448b9cd80b07303a6736 : SUCCESS in 38m 35s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1db71718188340919f89d3a8664ae9ce : FAILURE in 2h 06m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/3b60db8205084d7cac994d454d0ada84 : SUCCESS in 2h 43m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e076c7da40ef46eaae48941a34db1f6c : SUCCESS in 24m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e3460f9070c042ec9aa7d823705c715f : SUCCESS in 17m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3a42b83bd63140699c6b4e6f934293ad : SUCCESS in 1h 09m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec6fec45df654950a3c06079c327d12c : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3cd6a2d283d4bbc8ec82dec80d97f4f : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f28bb5d90f7460da03f87ed6b7676df : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24e0d4defebb45e3ada45709263ec57d : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6db72fbb7af243f5b925bed741dc57b6 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b1a851cd604f1e83acc20212b01ddd : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7afa04e4a83446258affa0a83b1f73cb : SUCCESS in 4m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/975a2190b66d486da20ef1b1d657c29e : SUCCESS in 22m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d644d2b46fa4951be08961d4c686543 : SUCCESS in 25m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee7d10152ebc4e01a8dd79b3225adc32 : SUCCESS in 1h 09m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/249e7c3d7bbf42aaa673a361e38dfcc4 : SUCCESS in 53m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/314cf1c363f04b76ad701d160806f8de : SUCCESS in 1h 06m 06s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d394b9a977634af29040a14e4cdf60cf : SUCCESS in 1h 43m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/fc9d4fad11db4634ad583ce240ceb58f : SUCCESS in 1h 38m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aacf48e5bd8ffa3bc90168d6503e6a3c82a3eb51","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-04-02 07:31:33.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1ef2b6444e7a30098d27aa812028610a594c911e","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-02 07:52:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1d3198fdb2f134f6514af4cc8c7ee0fdf1ea2a2b","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-02 07:53:46.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nThanks for the review Michal.","accounts_in_message":[],"_revision_number":4},{"id":"6dadc597f5b4bb542b9907749476f14b7c25578c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-04-02 07:53:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e5d920e880d1c8b3a778fbb059abf5f406a19c0e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-02 08:01:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c51484c0166ac1077fa2a7d4a2d9252941ac2960","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-02 08:01:44.000000000","message":"Patch Set 4:\n\n(it will probably get blocked on monasca-grafana, please do not recheck until fixed)","accounts_in_message":[],"_revision_number":4},{"id":"bad480e9394458228cb9c2a387e6f2e8dce1a826","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2020-04-02 08:09:32.000000000","message":"Patch Set 4:\n\n\u003e (it will probably get blocked on monasca-grafana, please do not\n \u003e recheck until fixed)\n\nSure, thanks for the review.","accounts_in_message":[],"_revision_number":4},{"id":"98b03e64b10c5beb60bedb47ebd26154d5592541","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-02 08:20:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/13188af69275433da4b2c43df1f75954 : SUCCESS in 26m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"500fb9ff9a029c135d26dce2330ecb998bd80dcd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 08:42:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a3678e6e19ff4eaa89ed5d48edff2424 : SUCCESS in 47m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b627583794fb48979d689bbb1ae5a3cf66f81b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 10:48:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/79d19280266143349a178f6e03fec420 : FAILURE in 47m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6fe211d639104f68903680d86ab45379 : FAILURE in 52m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b1c67775b9214c8fb162827d79fcd960 : TIMED_OUT in 2h 03m 52s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3e861ccf5c054d19b6d418033a8a7622 : TIMED_OUT in 2h 03m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/dc6bb8f7f59a4f988303e73c522f5ad4 : TIMED_OUT in 2h 41m 44s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/709abbb0e0b24b93a8438a33867a3c9b : TIMED_OUT in 2h 07m 52s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/589b5a82ba7243caaa9971d7a7d8ae7d : SUCCESS in 2h 00m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9dd7b9f7c9c749a3bdd44c0d0555d704 : SUCCESS in 2h 19m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6cf2e84e1866474496ecb27c0422a071 : SUCCESS in 1h 52m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d1e62c067aff499ab2da05585ec4e912 : SUCCESS in 19m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e1fa34f789d14e9684936720dd99c1aa : SUCCESS in 1h 06m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e042caf6ad1e458cb62909f3dfb1b895 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49b025813ae747bf8266e455c286c509 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/599ce391ad094338bc2e586b74b73bec : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5cde8e5054c4fbe857e06c85a0d9f03 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd087d865ef54d29bfc449fe323deda6 : FAILURE in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a978edc6a9c4ee99ad44a2eef19f004 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a03a8ed8486b42dc82361dc5ba347b05 : SUCCESS in 6m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9f47d4469a946db99ffdd708022b226 : FAILURE in 22m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f01a3b2628f440f787a047e1ff038c20 : FAILURE in 27m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e08221b548345be808bc45236f75c92 : SUCCESS in 59m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2c1f535c2584e688f1454c3b625d057 : SUCCESS in 1h 01m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1b2b6630e148419fb5a504dece7a7e08 : SUCCESS in 1h 15m 32s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/21d7e975abf84590a9194a900aa5b889 : SUCCESS in 1h 37m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e3e12bcfa68b4b4db5ee66c2ec1a70ad : SUCCESS in 1h 49m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a27dabaaf006576abc09b587c82b8b39738401e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-04 10:26:46.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a8a31dd590114dffef97daf3d3a8b50d12f44525","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-04 10:55:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/f1a9ed7375e94047b09cd9521d9565f4 : SUCCESS in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"756735cf7480a3a9bc55c01faf3f184e0b673491","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 11:12:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9fea19e4b0a64191a5965f277df2da13 : SUCCESS in 44m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"833adbc9c810e22786aea3aa3a94645efbdca8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 12:25:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b4eabdf86f5544e6b76bd36b78dd719b : SUCCESS in 31m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/47e3d677bedc47ba9716f8aaa7e422c4 : SUCCESS in 35m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/87d1a771199a4022aa11cbe06793aa51 : SUCCESS in 1h 16m 52s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/770934dc7baf4d2bb886fce66ee039ad : SUCCESS in 1h 22m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/64cd85d93f074249a240837a0d845391 : SUCCESS in 1h 26m 14s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/1e122eb88a094fc49ffabbaa5b020ebd : SUCCESS in 44m 00s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/848adf7b78ea46d0b49ccb16fdeb6e94 : SUCCESS in 1h 50m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/349a43870e234b8282022ce1bf8979e2 : POST_FAILURE in 1h 07m 46s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/96cad86c05544bd5a5945837c980d4e9 : SUCCESS in 30m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a8503ac13be243da98715e7ddf3ea002 : SUCCESS in 42m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bbc215b5ade54ed3ae90f63fc638329b : SUCCESS in 59m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73707b0b65aa46708df47a4b014ccc3a : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9b8286b994648158b67030021767e40 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd280d9a188643c4954f3358838e3ef3 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fea4574dc7542deba4bbbdd0bfcf698 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56e2250075774b62a901cbfe959a67aa : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ef4657651f436984f5753b20db882c : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5758f798ee84ead8c56e4d5ef5f9144 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d351e7045c8b4295974c0f4b34675898 : SUCCESS in 59m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f1cf825016484a3fabde43c80f885d95 : SUCCESS in 35m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/66c558baeaf840419836d5b013ebcc4c : SUCCESS in 1h 10m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/112fb23f2152408d845ace5980eba35b : SUCCESS in 1h 19m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/30c7644c611f4db59cb2d7d94e7235b4 : SUCCESS in 1h 38m 16s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0721cdda73ae4a369165b66732032c1a : SUCCESS in 1h 33m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/540fa148fe904b0dacd533b965082c8d : SUCCESS in 1h 58m 03s (non-voting)\n\nWarning:\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/netaddr/strategy/__init__.py in the commit\u0027\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/netaddr/strategy/__init__.py","accounts_in_message":[],"_revision_number":4},{"id":"d53cdcc88f1166b03ca7c9212d4881fc9b9b00c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 12:26:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"33d47006367d5b25d18832685ea798a27191519b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 13:54:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"051ddabb666be63cfc269786eaffbe134271dced","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 13:54:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/edd16d770e4144358218feda45874135 : SUCCESS in 30m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0f21936c280b44b79782a25b4ae06401 : SUCCESS in 1h 22m 38s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4b7f9599e2574d7ab70be0368f00db9a : SUCCESS in 1h 27m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/998c6ce3310b411abc6aaa21470ae20b : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20c6e57e80f94e15a7ee1e67454a495d : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b579b5aeffba4e86a5201e778a453af3 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b3c840bbdae4235adf81f77578e02c4 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090cde0ecaaa4f389e1b5da301545755 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84fdf41e44cd4e0895d3f5fb9e9c7274 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc5db05bfb644956aa728ddfed4b6995 : SUCCESS in 26m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed1d63c49be9430c8b0485b89415c60b : SUCCESS in 1h 06m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92b53abc6384465b8020bdfddb0f5c2f : SUCCESS in 1h 15m 06s","accounts_in_message":[],"_revision_number":4},{"id":"27117270309dd76f71d4bac841ed46aee166f026","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 13:55:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/330ed82a7cd342259be99b8d4c32fa3d : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/738f4dec6c8345dcadb491f9576e8ad0 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cac9d46aa6dbac6631bdcae67db125a88e1c3234","revisions":{"9af8b7e8bbd3f580a85ad48abb8cefd6b25ed87d":{"kind":"REWORK","_number":1,"created":"2020-03-23 07:37:58.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/91/714391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/714391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/714391/1"}}},"commit":{"parents":[{"commit":"fe8820be560c32832968d939ea7733ebd0848acd","subject":"Merge \"CentOS 8: enable monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe8820be560c32832968d939ea7733ebd0848acd"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-03-23 07:33:03.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-03-23 07:37:31.000000000","tz":0},"subject":"Add nfs dependencies to glance container","message":"Add nfs dependencies to glance container\n\nWith a recent change[1], the nfs mounts will happen in the glance\ncontainer(and not in cinder container) due to which the glance\ncontainer requires the appropriate nfs packages to do the mount.\nThis patch adds those dependencies.\n\n[1] https://review.opendev.org/#/c/709300/\n\nChange-Id: I3aac1e6a7f7c5f839b036f15713af949fba206cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9af8b7e8bbd3f580a85ad48abb8cefd6b25ed87d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9af8b7e8bbd3f580a85ad48abb8cefd6b25ed87d"}]},"branch":"refs/heads/master"},"22f949d5232190902a8c6b9cb5839c902b8a6c90":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-23 15:06:23.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/91/714391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/714391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/714391/2"}}},"commit":{"parents":[{"commit":"fe8820be560c32832968d939ea7733ebd0848acd","subject":"Merge \"CentOS 8: enable monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe8820be560c32832968d939ea7733ebd0848acd"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-03-23 07:33:03.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-03-23 15:06:18.000000000","tz":0},"subject":"Add nfs dependencies to glance container","message":"Add nfs dependencies to glance container\n\nWith a recent change[1], the nfs mounts will happen in the glance\ncontainer(and not in cinder container) due to which the glance\ncontainer requires the appropriate nfs packages to do the mount.\nThis patch adds those dependencies.\n\n[1] https://review.opendev.org/#/c/709300/\n\nCloses-Bug: #1868574\n\nChange-Id: I3aac1e6a7f7c5f839b036f15713af949fba206cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22f949d5232190902a8c6b9cb5839c902b8a6c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22f949d5232190902a8c6b9cb5839c902b8a6c90"}]},"branch":"refs/heads/master"},"6b4d33be7f0ae0453b080e05b7a9e65542b04b93":{"kind":"REWORK","_number":3,"created":"2020-03-29 14:19:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/91/714391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/714391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/714391/3"}}},"commit":{"parents":[{"commit":"fe8820be560c32832968d939ea7733ebd0848acd","subject":"Merge \"CentOS 8: enable monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe8820be560c32832968d939ea7733ebd0848acd"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-03-23 07:33:03.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-29 14:17:21.000000000","tz":120},"subject":"Add nfs dependencies to glance container","message":"Add nfs dependencies to glance container\n\nGlance API image lacked nfs programs/utils to be able to mount\nNFS exports.\nThis affected Glance usage with Cinder backend when NFS was used\nas Cinder backend.\n\nCloses-Bug: #1868574\n\nChange-Id: I3aac1e6a7f7c5f839b036f15713af949fba206cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b4d33be7f0ae0453b080e05b7a9e65542b04b93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b4d33be7f0ae0453b080e05b7a9e65542b04b93"}]},"branch":"refs/heads/master"},"cac9d46aa6dbac6631bdcae67db125a88e1c3234":{"kind":"REWORK","_number":4,"created":"2020-04-02 07:52:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/91/714391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/714391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/714391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/714391/4"}}},"commit":{"parents":[{"commit":"fe8820be560c32832968d939ea7733ebd0848acd","subject":"Merge \"CentOS 8: enable monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe8820be560c32832968d939ea7733ebd0848acd"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-03-23 07:33:03.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2020-04-02 07:52:42.000000000","tz":0},"subject":"Add nfs dependencies to glance container","message":"Add nfs dependencies to glance container\n\nGlance API image lacked nfs programs/utils to be able to mount\nNFS exports.\nThis affected Glance usage with Cinder backend when NFS was used\nas Cinder backend.\n\nCloses-Bug: #1868574\n\nChange-Id: I3aac1e6a7f7c5f839b036f15713af949fba206cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cac9d46aa6dbac6631bdcae67db125a88e1c3234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cac9d46aa6dbac6631bdcae67db125a88e1c3234"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
