)]}'
{"id":"openstack%2Fkolla~714667","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I8cac0ffa5c52b5b6060b8072d7d11c6e8bd9943b","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I8cac0ffa5c52b5b6060b8072d7d11c6e8bd9943b","subject":"Bump service versions (Train)","status":"MERGED","created":"2020-03-24 14:57:32.000000000","updated":"2020-03-24 19:52:47.000000000","submitted":"2020-03-24 19:49:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714667-1585079351820-a663f06a","meta_rev_id":"bec5edabc11680d862f8e02f5e27a9ebfa52584a","_number":714667,"virtual_id_number":714667,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-03-24 16:31:21.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 19:49:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-24 15:03:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-03-24 17:43:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 14:58:47.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-24 17:43:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-24 19:49:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f78637a5190eba058acebe9fa579f44ee0b73313","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-24 14:57:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28e9dc387599eebaad3422a02e2b56b0e1a4e8f5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-24 14:58:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7df8920ad1da7128546aa20962ae3eae43a09c39","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-24 14:58:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ede6d3cc3690d997651ece3b8a3b85eb9668386a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-24 15:00:37.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"629d22f413ecaf63c5d202ff9ea571d94bfa9a1d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-24 15:01:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ef3babef63b0b08292b46240c601db3822017eac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-24 15:02:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c194eb6a74c5cd460f212c5a09d32c7518bcff9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-24 15:03:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1bd8dd58bd3b0f47b87d1e2f4246c54ae3f2aa56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 16:29:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/227130269ab440c1b3f64202c647126b : SUCCESS in 50m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/feaad96ef25449fda8e018b229e76876 : SUCCESS in 50m 29s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fbd3abe7a5d7499987fd5c587be501f4 : SUCCESS in 26m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a3812f952f464941ac8d0f6d0d43e0cf : SUCCESS in 31m 46s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5de30b33e3c04e238595df4ea849e4cf : SUCCESS in 1h 25m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b0d17fb94e824a0686192768ac0633ca : SUCCESS in 1h 16m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a45ea8232c464b2492f148af07bb1935 : SUCCESS in 1h 19m 18s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/22b8d2bdcc5d42b880aaf21f3c8477df : FAILURE in 52m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e6e4e7364c644088900b3042d3c24c09 : SUCCESS in 1h 19m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bd18b0a4659d43bd933b7e596a52e088 : SUCCESS in 1h 04m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b844c04ea0314d399ba0b4131146b423 : SUCCESS in 32m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a52e6072cbb649a39dc426a77c931fbc : SUCCESS in 31m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cc0d890d0c334873a4354f3d8cd2e5e6 : SUCCESS in 59m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c76cc742ea5546728fb4aed210246374 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18652ea76eff428689cd5b2b94c7270f : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cffc18218dd74372ba6391ba2e079582 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cba98f5aada416aacdd7fbedb910226 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9bfa38266454f84aee46b0815ad0924 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc6ab18e0ce64bbf8103b9780bef212e : SUCCESS in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/76e3cee0f8fa448282292b06e2ebcfe4 : FAILURE in 10m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b9ac9f5edadc4295ada5526001a26962 : SUCCESS in 27m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a26a2824d69f43ce817097a297bf041d : SUCCESS in 1h 08m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/30689df29ece46a9b049ea42ca8f4e3b : SUCCESS in 1h 03m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f8af492f71d74a6ab0d203cb9f0065ad : FAILURE in 49m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e9df6db723babb9eaeda7bc44ae6b479922a611b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-24 16:31:21.000000000","message":"Patch Set 2:\n\nrecheck\n\nnetwork notwork","accounts_in_message":[],"_revision_number":2},{"id":"9956416c741e42f2949690b7531efe7e1a732c03","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-24 17:43:47.000000000","message":"Patch Set 2: Workflow+1\n\nlet it go when ready","accounts_in_message":[],"_revision_number":2},{"id":"e086c985a25fc703249e3d2bd98a320a0b2d793e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 18:13:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0de4cf46f17a40c89d097a71e803e7e9 : SUCCESS in 46m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ca7497d5478d4dc2bcdee9020660aee8 : SUCCESS in 1h 01m 52s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f0cb37790dbc4ae791c2fc23c5bb745c : SUCCESS in 30m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9c32bc19c6464479858d0990ed1fc26d : SUCCESS in 35m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fdc9077492d64fe7a048e4a781941622 : SUCCESS in 1h 32m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/03a9ab404dec41d38665d7bd63efe352 : SUCCESS in 1h 22m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a8e1727b0d5c4268ba1aeb90f53c04d0 : SUCCESS in 1h 28m 50s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6f430753c00b4354bbaf35cb41ad49ac : FAILURE in 1h 01m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aa71c0654ebc4567a32ba9c0dda63960 : SUCCESS in 1h 38m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3661934a363b4edabc565ef836e9ecc7 : SUCCESS in 52m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f478d06cae7d44479d15dfea5eaa26b7 : SUCCESS in 30m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3999353b19724448ab3aec103b67f58e : SUCCESS in 22m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9fff0826d8fe452b9af910997cae0682 : SUCCESS in 57m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87d0d3cfc52c413fb0f9b6b1f2151424 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d354aa5e2afb4e9e91ea5e68307ddf06 : SUCCESS in 7m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3d7602bbea7495e980ae6ddd9dfa4ac : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b032d78602654b35b0ff9a1499d8bbe0 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/094f94d053fb4af7b66d4e2f9fd7f0e1 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02b557893fd34212aebae00f34e3320b : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b6941f83233455fa8107ea3d074437d : SUCCESS in 27m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e09b6d4c92cc47cca18425658f2edb9c : SUCCESS in 35m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bee27676bb9f43afa6f25d2e8d917369 : SUCCESS in 1h 01m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3c59c9757e2471f93245ab4772e3c31 : SUCCESS in 58m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4d8e2cc24ed7424a8c47311d1c6565ef : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":2},{"id":"8ce9ed0781e6196b2ab7eae3e31a9d8de5e6c6b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 18:13:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c195bd030262c107c6ebb64623de4b69b4a76633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 19:49:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6baf2f0ca11149ccabc4ad38c20dce87 : SUCCESS in 45m 25s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3612f00be4eb4728942721cd930e71ef : SUCCESS in 33m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7ed6e10dc3e84d1d8670ac610d97b779 : SUCCESS in 1h 31m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9189813f512c4e5db6413b8cd09bde13 : SUCCESS in 1h 10m 03s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/17acb9bba5e647d08c21378ba8449af6 : SUCCESS in 1h 23m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b587ad08cd09439c94108560f712254b : SUCCESS in 49m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9355d3192a10459994d211b03c5e52a0 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d5ebdd8ad74051bd7881c8bb0b06e1 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8fffc0f6e586476d905a7abfb49e3177 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/650d3e6958d9417c8d41fa1d84c6ccfd : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3c3ebf7e337476ab3d040671315a4c7 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213e1ece52ea4bec9501be74d5e72be8 : SUCCESS in 5m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ecfc134684f4fd8bbdf114a72fb2033 : SUCCESS in 31m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a5a5b99b3f84fa4ae621fc0053839e6 : SUCCESS in 58m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2608dedfae5f49f7a38cb6505762b253 : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":2},{"id":"2d714884d81fa8e1caff18c901a96f60735a023d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 19:49:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bec5edabc11680d862f8e02f5e27a9ebfa52584a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 19:52:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512c06531a5f4bf898b939e4ae51a3b4 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b50ca0a1728306b8af2c73e85394305c1350443a","revisions":{"7ea2b754aeb8aa9f606d80d364fe65d376d90d98":{"kind":"REWORK","_number":1,"created":"2020-03-24 14:57:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/714667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/714667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/714667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/714667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/714667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/714667/1"}}},"commit":{"parents":[{"commit":"2d87de55aa51549de5746539a257f0f035affd0d","subject":"Merge \"CentOS 8: Enable hacluster and masakari images\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d87de55aa51549de5746539a257f0f035affd0d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-24 14:57:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-24 14:57:20.000000000","tz":0},"subject":"Bump service versions (Train)","message":"Bump service versions (Train)\n\nChange-Id: I8cac0ffa5c52b5b6060b8072d7d11c6e8bd9943b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ea2b754aeb8aa9f606d80d364fe65d376d90d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ea2b754aeb8aa9f606d80d364fe65d376d90d98"}]},"branch":"refs/heads/stable/train"},"b50ca0a1728306b8af2c73e85394305c1350443a":{"kind":"REWORK","_number":2,"created":"2020-03-24 15:01:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/714667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/714667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/714667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/714667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/714667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/714667/2"}}},"commit":{"parents":[{"commit":"2d87de55aa51549de5746539a257f0f035affd0d","subject":"Merge \"CentOS 8: Enable hacluster and masakari images\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d87de55aa51549de5746539a257f0f035affd0d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-24 14:57:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-24 15:01:52.000000000","tz":0},"subject":"Bump service versions (Train)","message":"Bump service versions (Train)\n\nChange-Id: I8cac0ffa5c52b5b6060b8072d7d11c6e8bd9943b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b50ca0a1728306b8af2c73e85394305c1350443a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b50ca0a1728306b8af2c73e85394305c1350443a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
