)]}'
{"id":"openstack%2Fkolla~714670","triplet_id":"openstack%2Fkolla~stable%2Fstein~I9c94d95287ef8956556de5d48b7e88b69457e60f","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I9c94d95287ef8956556de5d48b7e88b69457e60f","subject":"Bump service versions (Stein)","status":"MERGED","created":"2020-03-24 15:07:12.000000000","updated":"2020-03-24 18:08:52.000000000","submitted":"2020-03-24 18:07:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714670-1585073222879-795c03e7","meta_rev_id":"6a004c4316697218d7eaf023c1a36d500f699220","_number":714670,"virtual_id_number":714670,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-24 18:07:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-03-24 15:10:19.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-03-24 15:08:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-03-24 15:10:19.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 15:08:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-24 15:10:19.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-24 18:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d845f616cff61853e206fd77ec9dba2df7d9aaac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-24 15:07:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0d3279484980bf397cc2cbfa7a41e8739a663c3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-24 15:08:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c7d58f60704a2538553cf76158cdd65d7a3703e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-24 15:10:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6aa3d46be34a5a8b2b0d4dfa6faea2139ee0e063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 16:39:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/400244b26e7046e58a0e2a2570778649 : SUCCESS in 57m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/25f681f2b9174b19b4a16c42d719dfba : SUCCESS in 1h 01m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/628649d180f24d0d9bb4dc01b12570ad : SUCCESS in 1h 10m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1b05d6b021984732b1b145942fc1fd24 : SUCCESS in 1h 22m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a17127d215e04122a34c32765c501bc2 : SUCCESS in 1h 30m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/889a506f77a049dbb7a074bf5334a061 : SUCCESS in 59m 01s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/2241db2285604526bd8ee19b34f1a469 : FAILURE in 1h 05m 20s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/19f4ba26d72445b89c994a524b4a3b33 : SUCCESS in 1h 08m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fecf3d49c05b485b877ae86015ad5922 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/734db3b595404bd590c9dca7fda38477 : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/341e21b8cc5f4610a69ad936f500d5d3 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7298395b0a64a2682dabd18df984a0e : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12ba8175e5b6492f8b60f5c83fa3c83d : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/114ed711320744d18e86ec6bbbf79348 : SUCCESS in 5m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7ff19b5ecd884a7185163873177940b4 : SUCCESS in 27m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf4dce922de24f58b029018e4fb1c28b : SUCCESS in 33m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf03e5c712a3417191d86d98926e33e7 : SUCCESS in 56m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec70a025d73d420e8a0211135a392edf : FAILURE in 38m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f2680222f011439ab83776f94d350447 : SUCCESS in 1h 06m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b0515e6bc1a7bde9beadaa384c54af73b76dbc17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 16:40:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"823fa14614764845083a0920ed1c92417aa51245","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 18:07:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c22f21b2a6de574177c544d200c324c93fa19007","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 18:07:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2c9c0e12d070450ca45a14d8831f9bb3 : SUCCESS in 1h 00m 24s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/53b5b832a97e4b6083b46dc166b05b93 : SUCCESS in 45m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6902373efbe4926b2bb9988b15b5efe : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f734dfc9177460c897ea21239f21c37 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/301349db0cab436e801c160d5a1db90f : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a39803f1b094e30905b92bfc6dbe5a1 : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d3466c43300412ab571663efd0ebf51 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e015a2afccf4782979884ba80d65df0 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d33f6f2877124cd4b6c02e9e2eecb1fe : SUCCESS in 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6a004c4316697218d7eaf023c1a36d500f699220","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 18:08:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ba58fd0018482d9c5b0effd4e509b9 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58a8e81897a12e95f1b75c058e57afafd705b828","revisions":{"58a8e81897a12e95f1b75c058e57afafd705b828":{"kind":"REWORK","_number":1,"created":"2020-03-24 15:07:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/70/714670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/714670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/714670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/714670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/714670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/714670/1"}}},"commit":{"parents":[{"commit":"abd8b9401f7f0ab711a12656049d07f6a0d46d17","subject":"Merge \"mariadb: install mariabackup on Debian\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abd8b9401f7f0ab711a12656049d07f6a0d46d17"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-24 15:07:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-24 15:07:04.000000000","tz":0},"subject":"Bump service versions (Stein)","message":"Bump service versions (Stein)\n\nChange-Id: I9c94d95287ef8956556de5d48b7e88b69457e60f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58a8e81897a12e95f1b75c058e57afafd705b828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58a8e81897a12e95f1b75c058e57afafd705b828"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
