)]}'
{"id":"openstack%2Fkolla~714944","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I3e12ab51e9f01d383268d75ce72903686564ada6","project":"openstack/kolla","branch":"stable/train","topic":"bp/centos-rhel-8-stable/train","hashtags":[],"change_id":"I3e12ab51e9f01d383268d75ce72903686564ada6","subject":"CentOS 8: Drop trickle install to build freezer","status":"MERGED","created":"2020-03-25 13:39:09.000000000","updated":"2020-03-25 17:33:36.000000000","submitted":"2020-03-25 17:30:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714944-1585157417964-fe09d30a","meta_rev_id":"a831c41e420fac0ff96e773cff1b7f55c2f39444","_number":714944,"virtual_id_number":714944,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 17:30:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2020-03-25 15:56:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2020-03-25 15:56:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-25 15:56:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-25 17:30:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eff3088c1a6dff72046c5b380d78ea6882b31b20","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-25 13:39:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"97caa60ca498fc9ddce239e64e033a70ca5d657a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 15:14:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0337e8a0be4f44d5b31d38283a14305f : SUCCESS in 38m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e0a083b60989473c92576f50111d09a0 : SUCCESS in 56m 05s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cdd0cedc76c0467e921ea793cc59af21 : SUCCESS in 26m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2baf7d131dc94c0e95136a7ecac31e1e : SUCCESS in 32m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d20a466d699f41af8cfab954b4357329 : SUCCESS in 1h 18m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4532bccf40274b19909ba9417166d2f8 : SUCCESS in 1h 33m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/81665005517d4deabee0d16807f66607 : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e5504f0c8f4641b0b1971faff0c86927 : FAILURE in 49m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3836245185894fc096f5f5fee5685514 : SUCCESS in 1h 25m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f8b1a03fee2f4476929a720c919e6300 : SUCCESS in 55m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/87bb756fb5524420a0882b4fdc6f2f39 : SUCCESS in 29m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/71257bbd0e9b47eab07a3a86e2d52d80 : SUCCESS in 19m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2854a9749ee240879beb92b42e58f1f4 : SUCCESS in 1h 00m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b3fbcc5993a403abb4a92a6f3574702 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3efde43a8e314819a909f331f15ea9f6 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/271bb82d9b9a4a49842c63e60057ec7a : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d5449f9704245cda76583194504a82f : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06f0608e9f0f4b9e9c2ec1760094dbc4 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/719beb6edc024684b5786873f0225580 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/364279beb42b4b92b6b5f348a663ad1a : SUCCESS in 5m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5cd74afcc5614beea54a2aed0226f13f : SUCCESS in 28m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c63bb54980240c2b5ddaae1304bd8cf : SUCCESS in 33m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5dfa7060842f42eaa60b467fdbeb7b43 : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b56cf858d9174b948c5ae29ccc062175 : SUCCESS in 1h 00m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/35612cbcda3d4d6d9a21cd67c39d6dee : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d08d4b2c6e864d6c5d9fd6a177c8cfac8002e8b9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-25 15:56:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e79afd74419a185c74655f6ba6aac388d0503657","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-25 15:56:30.000000000","message":"Patch Set 1: Workflow+1\n\nclean one","accounts_in_message":[],"_revision_number":1},{"id":"c79c84c71e47b84b2695c5981045e4173eada88f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 15:56:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c4ba7d18c58fec386384f980c1868e869d7eaf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 17:30:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d43da0c0c96d4825b46c28ce8e0f1015 : SUCCESS in 57m 08s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ed0e398cb4d74eb39ee83f86348c8746 : SUCCESS in 32m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0b79bdde3fd74ae99df2ebaf10b6f619 : SUCCESS in 1h 26m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9067f0ba6db146c1be557d976c9352f6 : SUCCESS in 1h 05m 01s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3a1cb891700e426f8c4eb82bc46b6a8b : SUCCESS in 1h 28m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/08dc29f3499a448b9819ffb32f39968c : SUCCESS in 57m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccff83f2049641c0aa5eaa606ae621d9 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f31619c4309f490b9cdf50116d6ed1e0 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/06cce809bfda47c68caf60c9577fd3cb : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df68bf92f6dc4d8d90af1ae6f92ac7b2 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68b09a84c6dc485988c5f43eb54c338f : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb7688406c84b269d76d616642c5233 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14d287cc9bec488a88982a2bfdf79395 : SUCCESS in 5m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d7d139622964aa29ecca6e6c9b17032 : SUCCESS in 28m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0f0f3cf7cce4cd3a4dbf7bb0d5a5ce6 : SUCCESS in 55m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e46e483a633546e087d48f9b982dce80 : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c0678c1ac72e863d0a4cb771b2f095bc5cb7c258","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 17:30:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8eb12c2bfe8286594f1dddd23df83eadb1e1e2fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 17:33:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b1ab1e3d054a2f9dd4b531a5eb27f3 : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62d8c7cba22b481abbc2e2e3ea49e673 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d5a8fcd73e2b177609be64840757848f57a39cd","revisions":{"9d5a8fcd73e2b177609be64840757848f57a39cd":{"kind":"REWORK","_number":1,"created":"2020-03-25 13:39:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/714944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/714944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/714944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/714944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/714944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/714944/1"}}},"commit":{"parents":[{"commit":"052a9433659347a35d4f81b6a07bb83d97a17ad2","subject":"CentOS 8: Build bifrost image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/052a9433659347a35d4f81b6a07bb83d97a17ad2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 15:07:26.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-25 13:39:09.000000000","tz":0},"subject":"CentOS 8: Drop trickle install to build freezer","message":"CentOS 8: Drop trickle install to build freezer\n\nTrickle was previously installed from EPEL7, but is not available in\nEPEL8. It does not appear critical to the functioning of freezer, so\ndrop it to allow freezer images to build for CentOS 8.\n\nChange-Id: I3e12ab51e9f01d383268d75ce72903686564ada6\nPartially-Implements: blueprint centos-rhel-8\n(cherry picked from commit 548144fc3c2564c9c44a3a42372a1f3b8f2fef34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d5a8fcd73e2b177609be64840757848f57a39cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d5a8fcd73e2b177609be64840757848f57a39cd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
