)]}'
{"id":"openstack%2Fkolla~715051","triplet_id":"openstack%2Fkolla~master~Ia53657a415249882b93acca86b5e2a3bfa167820","project":"openstack/kolla","branch":"master","topic":"bug/1869072","hashtags":[],"change_id":"Ia53657a415249882b93acca86b5e2a3bfa167820","subject":"Fix glance-api privsep errors","status":"MERGED","created":"2020-03-25 20:02:15.000000000","updated":"2020-04-15 14:30:04.000000000","submitted":"2020-04-15 14:28:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715051-1586960887314-69063b05","meta_rev_id":"b701f05aee1c6dd639e9dd022ef06b5defac641f","_number":715051,"virtual_id_number":715051,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-04-03 17:13:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 14:28:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-03 17:44:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-04-06 05:28:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2020-04-15 12:33:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-14 13:58:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2020-04-15 12:33:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-04-03 17:39:23.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 17:39:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-03 17:44:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-06 05:28:44.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2020-04-14 13:58:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-15 12:33:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-15 14:28:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed2a06ba43689307e265845adc7d7936b722ea1a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-25 20:02:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6f2650468586e12cc7e9d725000215779c59002","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-25 20:02:51.000000000","message":"Patch Set 1:\n\ntesting https://review.opendev.org/699910","accounts_in_message":[],"_revision_number":1},{"id":"715af967e448461d97da526b8a4bd5a2f6333d07","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-25 20:03:38.000000000","message":"Patch Set 1:\n\noopsie: https://review.opendev.org/714999","accounts_in_message":[],"_revision_number":1},{"id":"e4b247918cbb5b6712d6fe4704f788a500213e94","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-25 20:30:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/c3c7a5edbe814b828a4c7aefba5b66bf : FAILURE in 7m 54s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/f06210eaa8954a898f9a46f5958c4126 : SUCCESS in 27m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7582358c611af9cb0a182dbb0e6a3a32c47a1cb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 20:53:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d904333bf22f408ca75b84d2ea7cab78 : SUCCESS in 49m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7edcdc042487f8df8bb555aa461528840b8b339f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 22:15:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7a38f387dfd74841abbbf791c37e3e8f : SUCCESS in 49m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7c3e28cd1d744e018675fc4a6df7135d : SUCCESS in 29m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9b43f9c55bdb47e5be44b2df3d81d4c6 : SUCCESS in 1h 17m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/51813e5281a74baab12e7737536bc482 : FAILURE in 57m 10s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1fa3e192e0f149339782963bd4d46427 : SUCCESS in 1h 23m 16s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/9a27715ee3e64bfcb7c8495de250e3da : SUCCESS in 40m 45s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c4b1eadbd36943768dc6b1d83b500084 : SUCCESS in 1h 53m 35s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/5ba8c29dfd2f46c78a765a1c2edc2aab : SUCCESS in 2h 10m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4dab3205dd5b4e6989bf37db0991288e : SUCCESS in 24m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/243f7151dfdc4f37a378882e6f31d970 : SUCCESS in 25m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6035a0fc4b5d4d66bb73c78a5e6910b2 : SUCCESS in 1h 00m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bb223e6872a40258accc4cd2afc76bc : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae7d44302c8f412d80ed8a5c436f55ee : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c4751d524d04d588e3ace3429dd0115 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfa534891d8548d7813aea6369d38e74 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca1df3da82bd4a22b92b42718e339f3e : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c464b16a9f419b900eb3ba137eca55 : SUCCESS in 7m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e2caf25c397848b9a485a7793df564ee : SUCCESS in 23m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7e06dc03b3c4f4ea8948aaa59c4d36a : SUCCESS in 30m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc7eed612e1e494e992595eaef9b0e1b : SUCCESS in 1h 08m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9339c950cdaf419d9a1b98f3dc6b03aa : FAILURE in 1h 02m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0465790d6df745db84bc240a8406fb4a : SUCCESS in 1h 08m 20s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5f3e4378c8fa4e7194fd67db07fd3f67 : SUCCESS in 1h 33m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ab8a2145cbc84f6f8e7ad4c3ff4942ca : SUCCESS in 1h 50m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1748278afe9909b0f910bd37f0ed8c5f2a6062c6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-28 11:13:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3a80531a0fe7ae99c95d474ea5d4607778822372","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-28 11:45:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/329a84a1cd534e09a25bc48886f07dfc : SUCCESS in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"186881e0271ecc3d6fb475a848e2d61d973d07a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 12:01:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8e31e288835b4de6ab83e7004c4e4e01 : SUCCESS in 47m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"06390dbaf1edfc3755b9e2481dfd32a2e722ac9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-28 14:31:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d2529ef4faf64d68be315f23c7b53f79 : SUCCESS in 23m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e28744a75bf445c69e3dc75ab14e972c : SUCCESS in 40m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f6dbb084b1a54181b29fa2650f4e9dfc : SUCCESS in 1h 13m 41s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/35fa88fa99f8417b8feda80bf0c1ff89 : SUCCESS in 1h 16m 22s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/624286a0d2c04fb4aeb6bc4a349042ee : SUCCESS in 1h 16m 24s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e09289b9db2f4e6388c970857e87e871 : SUCCESS in 41m 12s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/bd85fd511e99418dbda75bb02a98416b : SUCCESS in 1h 51m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8d8de7c06ae344668dfe56bc0760d994 : SUCCESS in 1h 58m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8aae338dc38743708e03d628acecd7a7 : SUCCESS in 32m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ff86f0de3f7343848024ae144c403a4d : SUCCESS in 16m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5bd64da9ee32485ba887e5b7a91152f2 : SUCCESS in 58m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9056754c651d4594aa767c360877f826 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/238385f1e9634b7fb5bfe4d47f84fc22 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/659ab663ac98435daa57f6f150583ff9 : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dcd899b295e34e48a8316baa129c04e2 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ae76f6dac93420c9269840b5b45ff76 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6473c7febe454689b70f820c6dc5f3d3 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c5659b8c24ce4508bbe259f3272e557a : SUCCESS in 33m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c233d2c28744c68abafba311258a1ff : SUCCESS in 22m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/47efc98ff52545c1aea7ec7623325707 : SUCCESS in 55m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/53d04ea3ef5e4bb783db4cdc3296be2e : SUCCESS in 1h 03m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5cd52106105a4f91b010076747727816 : SUCCESS in 1h 11m 35s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/53cd0c74f93747229defe1768838dafc : SUCCESS in 1h 32m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/07b3efb453c0404db660277cf5b51181 : SUCCESS in 1h 45m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b87d8a3ef38c82adb21ed2c96008438572f2a5d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-03 09:52:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c35cd977bce019d9a9831a5cbd59e80b960af45","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-03 10:08:18.000000000","message":"Patch Set 2:\n\nthanks for reminding me of this, will clean it up","accounts_in_message":[],"_revision_number":2},{"id":"4f83a8300bb4d92268c5eaf363f9db76de5dd7a1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-03 10:57:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee0eb0787b53467fd6b45a04c7fde2bd0853c79f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-03 11:26:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/5de4aa52cfc540289322aa807683ff28 : SUCCESS in 25m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7a522c75c4be314d5af884fc75a98cf9cf180455","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 11:44:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38551769cf484c8cbcfc3341e134b04f : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a00326d9d0b97c3d1779a5cfb8b72403ccce06d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 13:42:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b278d31ab3364a5a860fcc97b640100f : SUCCESS in 32m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9968a584603f40fab7a62bab4da6d4d6 : SUCCESS in 37m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ed0ac5c6ea464ca4b1af94838f29a6eb : SUCCESS in 1h 49m 45s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8a47baf4682040c1ae478a92278694d0 : SUCCESS in 1h 25m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fe281c42ce5b4907947758324866980c : SUCCESS in 1h 30m 07s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2ba0a97b605447a9ba2e75a8f0e3cde8 : SUCCESS in 49m 59s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/3fdde3b74db643e289f62c91459892c3 : SUCCESS in 2h 08m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ed84b0762b944f478b37fc3b16799d87 : TIMED_OUT in 2h 41m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b11429b191454f9f886ff3be6ca277ab : SUCCESS in 1h 09m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b47aa5e515e8489cbbc006a1cb01a9a8 : SUCCESS in 20m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bb66917552af44b08fefabc2e166be44 : SUCCESS in 1h 04m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/561d3943d4174e1b8f8397fa2f7ee43d : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53fae2118f394aeb86f9f089f5343ca4 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfee92f735f14303963d29efd0b9be3f : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/368c4512e6c0416a8fb0fbb1c3fa40fa : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a15252a5a0cc4631b561f6d52f9585b0 : FAILURE in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d185ff02f3014874af79802b121af5c3 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6a5c76461e58448c88c0ca3178887bf1 : SUCCESS in 23m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b567a160f7dc415984682901aaf9f7fb : POST_FAILURE in 45m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3cb605eefa6402198e8f05b72137c6d : SUCCESS in 1h 03m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e98ba97dfae45a7815f7368fba738e4 : SUCCESS in 58m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/51bbac5ff2f142569bc915be89206d18 : SUCCESS in 1h 22m 16s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/36e01ae0e3664ac79ec9bc030ea05ffb : SUCCESS in 2h 14m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/71185d4715934f478e3e6c8201b7aac2 : SUCCESS in 1h 43m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1c2a4cff1d38ed16a20aeec079f020b66888213c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-03 17:12:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8146a34781e01658872f75be77e63e3b9230a475","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-03 17:13:35.000000000","message":"Patch Set 3:\n\nhmm, \"Dump host info to logs\" module failure on ubuntu where all built fine","accounts_in_message":[],"_revision_number":3},{"id":"eb1e805618c4893e1adbcc753d7d41c0566e61ee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-03 17:39:17.000000000","message":"Topic changed from wip-glance-sudoers to bug/1869072","accounts_in_message":[],"_revision_number":3},{"id":"480062a632453d045357769a7502fc9aa77893a8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-03 17:44:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/5540f2aa67c940fda9cd8cb3fbae0e81 : SUCCESS in 25m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"550a4d395699b84d3f86382a47be348110abbe4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 18:05:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cc2f29504794e0daaab88fb27f7c8cd : SUCCESS in 49m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"880bd20888b6447465915fa30f03ac151d4b2112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 19:44:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d4fccb20c64a4b1ca07490df3d3fcbfd : SUCCESS in 59m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3809bf72b70843ae8d5f25c907bff998 : SUCCESS in 42m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f2ced9392aa54c56acb92b37b1bf17e7 : SUCCESS in 1h 25m 19s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/82e811d91a8f40b5b185afa47410419a : SUCCESS in 1h 21m 10s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c7bdde81ef4e450595390e3ae0c14555 : SUCCESS in 1h 27m 53s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f58f0015d39043bfa963de3dca26f409 : SUCCESS in 40m 00s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/40b269f2b9024665a30e7b355394bca7 : SUCCESS in 1h 46m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f1cbd3c9c22f4525aa907527ce48326e : SUCCESS in 2h 25m 19s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/570f2dd155364129818040dc0fccb304 : SUCCESS in 21m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3e7e29e532ad48b58230c957d90c41f5 : SUCCESS in 20m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6f2a19b77fbd43dbb52875d7ba26053a : SUCCESS in 1h 02m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e16e4fe2ad447f0b2d21a843cd0d195 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/059f1bacc8314eca8bd8c9aa5a4bdb86 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8a032f48a2b42c386b73f297482107e : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0a4827e60cc4ede8bcf36bec52cc581 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b028472e7c646248082c00209ea74a1 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df090b9a25064b95be43dd77356959a9 : SUCCESS in 8m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/68400c4726924b50934f64b1842d4ef1 : SUCCESS in 36m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/08d3743a7dd844fd836611333555585c : SUCCESS in 37m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d0f42a0ce8c40958df86d24d96a22a5 : SUCCESS in 1h 01m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/39f515555c414f5995be16dbca55ce4e : SUCCESS in 1h 16m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c499b90f743147cd9026d16bd7ebc2e9 : SUCCESS in 1h 31m 02s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/14b0db6a21354e8eb0303eb01cc82a30 : SUCCESS in 1h 35m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9ba8607c4fb24e9c9af15d951cb757d4 : SUCCESS in 1h 46m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6c030d33dfa84c656e0b41e629d28c8d292a4c7e","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-04-06 05:28:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0c48241512f1303b9920c2dd3a9115a197c07882","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-14 13:58:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"158c5bd53475803ed82cd1f3d453b57d5afb0a24","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-15 12:33:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7612082f14af0a86f64c156681701d0b06c27bde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 12:33:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ce248947c50d5370b3a63ec92b7eff9dfa52f36e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 14:28:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d8ac2d378cfc49f0d9eea91fdb683789290504dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 14:28:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7249d8a12e224da0809098ebef438114 : SUCCESS in 28m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/57a1d85335d44bfb92d4f60e4dcf28d6 : SUCCESS in 1h 31m 10s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/614ae1b53dea4957a338a101e4c8f530 : SUCCESS in 1h 35m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d2be7e53c554150ae7c817595c9f582 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9cad8d38c5a406b83ef60b15b2af609 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/670bddd883164f62b58f619df89adcca : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a303981b322c4c8d8230f7f6d2151498 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db15706077b42d0b37407fd6a2d6fe6 : SUCCESS in 6m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1aa5a0c6fda4ee19c8eac341b406eab : SUCCESS in 26m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c77e036e4dc4849adc05d5b44401340 : SUCCESS in 1h 00m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/69b222ce96ca45f09f29c009956deb9b : SUCCESS in 1h 52m 04s","accounts_in_message":[],"_revision_number":3},{"id":"b701f05aee1c6dd639e9dd022ef06b5defac641f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 14:30:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56e137f28724167815b0e64047c5174 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d7d40d6c8695f79da5aad58c45231de7e9f738a3","revisions":{"ac25a5018033c50e97c359d4bbffb04981ee67e8":{"kind":"REWORK","_number":1,"created":"2020-03-25 20:02:15.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/51/715051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/715051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/715051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/715051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/715051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/715051/1"}}},"commit":{"parents":[{"commit":"fe8820be560c32832968d939ea7733ebd0848acd","subject":"Merge \"CentOS 8: enable monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe8820be560c32832968d939ea7733ebd0848acd"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-25 20:01:38.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-25 20:02:12.000000000","tz":60},"subject":"[WIP] glance sudoers","message":"[WIP] glance sudoers\n\nhttps://bugs.launchpad.net/kolla/+bug/1869072\n\nChange-Id: Ia53657a415249882b93acca86b5e2a3bfa167820\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac25a5018033c50e97c359d4bbffb04981ee67e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac25a5018033c50e97c359d4bbffb04981ee67e8"}]},"branch":"refs/heads/master"},"526459a9f609a1df6d4fcf9078d11c7d38fce842":{"kind":"REWORK","_number":2,"created":"2020-03-28 11:13:48.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/51/715051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/715051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/715051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/715051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/715051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/715051/2"}}},"commit":{"parents":[{"commit":"fe8820be560c32832968d939ea7733ebd0848acd","subject":"Merge \"CentOS 8: enable monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe8820be560c32832968d939ea7733ebd0848acd"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-25 20:01:38.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-28 11:05:22.000000000","tz":60},"subject":"[WIP] glance sudoers","message":"[WIP] glance sudoers\n\nhttps://bugs.launchpad.net/kolla/+bug/1869072\n\nChange-Id: Ia53657a415249882b93acca86b5e2a3bfa167820\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/526459a9f609a1df6d4fcf9078d11c7d38fce842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/526459a9f609a1df6d4fcf9078d11c7d38fce842"}]},"branch":"refs/heads/master"},"d7d40d6c8695f79da5aad58c45231de7e9f738a3":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2020-04-03 10:57:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/51/715051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/715051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/715051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/715051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/715051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/715051/3"}}},"commit":{"parents":[{"commit":"69f74554671a49c75bebeeed9d83816ffbd7efec","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69f74554671a49c75bebeeed9d83816ffbd7efec"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-25 20:01:38.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-03 10:57:35.000000000","tz":120},"subject":"Fix glance-api privsep errors","message":"Fix glance-api privsep errors\n\nThis fixes usage of cinder as glance backend.\n\nGlance API source containers were missing both sudoers and\nrootwrap config. rootwrap config gets included in Kolla\nbecause otherwise it is clunky to get it from glance_store\nwhere it lives. The positive side is that it has not\nchanged since July 2016.\n\nDebian/Ubuntu binary have weird double-glance path to\nrootwrap which gets fixed now by symbolic link.\n\nCloses-bug: #1869072\n\nChange-Id: Ia53657a415249882b93acca86b5e2a3bfa167820\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7d40d6c8695f79da5aad58c45231de7e9f738a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7d40d6c8695f79da5aad58c45231de7e9f738a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
