)]}'
{"id":"openstack%2Fkolla~715111","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I3ffb69650fec7288b17d5c758a9af3ce0d4fe74f","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I3ffb69650fec7288b17d5c758a9af3ce0d4fe74f","subject":"Deprecated shade client library","status":"ABANDONED","created":"2020-03-26 01:44:25.000000000","updated":"2020-03-26 09:58:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"739fe9872ca844dfb91d0d4af807e5f613605432","_number":715111,"virtual_id_number":715111,"owner":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-26 07:15:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"disliked":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},{"value":-2,"date":"2020-03-26 09:44:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":-1,"date":"2020-03-26 08:56:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"date":"2020-03-26 07:49:40.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-26 07:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-26 07:49:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-26 08:56:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-26 09:44:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"e1f5d1ce6aaa7ce4162b5ecc8c7b1d5307e43310","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-03-26 01:44:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a870fc5899db7704171195d09d4993fc66b11f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 03:04:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ed556776676c485f9463070ed250fa84 : FAILURE in 6m 29s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ad0cb7791f8c4e6ebb0a9642813ebfbd : FAILURE in 6m 22s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ca36e9c98c754a6088124cefdda7b893 : SUCCESS in 28m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/41215fed1d6543beba9835301f91e4ca : SUCCESS in 36m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ef8b7c0d16354bea9dc9a26d5740dc4f : FAILURE in 11m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5cec398c76d6411590147acd536bacff : SUCCESS in 1h 10m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2c68c61182b948ab96ed5d2337130724 : FAILURE in 10m 51s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a2b899f554004a09b683874a7d819539 : FAILURE in 51m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2708c246c08c4f9bb8872adf6ac16acd : FAILURE in 11m 13s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/435fc5a810c64d1496e2be1f8bdd2a39 : SUCCESS in 1h 17m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/63f30e90f2a54f5196b904356f53b8c0 : SUCCESS in 26m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5c6f0fe356ee4041895e970e1a3f231f : SUCCESS in 27m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c750fdb70215489da1fe6596a490273d : SUCCESS in 1h 03m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89a5e5e32b01440e8b0710647e100f8e : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f90300b5e61425185983d2e1d23a3b8 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6572b3701e1a446cbe4416f4698b4e40 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51e71118552f4350b558d93a77aec888 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9437c583af4e49fabd37cb3151a12df8 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd99bbaad6b47798414b098f1a5eacf : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1136457e76344bc2b6bf7807e2aeea87 : SUCCESS in 36m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcbf65761db943f3bdd4dbe56d717abf : SUCCESS in 27m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e68fb9636267434db8447373e01be1cd : SUCCESS in 1h 00m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/03d933ade3f14016bffc8cc9eac57be2 : SUCCESS in 59m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aa6300bcdad54065863eadc7f50a1cbc : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c2fc8c736c7be2cb691b17be4ba0325bc89264fd","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-03-26 03:12:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"35f0abdae984ac3a7979cd8e542d3944b835e899","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-03-26 03:12:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"22a0a8b749878b80bb7635b4aa7f2518b9759879","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-03-26 04:05:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b812efbfdb27dceabbd880871f8805f09e2fd49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 04:24:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a47ae51c4cfc4e8f98bfe817a7bf4d5d : FAILURE in 4m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0a8a62833b56464b8b672aba3960bfa2 : FAILURE in 4m 32s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c7f2f40c251447ca860db2c6210f05bb : FAILURE in 4m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0626afca3c6e49dfa788590237e4d5fa : FAILURE in 4m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7398dc9929e04512b25a149550b4b1f2 : FAILURE in 9m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bd427d5d20a64a3fa0b1df5082d9c972 : FAILURE in 10m 19s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/50381d33a9c640cda26b8ba415d5c122 : FAILURE in 9m 38s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/95149f56df57423dac37b200b8c03b55 : FAILURE in 9m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3467ec065936410e9b5aee94fa6910e0 : FAILURE in 9m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e61c7cd6d7504f9b9d123924684720ff : FAILURE in 14m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/195df1be0ad942b6bb8f8222e3ef7cb9 : FAILURE in 3m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3cb1fabaf9e943c18ef8df11971921af : FAILURE in 3m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/22d693787c914ee19631ee8f725fcfd3 : FAILURE in 7m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32b0300372c5472e93f64f1e7db4a512 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c81dfdfd7e448218bddfa6564bbb11d : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a377320502a4f118cffdc08900023d8 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9b53a5453c94ad8907f2b816af61745 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bde98b18eba45fdb2273cdf26af31a7 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/468235b03bcc4698835e9aece2d6b957 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c67e4d0c8d94ecfa5b0e1a11c798d1a : FAILURE in 3m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4c97c8a3f3f4a89b6e38af3750140bc : FAILURE in 3m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bdb9bc80483647b88f566b5a6c93bb08 : FAILURE in 6m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24f630a0857d4bd2ac6757acf928b967 : FAILURE in 7m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2efaaa2b433747fa874592f6d0034edc : FAILURE in 6m 54s","accounts_in_message":[],"_revision_number":2},{"id":"5eea2856c15cb5732f9ad31b10a0785675165414","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-03-26 04:36:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f26ce6e0218f9b0db37e2f75ba37063b0f521281","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-03-26 05:41:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e7718ae6f8731a1ac1b94b28fecdbf0acb884050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 07:15:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/43c5b7ea1dd0475692d57723692fd07b : FAILURE in 6m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e8236a009da943db9b5e9e0bf1aa40d6 : FAILURE in 6m 38s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1297c0e901e748acb2a7cb53c2c11d9a : SUCCESS in 31m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4e1148e4e1ab4af1b73001d6118a4948 : SUCCESS in 40m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d24517b94d604975b2424f4897df4cc1 : FAILURE in 6m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0453260829404782a8fb96aa48a98c22 : FAILURE in 46m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c7a347da419c4ba290d4ecb5b66d7f29 : FAILURE in 6m 53s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/89d997635fee40d7b2e00a32cbca9a39 : FAILURE in 57m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3b7e40f54e294740824d55700205a59c : FAILURE in 6m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b2c29a4b9d884b7f85fca87c0eabfcb2 : SUCCESS in 48m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/42e6ac61f687466cb2713ca4accfd3c4 : SUCCESS in 29m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/58c931083162467a9b7d4cab9c7d9e42 : SUCCESS in 17m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f31a2c9a45894f50a24892ea35c0773a : SUCCESS in 1h 01m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f17c1e241b554851a29ee82fecaf4316 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fc90fe000244bd7ac7e78661f072466 : SUCCESS in 8m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8aa0c6a1e4a54d7c80b013e28ffa820e : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0899b1ce2eef47d580d8b55a3964ff9c : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c14cc4d6b8ea413582554accacee9a2f : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7433f54f6a4b5ab40f209bdbd49f3f : SUCCESS in 9m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d511090363ce4dc58d24acfd650de459 : SUCCESS in 25m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/79f6a9176dcd4088b770db8183881b98 : SUCCESS in 27m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7fcaf4fba91c472789a96bde92d0c31d : SUCCESS in 56m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8eb4d28564f6420c95675fbf12f10a3a : SUCCESS in 58m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f0039e1e109948fd8be99b17e6969793 : FAILURE in 6m 20s","accounts_in_message":[],"_revision_number":4},{"id":"5ef016676b6bd7f5dbca3845c7d6cfc83ee80df2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-26 08:56:46.000000000","message":"Patch Set 4: Code-Review-1\n\nshould go via master","accounts_in_message":[],"_revision_number":4},{"id":"65e3c16b7f16fd90f8d323b17c5e51ae95fec55a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-26 09:44:39.000000000","message":"Patch Set 4: Code-Review-2\n\nThese modules were removed in master.","accounts_in_message":[],"_revision_number":4},{"id":"739fe9872ca844dfb91d0d4af807e5f613605432","author":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"date":"2020-03-26 09:58:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bca84875f199a34e79a57ba691e079d399372d95","revisions":{"54c9dc8b9062ce8472ea14cdf84be16251bbd640":{"kind":"REWORK","_number":1,"created":"2020-03-26 01:44:25.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/11/715111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/715111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/715111/1"}}},"commit":{"parents":[{"commit":"052a9433659347a35d4f81b6a07bb83d97a17ad2","subject":"CentOS 8: Build bifrost image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/052a9433659347a35d4f81b6a07bb83d97a17ad2"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-03-26 01:25:26.000000000","tz":-240},"committer":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-03-26 01:40:30.000000000","tz":-240},"subject":"Deprecated shade client library","message":"Deprecated shade client library\n\nThe kolla_keystone_service and kolla_keystone_user modules\nof kolla-toolbox use the shade client library,\nbut the shade has been superceded by openstacksdk and\nno longer takes new features, so the shade is deprecated\nand the openstackSDK is used instead.\n\nChange-Id: I3ffb69650fec7288b17d5c758a9af3ce0d4fe74f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54c9dc8b9062ce8472ea14cdf84be16251bbd640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54c9dc8b9062ce8472ea14cdf84be16251bbd640"}]},"branch":"refs/heads/stable/train"},"9fd207ba4518c7fe8a732282c0781e3de0b27a7e":{"kind":"REWORK","_number":2,"created":"2020-03-26 04:05:26.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/11/715111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/715111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/715111/2"}}},"commit":{"parents":[{"commit":"052a9433659347a35d4f81b6a07bb83d97a17ad2","subject":"CentOS 8: Build bifrost image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/052a9433659347a35d4f81b6a07bb83d97a17ad2"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-03-26 01:25:26.000000000","tz":-240},"committer":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-03-26 04:03:01.000000000","tz":-240},"subject":"Deprecated shade client library","message":"Deprecated shade client library\n\nThe kolla_keystone_service and kolla_keystone_user modules\nof kolla-toolbox use the shade client library,\nbut the shade has been superceded by openstacksdk and\nno longer takes new features, so the shade is deprecated\nand the openstackSDK is used instead.\n\nChange-Id: I3ffb69650fec7288b17d5c758a9af3ce0d4fe74f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fd207ba4518c7fe8a732282c0781e3de0b27a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fd207ba4518c7fe8a732282c0781e3de0b27a7e"}]},"branch":"refs/heads/stable/train"},"77cf8f199d1a5516ff57cac1122e1829f57b53d2":{"kind":"REWORK","_number":3,"created":"2020-03-26 04:36:31.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/11/715111/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/715111/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/715111/3"}}},"commit":{"parents":[{"commit":"052a9433659347a35d4f81b6a07bb83d97a17ad2","subject":"CentOS 8: Build bifrost image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/052a9433659347a35d4f81b6a07bb83d97a17ad2"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-03-26 01:25:26.000000000","tz":-240},"committer":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-03-26 04:29:42.000000000","tz":-240},"subject":"Deprecated shade client library","message":"Deprecated shade client library\n\nThe kolla_keystone_service and kolla_keystone_user modules\nof kolla-toolbox use the shade client library,\nbut the shade has been superceded by openstacksdk and\nno longer takes new features, so the shade is deprecated\nand the openstackSDK is used instead.\n\nChange-Id: I3ffb69650fec7288b17d5c758a9af3ce0d4fe74f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77cf8f199d1a5516ff57cac1122e1829f57b53d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77cf8f199d1a5516ff57cac1122e1829f57b53d2"}]},"branch":"refs/heads/stable/train"},"bca84875f199a34e79a57ba691e079d399372d95":{"kind":"REWORK","_number":4,"created":"2020-03-26 05:41:52.000000000","uploader":{"_account_id":31506,"name":"Lin PeiWen","email":"591171850@qq.com","username":"LinPeiWen"},"ref":"refs/changes/11/715111/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/715111/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/715111/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/715111/4"}}},"commit":{"parents":[{"commit":"052a9433659347a35d4f81b6a07bb83d97a17ad2","subject":"CentOS 8: Build bifrost image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/052a9433659347a35d4f81b6a07bb83d97a17ad2"}]}],"author":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-03-26 01:25:26.000000000","tz":-240},"committer":{"name":"LinPeiWen","email":"591171850@qq.com","date":"2020-03-26 05:36:51.000000000","tz":-240},"subject":"Deprecated shade client library","message":"Deprecated shade client library\n\nThe kolla_keystone_service and kolla_keystone_user modules\nof kolla-toolbox use the shade client library,\nbut the shade has been superceded by openstacksdk and\nno longer takes new features, so the shade is deprecated\nand the openstackSDK is used instead.\n\nChange-Id: I3ffb69650fec7288b17d5c758a9af3ce0d4fe74f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bca84875f199a34e79a57ba691e079d399372d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bca84875f199a34e79a57ba691e079d399372d95"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
