)]}'
{"id":"openstack%2Fkolla~715163","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ie46da709db0de09a9cf641b3c154275282f213fc","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1866827-stable/train-stable/stein","hashtags":[],"change_id":"Ie46da709db0de09a9cf641b3c154275282f213fc","subject":"Fix invalid test in cinder data migration script","status":"MERGED","created":"2020-03-26 10:54:51.000000000","updated":"2020-04-03 18:48:32.000000000","submitted":"2020-04-03 18:45:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715163-1585939513022-fe09d30a","meta_rev_id":"102fcf17693af419ac39f16719baa96ea70d9f66","_number":715163,"virtual_id_number":715163,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7923,"name":"Gauvain Pocentek","email":"gauvainpocentek@gmail.com","username":"gpocentek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 18:45:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":7923,"name":"Gauvain Pocentek","email":"gauvainpocentek@gmail.com","username":"gpocentek"},{"value":2,"date":"2020-04-03 15:54:27.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-03 15:53:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":7923,"name":"Gauvain Pocentek","email":"gauvainpocentek@gmail.com","username":"gpocentek"},{"value":1,"date":"2020-04-03 15:54:27.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7923,"name":"Gauvain Pocentek","email":"gauvainpocentek@gmail.com","username":"gpocentek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7923,"name":"Gauvain Pocentek","email":"gauvainpocentek@gmail.com","username":"gpocentek"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7923,"name":"Gauvain Pocentek","email":"gauvainpocentek@gmail.com","username":"gpocentek"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-26 10:54:51.000000000","updated_by":{"_account_id":7923,"name":"Gauvain Pocentek","email":"gauvainpocentek@gmail.com","username":"gpocentek"},"reviewer":{"_account_id":7923,"name":"Gauvain Pocentek","email":"gauvainpocentek@gmail.com","username":"gpocentek"},"state":"REVIEWER"},{"updated":"2020-03-26 11:01:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-03 15:54:27.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-03 18:45:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b12b807a142f0dce379d8001ef1a624ce24adc88","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-03-26 10:54:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"6573bcaea485a0d168af755524d1b4034841327a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-26 11:01:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0093e1ba5b935093cd095caaf691f25d99a44b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 12:03:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3f3820d4a9b54f06a7fc3ee986ac83b6 : FAILURE in 9m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/224dc164011c4f0a9e9f72a5e8117afd : FAILURE in 9m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e34d81a4c6764558b50931b85c7bb488 : FAILURE in 5m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9603195add8e4f92bac523aa966e4060 : FAILURE in 4m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e641aec8f72e411f88e59e25382ea1cb : FAILURE in 3m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5dfe4ebea8ff43eb8be8e9cbf79ba9db : SUCCESS in 52m 49s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/d366683dd9974c059d4a82cca60c543c : FAILURE in 6m 44s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/0c5001ef02564b68a3ed148b1991273b : FAILURE in 53m 24s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/551587b143684e8ab4b45e1000bf6162 : SUCCESS in 50m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e76c62285c55451a89b717937449538a : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/345f819999a84689a12f7405a119a21d : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5379e901992d47089f1c77aa9157ca4e : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb2bd282e1fe47d2a0e2c178d11cd980 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e01c2fb900dd49ba82056bd9122cb804 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb2080bc9e745f88f52ca37ed14dade : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33dfdd14e38d4b389834e69b2ac32ed7 : SUCCESS in 5m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f14262bf21394f539c393c12dd29d361 : SUCCESS in 24m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e517c19127244f0a9f6e26fa6aa50bba : SUCCESS in 27m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/78e57603b76744d68064af213abdad7b : FAILURE in 3m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f8be473befd43f4b6c9150d2b271fb1 : FAILURE in 4m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dd833c6b750543cc86bda531bc3ecd5a : SUCCESS in 1h 07m 06s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/2b4468d9ea2d45bda19259ef277361f8 : FAILURE in 9m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f40f700f696e98056e99cfe327999be21c3c8f0","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-01 11:13:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"16ff21f0af6a317c90285878e1f085e3d195ac84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 13:21:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d590161d7090464c9523782d20b23e7c : FAILURE in 9m 40s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0ded95b8e58445e4b78519797abe9808 : FAILURE in 10m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5ec6f23b79ca4e7db4b0d8c89b7ad30d : FAILURE in 14m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0f3b119467024f16b89984f6a3ce49fe : FAILURE in 13m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/016a383f08e040279533ba22fc1cea54 : FAILURE in 12m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8b1a72eed5994b2997672281644af2f3 : SUCCESS in 54m 50s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/829a78d4e0a54157bfaef144bbb43870 : FAILURE in 16m 19s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/eb068e12b4e74557b7b656ece0777aef : FAILURE in 58m 50s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/cda5124985864cd983e9c8f5e4f20181 : SUCCESS in 52m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47463cfd4bd3465ab4d9a33bd6f9cd49 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29545b49b03481c82660ccfbd87b467 : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/563a34a699144260938e81960ec9d033 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e53843cfde74538918ae08a74b0bb77 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fd08313c5034321b9220f7f3767ceaf : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/617036ebe3014cc4bedda06ee2fba72d : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26bf7623068c4f75a05ca54753ecccd0 : SUCCESS in 4m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a9b98858ec9247d0a204143f94189ef3 : SUCCESS in 23m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1dd721c944c446482f0f73245e274b3 : SUCCESS in 26m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c6492bdbfe064916bc370942a3a1f874 : FAILURE in 39m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c09bf85daf2c44c4895e22e6fa37512c : FAILURE in 33m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/28748a0572be4b48b0ca8f7f8ace0304 : FAILURE in 40m 29s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/b4a8b40b217d494098ba78e3ebd481f2 : FAILURE in 2h 01m 15s (non-voting)\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"bdd0965d8122adfdfe9821b38e59dd71ebbcfe13","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-03 15:53:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fd85bdac5f8cf0a7a5aff94993c5f4b1b38447f7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-03 15:54:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"7e785f7175b4d2ae2423870ec88652d41aa86495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 17:24:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8e27966375e04aa2864196568f8cda8f : SUCCESS in 49m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7350b454aedc4bb1aa25bdb22c101826 : SUCCESS in 43m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d685b470da784e1f889a3013e2b72e3d : FAILURE in 56m 29s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dd6a7eef62ba40b09a5c0e07f320307f : FAILURE in 57m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f6de5a67226f46068fee6042598a0708 : FAILURE in 56m 33s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a48001de834e4da695e3e413cd14a831 : SUCCESS in 1h 06m 52s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/680d0394a0ea4ec480dfda5b4ec17f1f : FAILURE in 1h 16m 33s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9dcd0ed5551a45279a74e800c0eb8feb : FAILURE in 47m 59s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/b07785ee3c734dc5b295efe8186cf703 : SUCCESS in 1h 20m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c88cdf6a1ae5464192324cd159836357 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea58ba6179fe47d5a0bfae959ec37d42 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dde4f0d787c64d108961a16a42860cf1 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6db5e936e86f405da920ab9b4b16c09e : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8e6bd5dc5b6435fb3cb1d7c34c59ff7 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a49ff64e2d5c43e0ab69f1a6f0c1c289 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c2da143c9354f2eb260c78a198a8a68 : SUCCESS in 7m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/09b7e5f129984ec49baa4476795cb8be : FAILURE in 10m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae4f0e3afbd34d5e80cff454e673e946 : SUCCESS in 27m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f12b09b0511409aaf08db563460658d : FAILURE in 51m 57s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed8dbb9ce5d74b3f862d00c570b59806 : FAILURE in 35m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/244c13de984640ccb0ddceeba876a39d : FAILURE in 43m 04s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/ce19924d4c1640f0b861d1b5d69bf4d8 : FAILURE in 1h 07m 19s (non-voting)\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"91c727a7b1d26cc93751de1274596d60813eaebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 17:24:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3d342e5b2638909c43800dce92a2dff8febc68ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 18:45:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3a214d3d6ca244ef8d47211b4a8bfafa : SUCCESS in 1h 09m 29s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/68c101ea92434ec2b94fa976c0ac4cea : SUCCESS in 1h 19m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eff0e2968301431e82af4b0380dc980b : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4447e185b5db4dcc86eef1875827b412 : SUCCESS in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2b6b1977f1c4891bac6f47d82095d1f : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/887b9d4e18134b48b8e3f8eb92e90e41 : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9769470f5b8424083251e2b2a0a45b2 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f358488aa5f740a3be2542aa99bc5b2b : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/541dfaad12fc400496c42cf890c6c8ea : SUCCESS in 7m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9fa88af98834217aafd5c2ce33d386a : SUCCESS in 32m 32s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"0b8e2ed2a739c42e70721f0ace717f3ecb0dedba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 18:45:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4d04b3aaa0555dc98b0cdb6e7c17ff5aff0b6a9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 18:48:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ee1f221f464248b58d1bab4c37d8d6 : SUCCESS in 2m 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e367bf96d3404968aed90b890035bfcd : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4696fc0f322bdc4327b0ac4ec58f2ddc9d014ec1","revisions":{"72943ce579e5b268e566382f02a4693da2491a90":{"kind":"REWORK","_number":1,"created":"2020-03-26 10:54:51.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/63/715163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/715163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/715163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/715163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/715163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/715163/1"}}},"commit":{"parents":[{"commit":"58a8e81897a12e95f1b75c058e57afafd705b828","subject":"Bump service versions (Stein)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58a8e81897a12e95f1b75c058e57afafd705b828"}]}],"author":{"name":"Gauvain Pocentek","email":"gauvainpocentek@gmail.com","date":"2020-03-10 11:46:17.000000000","tz":60},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-03-26 10:54:51.000000000","tz":0},"subject":"Fix invalid test in cinder data migration script","message":"Fix invalid test in cinder data migration script\n\nThe extend_start.sh script for cinder-api uses an incorrect test to\ncheck if the MAX_NUMBER variable is set. This leads to a incorrect\ncinder-manage command being run if this variable is not set, making the\ndatabase migration fail.\n\nThis change fixes the test to properly use the MAX_NUMBER variable.\n\nChange-Id: Ie46da709db0de09a9cf641b3c154275282f213fc\nCloses-Bug: #1866827\n(cherry picked from commit ff42ff20048ef3b7d2f8df04ada8ff4bf7b710e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72943ce579e5b268e566382f02a4693da2491a90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72943ce579e5b268e566382f02a4693da2491a90"}]},"branch":"refs/heads/stable/stein"},"4696fc0f322bdc4327b0ac4ec58f2ddc9d014ec1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-03 15:53:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/715163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/715163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/715163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/715163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/715163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/715163/2"}}},"commit":{"parents":[{"commit":"519eced6147085151a3e53a55c734c28d1e35eeb","subject":"Fix multiple CI issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/519eced6147085151a3e53a55c734c28d1e35eeb"}]}],"author":{"name":"Gauvain Pocentek","email":"gauvainpocentek@gmail.com","date":"2020-03-10 11:46:17.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-03 15:53:35.000000000","tz":0},"subject":"Fix invalid test in cinder data migration script","message":"Fix invalid test in cinder data migration script\n\nThe extend_start.sh script for cinder-api uses an incorrect test to\ncheck if the MAX_NUMBER variable is set. This leads to a incorrect\ncinder-manage command being run if this variable is not set, making the\ndatabase migration fail.\n\nThis change fixes the test to properly use the MAX_NUMBER variable.\n\nChange-Id: Ie46da709db0de09a9cf641b3c154275282f213fc\nCloses-Bug: #1866827\n(cherry picked from commit ff42ff20048ef3b7d2f8df04ada8ff4bf7b710e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4696fc0f322bdc4327b0ac4ec58f2ddc9d014ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4696fc0f322bdc4327b0ac4ec58f2ddc9d014ec1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
