)]}'
{"id":"openstack%2Fkolla~715202","triplet_id":"openstack%2Fkolla~master~I5264746fd04276448f4668c3bb3a218a40e8660b","project":"openstack/kolla","branch":"master","topic":"hrw/ansible-setup","hashtags":[],"change_id":"I5264746fd04276448f4668c3bb3a218a40e8660b","subject":"CI: Ansiblify setup of builders","status":"MERGED","created":"2020-03-26 13:25:45.000000000","updated":"2020-04-04 16:36:31.000000000","submitted":"2020-04-04 16:31:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715202-1586017888880-73754ddb","meta_rev_id":"90df640f90464289f611386fdb895ebc1a544f59","_number":715202,"virtual_id_number":715202,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-04 16:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-04-03 14:00:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-03 14:00:06.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-04-03 16:04:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 09:52:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-31 17:27:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-03 16:04:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-04 16:31:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5838538dc2c42ce2648298871c4f0a257fc38665","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-26 13:25:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfc24db270eb8a72bf4a54efa0a69ad8fea23ce4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 14:06:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bec4d6545a1aa222e7d426fe6c09832f65e0d038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 15:26:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/aed36d93fcc64bbf917ded6c0cd3a63d : SUCCESS in 24m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a8db4e77bfb143c9b89076a51b192c08 : SUCCESS in 58m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/efdb53c852044d27ae7280b506134ec0 : SUCCESS in 1h 50m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b4e5086560c94fa780840951b1bb58df : FAILURE in 55m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/975ecdcd9eca4a5eb339a2cf18fd8c47 : SUCCESS in 1h 27m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bc89eef626924a76940b3767e6a2268a : SUCCESS in 33m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/93b733e31b9241b9886fb2c0d5574eca : SUCCESS in 33m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cd04253cd94c44a88567fc49171ca36e : SUCCESS in 1h 12m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/133f2eb40a61498e9d9f8edc81c8f090 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad2aa6242af44b78fd83907ae8445ca : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/281ad3863ff84842b25f67f20e315d44 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17a6ed7c1e3748108cebfe4605680858 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3af389120b0d4f9d83559d4f35ea8485 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d07c9e285546e187263a2f3f4b58f1 : SUCCESS in 7m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c896d05a46ed47b08f74626c599fb8d0 : SUCCESS in 24m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b40fd6f4f8a4a1790699673be9baacd : SUCCESS in 38m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4504fbb02f8a4ec7876baadb21e9ee3e : SUCCESS in 1h 04m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d5e5361b7ce345deb058e01def77e0d0 : SUCCESS in 1h 04m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dcc56f6ae547442cb114390883a9bb21 : FAILURE in 8m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a35f2b098ba1567802d9044dabda76d3790a8f7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 15:27:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/aed36d93fcc64bbf917ded6c0cd3a63d : SUCCESS in 24m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a8db4e77bfb143c9b89076a51b192c08 : SUCCESS in 58m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/efdb53c852044d27ae7280b506134ec0 : SUCCESS in 1h 50m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b4e5086560c94fa780840951b1bb58df : FAILURE in 55m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/975ecdcd9eca4a5eb339a2cf18fd8c47 : SUCCESS in 1h 27m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bc89eef626924a76940b3767e6a2268a : SUCCESS in 33m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/93b733e31b9241b9886fb2c0d5574eca : SUCCESS in 33m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cd04253cd94c44a88567fc49171ca36e : SUCCESS in 1h 12m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/133f2eb40a61498e9d9f8edc81c8f090 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad2aa6242af44b78fd83907ae8445ca : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/281ad3863ff84842b25f67f20e315d44 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17a6ed7c1e3748108cebfe4605680858 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3af389120b0d4f9d83559d4f35ea8485 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d07c9e285546e187263a2f3f4b58f1 : SUCCESS in 7m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c896d05a46ed47b08f74626c599fb8d0 : SUCCESS in 24m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b40fd6f4f8a4a1790699673be9baacd : SUCCESS in 38m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4504fbb02f8a4ec7876baadb21e9ee3e : SUCCESS in 1h 04m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d5e5361b7ce345deb058e01def77e0d0 : SUCCESS in 1h 04m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dcc56f6ae547442cb114390883a9bb21 : FAILURE in 8m 02s","accounts_in_message":[],"_revision_number":1},{"id":"14c008f6eca7a504d3c8775e667cbe22caca2540","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-29 14:46:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"06b9e116f05f7db5e79487dbaa5a8113ebd7fbdc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-29 14:47:56.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"9d9b91056d1eaf8bd6a0a9cc893c6c1bb10cccda","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-29 14:51:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"42118db9b9a7e5503f26a7a4ae095558f7eca8eb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-29 14:54:33.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"82aa35c8f82585f24b6801344ecbacd72f1995bc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-29 14:54:38.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"29c2b2a0018237b32a53fac162c9923a00f832c8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-29 15:07:04.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"e5352d711703c221e4b158f4095ed09cc33109e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-29 15:14:57.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"10159a7698d655a474d5b4b54e72e62dbc739838","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 15:22:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fd2b8ee3b65b69a889b611f51a1822677adcbe6b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-29 15:46:06.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"9de9e2fee72b5d42023f539019530c16994c5c03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 16:32:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/674ca7015da2449a85453d25b11ac9e9 : SUCCESS in 45m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f3a3d437516fede4f6ac2edf337d74e59b60896e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 17:15:25.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ed19127d4b2d4d56a0670f9e4956194b : SUCCESS in 22m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b1102a4029c24376a3a3a590b7a7d24d : SUCCESS in 28m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fdff27bd5faf456696fa27439d326a96 : SUCCESS in 1h 24m 25s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2010706dbc4943b6928643234993ea9d : SUCCESS in 1h 13m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/030022d4e9c04ac5895dc060f90cba7a : SUCCESS in 1h 28m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/874f7b09e21e4e92b8d2daa154a947d2 : FAILURE in 3m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9655cf91b14a414d90768df21bc81c4c : SUCCESS in 19m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5fc6e591d4a9443b850be34fc9e858ce : SUCCESS in 56m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/347ee3ca0f554bf3b0f27a5b6b7b6471 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed34572890a4937a367dad16a69acf2 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6117f37e40a4e9eb8267f3aaf45f51d : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4911a86b731b4483b20b465c80f3ff67 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c3e2561f7ad42df85e21e1a33f0da60 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db66ebc0adf84ca59007d3bc5bfedf75 : SUCCESS in 7m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/32c436e222d84fe5a5378816a4d785bc : SUCCESS in 32m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3fba3a3c65784574980b8b6814d59e62 : SUCCESS in 26m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/69f4824b22b14c638e187513d2b990f3 : SUCCESS in 55m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/33b0b9878c8947f3a69327ba515bd72a : SUCCESS in 1h 02m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/29fdd3df9d5e49ae9f2d82662cf6a6b4 : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":8},{"id":"f498ae065651c603b2ef0adcfa6b18a100b1d869","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-03-30 09:49:08.000000000","message":"Patch Set 8: Code-Review+1\n\nLooks good, I\u0027m not a Zuul expert - but we moved some actions from run.yml to pre.yml - does it make any difference?","accounts_in_message":[],"_revision_number":8},{"id":"0b4e85ca2785e2eeb2bfc44ed0ab03bdd88d3170","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-30 09:50:33.000000000","message":"Patch Set 8:\n\n\u003e Looks good, I\u0027m not a Zuul expert - but we moved some actions from\n \u003e run.yml to pre.yml - does it make any difference?\n\nretries for free; otoh, we need to care more about idempotency there - looks reasonably idempotent to me (obviously :-) )","accounts_in_message":[],"_revision_number":8},{"id":"d9e62ed21fa7759810d3592591f634e9a56da8f7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-03-30 09:52:47.000000000","message":"Patch Set 8: Code-Review+2\n\nI like things for free - looks idempotent ;-)","accounts_in_message":[],"_revision_number":8},{"id":"8744de93ae4526ff96a8b516b88cdff15c5e024e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-30 10:16:50.000000000","message":"Patch Set 8:\n\n\u003e Looks good, I\u0027m not a Zuul expert - but we moved some actions from\n \u003e run.yml to pre.yml - does it make any difference?\n\nDoes not make any. I moved them as they are preparing environment to run tests rather to being part of test itself.","accounts_in_message":[],"_revision_number":8},{"id":"2d82bf0e19258ae7f6305b031ff1d8d49203f77d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-31 17:27:59.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\ndue to core shortage and this being mostly hrw\u0027s change in fact","accounts_in_message":[],"_revision_number":8},{"id":"acccbd56be9e1295dc616f60cb40f6a50bae7659","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-03 14:00:06.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"e82c954a80ce6f41e8d14fb88952d7bd14a66b3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:47:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03bf5d9c45254089b32d0a92ed0ee7c0 : SUCCESS in 45m 52s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ed52882e78f6cca55a75845ba1002fe5e2b0beb5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-03 16:04:19.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"e5483dd53c71848ff830bc2186b752c7258afda2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:17:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c4f0079f2094478d87f8fda214ca3156 : SUCCESS in 28m 15s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/383bd33fccf14e40be31720733b5dd52 : SUCCESS in 1h 56m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7eb1c11ca9c14fb9bfa8c0a318519859 : SUCCESS in 1h 16m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/064340662a7345b5b93a795174e590d3 : TIMED_OUT in 2h 05m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bc1a2af15d1b41f38924ba3e5b11f28c : SUCCESS in 1h 28m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/37511c9a4a68433ca85f2f6de5375ae8 : SUCCESS in 21m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/860427ab3bc84983856105b8c24633af : SUCCESS in 1h 43m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a0adbd0428bb47b1943a33621efaf173 : SUCCESS in 58m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49df3141d9e1451dacb8ed013cb66929 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df2466149ed74b298ab72647bf141543 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e85ce4ce6fa4b3b858cfbeadda86073 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c76ea08a11a7465ca4af2dd67cca1ed5 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f64ba6caa8f44e01ae976b434d198b96 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d903981acfee4bc3b46a5444a83ae2c4 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c64aed25c11041fcac9694635104b290 : SUCCESS in 25m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c4d3e39824f41d1913db62da0b6b543 : SUCCESS in 27m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bac66750ab1a45919a6aa4afdce0a2aa : SUCCESS in 1h 59m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/339d38c464b142d6a324569e3a5f506d : SUCCESS in 59m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/03a01bc98876401cadf663d9d9575960 : SUCCESS in 2h 14m 31s","accounts_in_message":[],"_revision_number":9},{"id":"e8aa90d7e73554a7190505257a8c1b9d3f8c1107","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 14:21:07.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"ead541586004a8779331bea969b9a3d07bae48a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 16:31:28.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5de148d13d134c91828082893d71c957 : SUCCESS in 26m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/23e9c7a9a9ba4386857ee16c03405810 : SUCCESS in 1h 14m 57s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6162aa271d0740ada4ba2be69fad40ec : SUCCESS in 1h 20m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d119214e5d64420b6d1873e0d1b690f : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b8bd30811514f899d70b50d8c3d11b9 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d155f2e0e834886a0cbc3ad322f2b73 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2651c4384b04e809295df75e075264f : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878a003991f646169705eb79bd5120ce : SUCCESS in 5m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/73de708bb2e04511a258e103c0651663 : SUCCESS in 23m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a569a5f6a1214c3c9784bf720d7ee121 : SUCCESS in 51m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ebfe379b4e5484b940a15240128113b : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":9},{"id":"21a7914c18af920f3b8feefad0f972a2fb025707","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 16:31:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"b819074a5a4174c4971c7e12af7613d23cfdf1b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 16:36:31.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eded206ba03b412fb95cfda7db1cdc40 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"2cdcf1821d95133a4992d36db514d981ff225e31","revisions":{"dfe382dd6fca68dcf9df470e0e7e13badb9a0aca":{"kind":"REWORK","_number":1,"created":"2020-03-26 13:25:45.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/02/715202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/715202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/715202/1"}}},"commit":{"parents":[{"commit":"69f74554671a49c75bebeeed9d83816ffbd7efec","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69f74554671a49c75bebeeed9d83816ffbd7efec"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-15 14:36:47.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-26 13:22:59.000000000","tz":60},"subject":"CI: Ansiblify setup of builders","message":"CI: Ansiblify setup of builders\n\nCI is setup using Ansible playbooks. So let do everything that way.\nWithout using shell scripts on a way.\n\nChange-Id: I5264746fd04276448f4668c3bb3a218a40e8660b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfe382dd6fca68dcf9df470e0e7e13badb9a0aca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfe382dd6fca68dcf9df470e0e7e13badb9a0aca"}]},"branch":"refs/heads/master"},"6da54eb34270759e855ff309d0f39dd2bd870d01":{"kind":"REWORK","_number":2,"created":"2020-03-29 14:46:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/715202/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/715202/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/715202/2"}}},"commit":{"parents":[{"commit":"69f74554671a49c75bebeeed9d83816ffbd7efec","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69f74554671a49c75bebeeed9d83816ffbd7efec"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-15 14:36:47.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-29 14:46:08.000000000","tz":0},"subject":"CI: Ansiblify setup of builders","message":"CI: Ansiblify setup of builders\n\nCI is set up using Ansible playbooks now.\nThis also drops redundant shell scripts.\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nChange-Id: I5264746fd04276448f4668c3bb3a218a40e8660b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6da54eb34270759e855ff309d0f39dd2bd870d01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6da54eb34270759e855ff309d0f39dd2bd870d01"}]},"branch":"refs/heads/master"},"b061e381a11e60b01185ffa7d731339b7a0e32fc":{"kind":"REWORK","_number":3,"created":"2020-03-29 14:47:56.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/715202/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/715202/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/715202/3"}}},"commit":{"parents":[{"commit":"69f74554671a49c75bebeeed9d83816ffbd7efec","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69f74554671a49c75bebeeed9d83816ffbd7efec"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-15 14:36:47.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-29 14:47:54.000000000","tz":0},"subject":"CI: Ansiblify setup of builders","message":"CI: Ansiblify setup of builders\n\nCI is set up using Ansible playbooks now.\nThis also drops redundant shell scripts.\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nChange-Id: I5264746fd04276448f4668c3bb3a218a40e8660b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b061e381a11e60b01185ffa7d731339b7a0e32fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b061e381a11e60b01185ffa7d731339b7a0e32fc"}]},"branch":"refs/heads/master"},"6d49ad951e60a7d5a448217df8bf44c0e746574d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-29 14:51:27.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/715202/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/715202/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/715202/4"}}},"commit":{"parents":[{"commit":"4b3abf5dc4f8cd037966dfd49da56ec67c9a0b47","subject":"CI: Avoid tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b3abf5dc4f8cd037966dfd49da56ec67c9a0b47"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-15 14:36:47.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-29 14:51:27.000000000","tz":0},"subject":"CI: Ansiblify setup of builders","message":"CI: Ansiblify setup of builders\n\nCI is set up using Ansible playbooks now.\nThis also drops redundant shell scripts.\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nChange-Id: I5264746fd04276448f4668c3bb3a218a40e8660b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d49ad951e60a7d5a448217df8bf44c0e746574d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d49ad951e60a7d5a448217df8bf44c0e746574d"}]},"branch":"refs/heads/master"},"b28d54b9f3b862659e5553a6a34b5e522e95de04":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-03-29 14:54:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/715202/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/715202/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/715202/5"}}},"commit":{"parents":[{"commit":"69f74554671a49c75bebeeed9d83816ffbd7efec","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69f74554671a49c75bebeeed9d83816ffbd7efec"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-15 14:36:47.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-29 14:54:33.000000000","tz":0},"subject":"CI: Ansiblify setup of builders","message":"CI: Ansiblify setup of builders\n\nCI is set up using Ansible playbooks now.\nThis also drops redundant shell scripts.\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nChange-Id: I5264746fd04276448f4668c3bb3a218a40e8660b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b28d54b9f3b862659e5553a6a34b5e522e95de04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b28d54b9f3b862659e5553a6a34b5e522e95de04"}]},"branch":"refs/heads/master"},"e7057f47c8047f6a2fa57d594a99f2b30c939ad1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-03-29 15:07:04.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/715202/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/715202/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/715202/6"}}},"commit":{"parents":[{"commit":"cb3acb876c048c6cf657e021e5e11f143a1b480b","subject":"CI: Avoid tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb3acb876c048c6cf657e021e5e11f143a1b480b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-15 14:36:47.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-29 15:07:04.000000000","tz":0},"subject":"CI: Ansiblify setup of builders","message":"CI: Ansiblify setup of builders\n\nCI is set up using Ansible playbooks now.\nThis also drops redundant shell scripts.\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nChange-Id: I5264746fd04276448f4668c3bb3a218a40e8660b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7057f47c8047f6a2fa57d594a99f2b30c939ad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7057f47c8047f6a2fa57d594a99f2b30c939ad1"}]},"branch":"refs/heads/master"},"8da8ca0af7ac2bb06d39b391e6f23c99d32ae9e1":{"kind":"REWORK","_number":7,"created":"2020-03-29 15:14:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/715202/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/715202/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/715202/7"}}},"commit":{"parents":[{"commit":"cb3acb876c048c6cf657e021e5e11f143a1b480b","subject":"CI: Avoid tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb3acb876c048c6cf657e021e5e11f143a1b480b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-15 14:36:47.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-29 15:14:52.000000000","tz":0},"subject":"CI: Ansiblify setup of builders","message":"CI: Ansiblify setup of builders\n\nCI is set up using Ansible playbooks now.\nThis also drops redundant shell scripts.\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nChange-Id: I5264746fd04276448f4668c3bb3a218a40e8660b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8da8ca0af7ac2bb06d39b391e6f23c99d32ae9e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8da8ca0af7ac2bb06d39b391e6f23c99d32ae9e1"}]},"branch":"refs/heads/master"},"35ea453027450821f09cc9d9372129c76f02541f":{"kind":"REWORK","_number":8,"created":"2020-03-29 15:46:06.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/715202/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/715202/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/715202/8"}}},"commit":{"parents":[{"commit":"cb3acb876c048c6cf657e021e5e11f143a1b480b","subject":"CI: Avoid tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb3acb876c048c6cf657e021e5e11f143a1b480b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-15 14:36:47.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-29 15:45:58.000000000","tz":0},"subject":"CI: Ansiblify setup of builders","message":"CI: Ansiblify setup of builders\n\nCI is set up using Ansible playbooks now.\nThis also drops redundant shell scripts.\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nChange-Id: I5264746fd04276448f4668c3bb3a218a40e8660b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35ea453027450821f09cc9d9372129c76f02541f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35ea453027450821f09cc9d9372129c76f02541f"}]},"branch":"refs/heads/master"},"2cdcf1821d95133a4992d36db514d981ff225e31":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-04-03 14:00:06.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/02/715202/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/715202/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/715202/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/715202/9"}}},"commit":{"parents":[{"commit":"c089eab392de39ff01f9eba65ef4855a14e2076a","subject":"CI: Avoid tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c089eab392de39ff01f9eba65ef4855a14e2076a"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-15 14:36:47.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-03 13:59:20.000000000","tz":120},"subject":"CI: Ansiblify setup of builders","message":"CI: Ansiblify setup of builders\n\nCI is set up using Ansible playbooks now.\nThis also drops redundant shell scripts.\n\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n\nChange-Id: I5264746fd04276448f4668c3bb3a218a40e8660b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cdcf1821d95133a4992d36db514d981ff225e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2cdcf1821d95133a4992d36db514d981ff225e31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
