)]}'
{"id":"openstack%2Fkolla~716339","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I2765fe5c4a749aa9249b30a9d235d11cf635d331","project":"openstack/kolla","branch":"stable/train","topic":"716339","hashtags":[],"change_id":"I2765fe5c4a749aa9249b30a9d235d11cf635d331","subject":"Bump up magnum and manila versions","status":"MERGED","created":"2020-03-31 15:21:54.000000000","updated":"2020-04-07 11:09:39.000000000","submitted":"2020-04-07 11:04:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716339-1586257491623-e3c9830e","meta_rev_id":"3391fbeeb2eacba554577807c222366e8e868dc8","_number":716339,"virtual_id_number":716339,"owner":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-04-07 07:01:46.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 11:04:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-07 07:28:25.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":2,"date":"2020-04-03 12:56:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-03 12:56:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-04 10:25:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-04-03 12:56:01.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 15:24:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-01 11:36:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-01 15:16:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-04 10:25:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-07 07:01:46.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-07 07:28:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-07 11:04:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f27ce47c75bb6a6860326acee5957df60869db42","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-31 15:21:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5be9f746b8bd088aa2aa64dcaa12a10cf0e40046","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-31 15:24:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"feadf748860bb67922b376dc2704f35635a9bf61","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-31 15:25:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7a80cebdc084ae49c97f90e08e21220657a95873","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-03-31 15:26:14.000000000","message":"Patch Set 2:\n\nDone.","accounts_in_message":[],"_revision_number":2},{"id":"db900cf1033341d410ed57b250d152e7412acb35","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-31 15:52:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/3db05b99c95241d9b5a38dbc341db635 : SUCCESS in 26m 49s","accounts_in_message":[],"_revision_number":2},{"id":"a0e381803ce9bec6c37c56c5716f66ff6a27a9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:54:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2dfe4f9a77a54205a24b6b19313453ef : FAILURE in 9m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9f8dcbb202334c5492a33184a6de1f08 : FAILURE in 9m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d170b024788a4db19b77142669bad490 : SUCCESS in 27m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/24d16f367a7a484aaa39a7cc20f41459 : SUCCESS in 36m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8ac58522efab4372bd3092e81ad4759f : FAILURE in 11m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3365992409f44917bb95cc391081e796 : SUCCESS in 1h 08m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/961329e0c4be495abe8976e71f3e539b : FAILURE in 15m 56s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/68ab0445c4af41139ecdb338f2a9a613 : FAILURE in 1h 21m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e4974b95ae6b43dd96b50c6a88ecca45 : FAILURE in 13m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/eb18c0909c7d47679c9ef979de258f12 : SUCCESS in 1h 03m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cfa06876dd564a49b76b8dde5a223a2c : FAILURE in 5m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c2ffdebc17ab46879c664e40a53ae4e8 : SUCCESS in 23m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/21e15cc4ded344d7ba08ab4647b75eee : SUCCESS in 57m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca4bbf19c1d748469906c3cf5e0ef5d7 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ebea51df9584de99d91d012153ee5de : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/80e4e8596e83430e94c24c3d1ab4e30c : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97b27a671fc54329a09ff68694277c5f : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48587972eb364a30bbb286e7fbff82cc : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb756ce6fbea4158a4effeaa5db013a4 : SUCCESS in 10m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b7c91711af8c44f5af2cdebe07d2b482 : SUCCESS in 1h 03m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a03bf80f90e4abb9b97946eb65e275d : SUCCESS in 40m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/654b535809094a78a5ddaa6b48adf253 : SUCCESS in 1h 00m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/84a335fbdc9645f8affb7758fb7dca36 : SUCCESS in 59m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/81fe035cdce8420784e4cf4ed71071b1 : FAILURE in 41m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a369faf353e76b65751f358021c00a8390b70d26","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-01 07:33:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"084c9c341e80158576557794e518f6644182951c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-01 08:04:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/774a2f6dd0214ce8b86465a9fb14b078 : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":2},{"id":"8114574e53e52f094da3aea019678506f0ad0ba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 09:24:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c493983b4eb14876bada13765438b925 : FAILURE in 7m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7338c9e1d63a434a8aee7bbbab77b13b : FAILURE in 8m 43s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/10d4483dce5a4c2fbd87ecd1868bb5d3 : SUCCESS in 32m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2fc56d4b47d4465f9ee8b38de6891995 : SUCCESS in 33m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9f0d1161f5ac49c6be49325cf2a0abc1 : FAILURE in 13m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3bf282dcdf304616a94a6daa24631a49 : SUCCESS in 1h 16m 04s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8fa6b9fe391743afa9cf40b05da06702 : FAILURE in 11m 30s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2127565c01bb47c784da7335d27f86d1 : FAILURE in 54m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b8bcbae95bb64062ad2bc453aac2da25 : FAILURE in 11m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/93e186ff4bfd4c3f9c3595f17b4c0926 : SUCCESS in 57m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/88f323d2d27644c29d0f54aede8d710f : SUCCESS in 25m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f10e9d90e1a46a393b2797058468296 : SUCCESS in 19m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6574ccedf39546e0abecb5b4d02d6c46 : SUCCESS in 57m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c86cb68cf2e641c3b149e797ca9048d7 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a98d7641e148108af556391e050d0b : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d759aa539770455db25c3386a91011a2 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6835a08ea6eb42d28d040a92f568c9f7 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96472f0be61c40dda0a6fbc487116b71 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/567b68f32b1b4f638bab1bdbdde79af7 : SUCCESS in 7m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5398c37f2e7e4bde85fdf9c07201d987 : SUCCESS in 1h 16m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f5c92f72d5a4304a51df4ab6e46d567 : SUCCESS in 1h 35m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cb991b826af14cfcb4353a95ee49e5dd : SUCCESS in 1h 44m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4427b8b0f929405e97c89303df31114d : SUCCESS in 1h 48m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/64b083ee4bec44dd8cb2c9b0c2d5c13b : FAILURE in 1h 31m 06s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"fc1f403d9929f582182e46539d3d860282973851","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-01 10:48:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2bfc01ba756053cfe9d50eccf2d17c87609e12fc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-01 11:18:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/4eed0ae9ee2d4fe3b32834a16174a890 : SUCCESS in 27m 04s","accounts_in_message":[],"_revision_number":2},{"id":"ca4ebd4cd99382688acf1fd9baf83eb74111431b","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-01 11:36:52.000000000","message":"Patch Set 2:\n\nwaiting for https://review.opendev.org/716548 to be backported to train first.","accounts_in_message":[],"_revision_number":2},{"id":"ac91864dc6ecb3a224dc5c821dff4beb347f7cc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 13:12:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a873a27d3aeb419f9afb6144d61968a3 : FAILURE in 10m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/05c06d4e4fe04e96bfeb3de16bc029b2 : FAILURE in 9m 22s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8bdfc18dfcaf44a093460947f17bc177 : SUCCESS in 1h 02m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/43dcd762cfd6461181b506769d2f7863 : SUCCESS in 32m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/042e534a9d244d629920728274b84259 : FAILURE in 17m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9343fae36eae4d12a1263a0cd1227078 : TIMED_OUT in 2h 01m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5fbb01e444954d9e8f0e5de80780a84e : FAILURE in 16m 27s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3b0cee7eb5c5456ba4a82ba6c743cdaa : FAILURE in 1h 53m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/144efaf82d8e45f7bca4e84cc1cfe68a : FAILURE in 18m 46s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4b49b5c621af4a53909cd9c59bc0432d : SUCCESS in 1h 55m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/713cee65c0de4f6faefb848a7d986e81 : SUCCESS in 44m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8d097c3e95474a608939ba01a5345020 : SUCCESS in 1h 33m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d46b4f325e5a4d9e945077a2887c4241 : SUCCESS in 56m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5caf3591dd0e4b76a4ffa69ba3d53407 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645fda472c5443bf8c5865df0caeb3eb : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4980f9740564021a4f74fddaa556831 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89e87414237a44eb81a04f554072ae6f : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b3a884035b04c8cb2de8004e0a3ddf4 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/284b0f3a5a7e49bc833072fd63ba49d6 : SUCCESS in 8m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b0b57a8faad438db21fe6ceb2e291c1 : SUCCESS in 46m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0972ef41b23a43a39c4ab2f686312d62 : SUCCESS in 2h 17m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de1a25e8cbd84a88be7ec1f9880a3482 : SUCCESS in 1h 02m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a0486d9943d84f0aacdd5229393ac5b7 : SUCCESS in 55m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8e0350deb8e745ae866f8bb95eba61d6 : FAILURE in 36m 49s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"3f512cb762b02f065b2a56f0b45d44f372ed83a2","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-01 14:54:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"f1a3b479875d85f9148b549abae9515d1cf71b62","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-01 15:12:20.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"443066fc008b1687ed5862ff55ce227ae087300f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-01 15:16:12.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c64185f01503bcbe69f9494ffe1a7363744c99cc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-01 15:40:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/e9eeca9d4dec47a981247c269d6cf1b7 : SUCCESS in 26m 51s","accounts_in_message":[],"_revision_number":4},{"id":"cf2e0b5ffc9df365912b9d42431bb90b8879e416","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-01 16:10:37.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"ccb80f66e23b2e1af078a0958cc86c26daf4ced3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-01 16:13:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"fbe95ce7a0c5b430abb38ed031fb47a5e549f68c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-01 16:39:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/58c388007e9c44b69687cf7f1f5c6d4e : SUCCESS in 27m 20s","accounts_in_message":[],"_revision_number":5},{"id":"c66cfcbf6b05a338d337ab47debd903ab20baa05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 17:48:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ecbbd10ad88c491a9cdd074d0f932c60 : SUCCESS in 1h 21m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3faf1d3ac1a44f21ba48d6fa111000cc : SUCCESS in 1h 25m 46s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/dc6cd2ea3a6d471e901a740c03ccf960 : SUCCESS in 53m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f870faf32ebb453aa3b3dc729ebc707a : SUCCESS in 1h 03m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/67bee1bdf837406ca30f7e7036234c5c : SUCCESS in 1h 32m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3b1e14dde20e4e0695833b32885e9b4e : SUCCESS in 1h 11m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dc0dbc77b1da4dcf8b7844cfa29ec7d4 : SUCCESS in 1h 31m 32s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cecf0579e4b542a78e88d91fe7bff537 : FAILURE in 59m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ddb835f3c2504d528e0d70f7e6b068a7 : SUCCESS in 1h 27m 51s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/aab28bec9a024da1ab39633ea83ca3fc : SUCCESS in 55m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ebbda2caf9ec41ee8a984fe876e88d95 : SUCCESS in 25m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/361d1038b63940b381be0cf97168446f : SUCCESS in 21m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/863e91c46e56446991af33b10425742d : SUCCESS in 1h 02m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fac2d5ba6c9498c8d3e0f13b3d07a8e : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/250ec79472fc4132a63e587dcdc8a1b9 : SUCCESS in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa4770f3bf7845b3b175e50c8b5b5b91 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b97d0c2b46b409d8f4ee9fc1a514f5d : SUCCESS in 7m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b62aea5f229d4c41b9346a5b7be04715 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ef1a1a5edf4139a4902515b2849e40 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c01417f23eca48339e974f35528aeee3 : SUCCESS in 28m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a5f8b139aa44672883a62cf7e1d9354 : SUCCESS in 29m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d137c038b7bc4c779b60f47b2f0f93ba : SUCCESS in 1h 03m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/59986d83c08246bf8ca2f3468ab90d63 : SUCCESS in 1h 12m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/99cd725e7d62407e931e35d75cd888d6 : FAILURE in 58m 59s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"6ff3c81662d7f3790380ded5a9984e3279b95ef3","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-03 11:11:24.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3620fbde5f006f3a00e1ffbaa17abe4d86748686","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-03 11:39:48.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/2ae6c434b91d458b8a5375f25d087a06 : SUCCESS in 27m 21s","accounts_in_message":[],"_revision_number":6},{"id":"ac6205e389a701f90e3422da2b11f032297ecb55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 12:52:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7f07fb98607a49d798215cf3af493e61 : SUCCESS in 49m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/18d4626d047341dc9cb0725902b1602a : SUCCESS in 57m 49s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4dcd6e6a62fe4343b6ec143e55be58ef : SUCCESS in 1h 01m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2d781735e35d403bbdc78d19a50cf401 : SUCCESS in 1h 21m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2e562982dbe74a87a58fed3c986d072e : SUCCESS in 1h 23m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7432af0ea9314eb6bb7a1949f9f1b607 : SUCCESS in 1h 13m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/868ae23759d046fd89ba5fcf9d8f2287 : SUCCESS in 1h 14m 37s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d61ac4038cf4436b9c37b5a879a4a332 : FAILURE in 47m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/09c79a3b4ab247a5bced78ec2bbdb01d : SUCCESS in 1h 35m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/92ccbb34ca8c43c5802e54512f821382 : SUCCESS in 1h 02m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0182a993ff7145f3874ffdbbca1804de : SUCCESS in 47m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/69a3138173d24c4e8986fba069ceb1c3 : SUCCESS in 22m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a190a732bc774073b7d6764b0e728a77 : SUCCESS in 1h 00m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/716178c82d5e4c05b9545bf36d43cfa9 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48feea31e5bc402a8c89340791628633 : SUCCESS in 7m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6a04cd96c1294d12b4a17c9b29509f6f : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cb18e28feed4ea8a6f56ceeb2fb2394 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af1fbefef4da4c56861a21f8c057a1db : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ebddcec0da1494a9cbdd385a50285d6 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a6e458d3c2e426c8f9a7ca62da77b49 : SUCCESS in 48m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f33a033aa7943f58b132f00a6795148 : SUCCESS in 30m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e01cca24da354cd196ab703e7c63657d : SUCCESS in 1h 01m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/02524ec266704875a1cb57b90e7a3c49 : SUCCESS in 57m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c81ae5a55c7f4a21aca9af1d463ebaa2 : FAILURE in 41m 58s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"909ed432bbeb877930b90547004730cd73d2e548","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-04-03 12:56:01.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"fceb18f6e41c0b20d764118222fcb584fb0ab502","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-03 13:25:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/ca3c62519c4e4300879b6afd5c2c0f98 : SUCCESS in 28m 12s","accounts_in_message":[],"_revision_number":7},{"id":"b58281b75ae183fa96e685bf253cdcb5536a113e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 15:35:38.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ba29c04987b14f79ab003b9a51081b86 : SUCCESS in 51m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0423f2aac4894187976a54872eae217c : SUCCESS in 1h 03m 59s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a04b3037b6534f2bb144c6794423e322 : SUCCESS in 1h 25m 26s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cac4ffdf9ef64481b01948fb48306d96 : SUCCESS in 2h 06m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5b28bbbc7671441faa834aa5d19aab8f : SUCCESS in 1h 52m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1c0e3203cb434e6ba58dd59b7b9efdf9 : TIMED_OUT in 2h 04m 19s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d20c32a5d0ec4dc18015542e1d38ea67 : SUCCESS in 1h 53m 51s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1aaddaef9a7c47e0a2fb8627872da7c0 : FAILURE in 1h 59m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e8054259b4af4e79bc06ef168519599c : TIMED_OUT in 2h 33m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/33314b84052a4d83a4e1aed7595a0804 : TIMED_OUT in 2h 05m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e4a2e65366b741d5aa0f2d2086281e35 : SUCCESS in 2h 25m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/140a7841fb3943afa666508bd4d6c6f2 : SUCCESS in 34m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a07e35ca881f47d588e8534805c1ddab : TIMED_OUT in 2h 03m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f9c52709592445887ee132cc6959198 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca342df4ed6406fa3d35e1255296ffe : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5cc1d464e02548d68adb2288d6b83f69 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb9a57333930462a8e83929980280dc0 : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e4496402d9e4ba1867b18c612368042 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4815184a5e1a447d8634c9f7701662fa : SUCCESS in 10m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c2038a6c305b4611988738b3621fe477 : SUCCESS in 2h 33m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e608f9ee833d415cb4a7d1023c07381d : SUCCESS in 2h 33m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac65298b841740b1abb61f7b0f91bd20 : TIMED_OUT in 2h 03m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7cbde2eaa33c4098bfd2f51fc23d8c0f : TIMED_OUT in 2h 02m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/87152ee23d5d48fdab1f2efb86b2cafc : FAILURE in 1h 59m 18s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"1ca72c063c7d2a76aa9b0a1f073e8f927147c585","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-04 10:25:23.000000000","message":"Patch Set 7: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"163c4755897799912f6c8b435525f0cfd9e28ae5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-04 10:53:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/6a7276c3f01641828289dbe788c3fd41 : SUCCESS in 26m 30s","accounts_in_message":[],"_revision_number":7},{"id":"05070c615ab330cca53dd8302dcaec91aeb2e194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 12:09:41.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3d2939352dbd4a358b34a260c85c9803 : SUCCESS in 42m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/06ab1dc4a41249a4afebd0058c7cf715 : SUCCESS in 56m 22s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/60fff0aa390840938d0fe6d6edf1eb35 : SUCCESS in 33m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ae862e51207e40ef8a8def3982f1d76a : SUCCESS in 37m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f673536453b1451695c25ac12ba69c11 : SUCCESS in 1h 42m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/75e7f1a1c3d1485cb06719047f6393c9 : SUCCESS in 1h 13m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/53a6f20cf454476cb6ade53df9ad1db7 : SUCCESS in 1h 16m 55s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/119bc18b9465440986ac27dc86dbb46f : FAILURE in 56m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/67ba8773a266444b92d821221c7b577d : FAILURE in 1h 08m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4d84b17dc3734f679efac41ff16d7502 : SUCCESS in 55m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2e3e47ea099d45ffa76becc6a9c4c1c5 : SUCCESS in 46m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/63926171f41945ef9701711d5790b8c1 : SUCCESS in 18m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f87ad968afb34c7d9260ded83d4ff351 : SUCCESS in 1h 02m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83e43f9af64e40d6866a1db04dcf4305 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00d75df33382464089e2b1d741aa054a : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/194ce753d5844c9d93d547b5dc0141e2 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8950aaf644734395bcee6837e34a7831 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9cbe67076524e65a029e61d0a6bcb29 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d90b2928b164532b191cf730b295f7d : SUCCESS in 5m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de93c3551c0c4b028d98424962fc932c : SUCCESS in 30m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/816f3fa0f2ba4c71a1a53fa0b7c02921 : SUCCESS in 50m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/31e4e91b7de34fd695a4f1abc10efb50 : SUCCESS in 1h 03m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a491a81317624715943444801786ecff : SUCCESS in 59m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/13d7673354bc4d81b9e05107592c5b48 : FAILURE in 52m 00s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"2fabfc230e507e819844547fc6e374e58271c081","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-07 07:01:46.000000000","message":"Patch Set 7:\n\nrecheck\n\nzuul, we love you!","accounts_in_message":[],"_revision_number":7},{"id":"f24094027c3c303428c6adf2941efd2d064e7b36","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 07:28:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/e5616962b9ac4d64ae86e68bd31d40be : SUCCESS in 25m 34s","accounts_in_message":[],"_revision_number":7},{"id":"d3660e88fb4d4bd4563e88c7cb9587272e27ecf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 08:52:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bbddd3ee34e04415abf25146c75e1ff4 : SUCCESS in 44m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e2dc4508788c4bff9dc21424e630226d : SUCCESS in 56m 48s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/17e20e3f5576428d9011529a088ca54c : SUCCESS in 30m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/deaa1b129b0d41978104f949b756d269 : SUCCESS in 36m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/420f070d863c461ab5e0879a55d8b7de : SUCCESS in 1h 27m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9c5179ab587f448a8a2fcdba0f38f698 : SUCCESS in 1h 14m 40s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1c810109fde4415ebd1077d5d03e5f09 : SUCCESS in 1h 14m 28s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cb49ed6f33f3453a9ddb153c48a04216 : FAILURE in 56m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0ee337ae99ee4dc7b3d687f030df6243 : SUCCESS in 1h 45m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/53d58e7938094e98a9292734c06187dd : SUCCESS in 49m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2b747e875ef6474c8cf48e1a5bc236d1 : SUCCESS in 25m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3bb5f560407847d8aae85dc26867365d : SUCCESS in 39m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/70bae984a8354f6bb1e08ba6d771df06 : SUCCESS in 1h 01m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e90140c33e8341e09649bf76924253e6 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90c59db4f0bd4b5ab134b7eaed741962 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64f1e3bcd79d44b883547fbcc420547e : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb781cb66a624456ade66157c88e46d2 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1251ffd70ca54bc5bde04b967cff8238 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78d19382a0a144bf99376e1cf765038b : SUCCESS in 12m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b696014ed06499bb10f7b8e46e94086 : SUCCESS in 44m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/accefab0faf34b008aabf61a274ebf7f : SUCCESS in 41m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d5207292938249f9a71ce474d1a5e92e : SUCCESS in 1h 02m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ad39ad125544bde871208d603c12d65 : SUCCESS in 1h 00m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/641e098e086a4601a093d04fdcb8e3bb : SUCCESS in 1h 09m 11s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"cf33798adad4624328fb767e7d16c2f2ade30440","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 09:09:06.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"efcd9324f35b9889052c119f2ffc8f5016de4566","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 11:04:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"c196d658719f4c1a701394bf30eaee58b35703cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 11:04:51.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/71adc2010b9746e889c55c613f3a99c8 : SUCCESS in 52m 31s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f494076848e74e46914f8343a998d715 : SUCCESS in 33m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e87e0d24127a410a965fd35829b02947 : SUCCESS in 1h 21m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f453ed3641034c7d8ff9872bab1f6efc : SUCCESS in 1h 17m 12s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e20d98070ea848c1b08e50fceb342417 : SUCCESS in 1h 30m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/23c5fa5e9cfd4774b26d573a55576d27 : SUCCESS in 56m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e571113a290423e812de9fe266f464c : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df72510d508b4b05a35e395dde69dced : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/523763be6fba42b7b75dfbd3cdaf77d9 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8082c9ab99c44cf83e1b144e3a15ba3 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93cbdac816654957ababf63e8059861a : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd8116023df4c42adccaac2965543e1 : SUCCESS in 10m 55s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/97a850016a634d7f80ab2bcabba57fee : SUCCESS in 1h 01m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4256b45cc1084122a1af8ee0609e1567 : SUCCESS in 1h 02m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d7eceb466d04138a7cde8b18373ffab : SUCCESS in 1h 07m 41s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"881944650658da816cfa41dcfd075f68bd5665fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 11:09:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b8df9299774179857ae8b785021ce1 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2bf5340063ad9de10e5ce32c10fcba76d28bfa14","revisions":{"1b4c10b6a2ef7b580a2c7b3f03e4c3225969ce70":{"kind":"REWORK","_number":1,"created":"2020-03-31 15:21:54.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/39/716339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/716339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/716339/1"}}},"commit":{"parents":[{"commit":"b912f94c5d69203d5f53291c2c96fb67a5a9aab0","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b912f94c5d69203d5f53291c2c96fb67a5a9aab0"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-31 15:21:07.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-31 15:21:52.000000000","tz":0},"subject":"Bump up mistral-dashboard, magnum and manila versions","message":"Bump up mistral-dashboard, magnum and manila versions\n\nChange-Id: I2765fe5c4a749aa9249b30a9d235d11cf635d331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b4c10b6a2ef7b580a2c7b3f03e4c3225969ce70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b4c10b6a2ef7b580a2c7b3f03e4c3225969ce70"}]},"branch":"refs/heads/stable/train"},"f090114e2a29cfcaae3a9b329508aef4bf5aab68":{"kind":"REWORK","_number":2,"created":"2020-03-31 15:25:00.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/39/716339/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/716339/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/716339/2"}}},"commit":{"parents":[{"commit":"b912f94c5d69203d5f53291c2c96fb67a5a9aab0","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b912f94c5d69203d5f53291c2c96fb67a5a9aab0"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-31 15:21:07.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-31 15:24:29.000000000","tz":0},"subject":"Bump up magnum and manila versions","message":"Bump up magnum and manila versions\n\nChange-Id: I2765fe5c4a749aa9249b30a9d235d11cf635d331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f090114e2a29cfcaae3a9b329508aef4bf5aab68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f090114e2a29cfcaae3a9b329508aef4bf5aab68"}]},"branch":"refs/heads/stable/train"},"180cd8da4e6ff4569eaf3cfa34a0faf074c067d8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-01 14:54:28.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/39/716339/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/716339/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/716339/3"}}},"commit":{"parents":[{"commit":"20f978dc6f0df8859fed35875a9baf01e7730e1e","subject":"CI: Add http[s] match for mirror.centos.org to template-overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20f978dc6f0df8859fed35875a9baf01e7730e1e"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-31 15:21:07.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-04-01 14:54:05.000000000","tz":0},"subject":"Bump up magnum and manila versions","message":"Bump up magnum and manila versions\n\nChange-Id: I2765fe5c4a749aa9249b30a9d235d11cf635d331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/180cd8da4e6ff4569eaf3cfa34a0faf074c067d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/180cd8da4e6ff4569eaf3cfa34a0faf074c067d8"}]},"branch":"refs/heads/stable/train"},"d490f1e91cf2af25be585f7e2f8353d473255a60":{"kind":"NO_CHANGE","_number":4,"created":"2020-04-01 15:12:20.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/39/716339/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/716339/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/716339/4"}}},"commit":{"parents":[{"commit":"68059d50a62b034db68212a30980940413b6071f","subject":"CI: Add http[s] match for mirror.centos.org to template-overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68059d50a62b034db68212a30980940413b6071f"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-31 15:21:07.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-04-01 15:11:49.000000000","tz":0},"subject":"Bump up magnum and manila versions","message":"Bump up magnum and manila versions\n\nChange-Id: I2765fe5c4a749aa9249b30a9d235d11cf635d331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d490f1e91cf2af25be585f7e2f8353d473255a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d490f1e91cf2af25be585f7e2f8353d473255a60"}]},"branch":"refs/heads/stable/train"},"8c066906875d2dcea275c30959ac064713a7b4e9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-01 16:10:37.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/39/716339/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/716339/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/716339/5"}}},"commit":{"parents":[{"commit":"57b916c17621c540ca3f7cde16b9de0bdc97c666","subject":"CI: Add https match for mirror.centos.org to template-overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57b916c17621c540ca3f7cde16b9de0bdc97c666"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-31 15:21:07.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-04-01 16:10:37.000000000","tz":0},"subject":"Bump up magnum and manila versions","message":"Bump up magnum and manila versions\n\nChange-Id: I2765fe5c4a749aa9249b30a9d235d11cf635d331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c066906875d2dcea275c30959ac064713a7b4e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c066906875d2dcea275c30959ac064713a7b4e9"}]},"branch":"refs/heads/stable/train"},"d35e72c49af7c7af3944008745c5920047caa73a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-04-03 11:11:24.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/39/716339/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/716339/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/716339/6"}}},"commit":{"parents":[{"commit":"0b37d821e06f7bc6543c14e44433babb977ba3aa","subject":"CI: Add https match for mirror.centos.org to template-overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b37d821e06f7bc6543c14e44433babb977ba3aa"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-31 15:21:07.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-04-03 11:11:24.000000000","tz":0},"subject":"Bump up magnum and manila versions","message":"Bump up magnum and manila versions\n\nChange-Id: I2765fe5c4a749aa9249b30a9d235d11cf635d331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d35e72c49af7c7af3944008745c5920047caa73a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d35e72c49af7c7af3944008745c5920047caa73a"}]},"branch":"refs/heads/stable/train"},"2bf5340063ad9de10e5ce32c10fcba76d28bfa14":{"kind":"NO_CHANGE","_number":7,"created":"2020-04-03 12:56:01.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/39/716339/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/716339/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/716339/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/716339/7"}}},"commit":{"parents":[{"commit":"169b3aa7c3569194b33747403fee4535044885ff","subject":"CI: Add https match for mirror.centos.org to template-overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/169b3aa7c3569194b33747403fee4535044885ff"}]}],"author":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-03-31 15:21:07.000000000","tz":0},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-04-03 12:56:01.000000000","tz":0},"subject":"Bump up magnum and manila versions","message":"Bump up magnum and manila versions\n\nChange-Id: I2765fe5c4a749aa9249b30a9d235d11cf635d331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bf5340063ad9de10e5ce32c10fcba76d28bfa14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bf5340063ad9de10e5ce32c10fcba76d28bfa14"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
