)]}'
{"id":"openstack%2Fkolla~716691","triplet_id":"openstack%2Fkolla~master~I67e5ca24c550163c218fb06692de0b7c528d8318","project":"openstack/kolla","branch":"master","topic":"update-hacking","hashtags":[],"change_id":"I67e5ca24c550163c218fb06692de0b7c528d8318","subject":"Update hacking for Python3","status":"MERGED","created":"2020-04-01 17:53:28.000000000","updated":"2020-04-04 11:55:12.000000000","submitted":"2020-04-03 20:03:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716691-1585944192680-73754ddb","meta_rev_id":"f7fb7777c6b67827c1f93bc80ec6e0ae7c392375","_number":716691,"virtual_id_number":716691,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 20:03:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-03 16:57:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2020-04-03 17:07:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2020-04-03 18:26:57.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-02 15:31:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-04-03 18:26:57.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 15:31:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-03 16:57:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-03 17:07:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-03 18:26:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-03 20:03:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e597db0cadebee3b4631ef4e1822edf780242433","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-01 17:53:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f64f2cc8061d897c24d0627fa46806e9745beaa1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-01 18:21:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/3f247349334b4fbfa1dbab61998f9b4a : SUCCESS in 26m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fb5c221ad59e9b414d3b8835bd3643287a9773a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 18:40:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f8be5a9419647489fdcf1ac601fe379 : SUCCESS in 46m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dedf5ff824caf871442ebbb7cb11397dfadbe7a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 19:31:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a939c6b6f3204d68aa2c3aa0fc4c3ad5 : FAILURE in 19m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/bdffb4458cc94c809f0c6a8cdb1f06b0 : FAILURE in 27m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/605be67a521648b495d6a770763f6d83 : SUCCESS in 1h 16m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/29bfe2662a584941b927710a4c819b2f : FAILURE in 18m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c7704ff563ab4335a1e121b3cb8d88f0 : FAILURE in 23m 50s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/41593c488e99437ca3abed31e0a36096 : SUCCESS in 43m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ba53b29d0e1747d4ae793ae96c0fc81f : SUCCESS in 20m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f08e616f073a4ff2bfb2137f14f427dd : SUCCESS in 21m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e850ceece48f433b81546d7a488c880e : SUCCESS in 58m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32a78e212aec4b26bc57cc5f7097f27d : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeb421214dc84a71823fd4db3bdab54e : FAILURE in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02076ccd504a485b9dc6fc0310c15b23 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45cac9c17bda4f38b767dfd51719ab0d : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f94d56be3094c219c0f4785986cd562 : FAILURE in 4m 11s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/b87b6f96225448e8a05c32ba64f2e392 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11158fec51a64bc0b65b4d69ca399d7a : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/328f284ac9b94db09c16156669554686 : SUCCESS in 5m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b2940a9d3fa749c8a975e41fde005936 : FAILURE in 21m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/beb955b88b29494f8b399a4541954068 : FAILURE in 35m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ccb648e1b3654508a62b3cdcc5f8fbda : SUCCESS in 1h 00m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ac08028b5cfe46409a49a64460bab736 : SUCCESS in 1h 17m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/51f4a493ec2040498abed1f637bac01e : SUCCESS in 1h 34m 43s","accounts_in_message":[],"_revision_number":1},{"id":"115a5c4704d19c3062f56cd2d8a798c818e23c1e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-01 19:42:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e1c7232e01eb7d9b7ef20c3e913849218e30f836","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-01 20:10:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/aed24f173a334b1095059df9cd4c3b67 : SUCCESS in 26m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"342ade9646c969c3d507224b636b4bcb30027d4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 20:28:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dee12b4189a6407c84ef85f47e72410d : SUCCESS in 45m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6e19c673d8f58ae4997bf964b252689f653b85c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 21:24:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d5b6a308af374cc6bb786b288577efd0 : FAILURE in 20m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/badd1bdd4bda407f900a172233973c07 : FAILURE in 53m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/148d407e8397405ca3f046ffff08e34e : SUCCESS in 1h 21m 44s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f156d523915d4b608c30aeb4b48170bd : SUCCESS in 1h 40m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0e374d25523d4963b442dbcbda8ef59a : SUCCESS in 1h 18m 07s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/65406734cdc4423884e3822c00a35dc8 : SUCCESS in 41m 00s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6b423398dc954cc9bcb36d10ed436140 : SUCCESS in 19m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f72daa6322d4ec5b8f35afdeeab38b5 : SUCCESS in 26m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/be24cb316b924a4990ced6d468e54224 : SUCCESS in 1h 02m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4647970fb1264014a1946bdc0cb7318c : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8955c0a9022c49338dfc1c0025b04625 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2a2ea7a74834310b89eae642ebb828b : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1b42da4dc8647f18333c7bdb1c5b778 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a90b286c6024860b6a471b2e0056151 : FAILURE in 4m 02s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/e5ca5b1cac874e6e951cdd3932bf3c7d : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc50b504cebc425e925bf30ef56ffe48 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72b71def16484810bfe59b8871bdb9bc : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/40bb9ba7d8294d21bfc90c2bed5421c0 : FAILURE in 56m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a28f3552709d43e1976a9990d7581d52 : FAILURE in 24m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/74c1080b45fe409ab7bd11ce394ae8c5 : SUCCESS in 57m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91e650c3ffa941c5aeada41cd8cd3128 : SUCCESS in 55m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bddf09c783cc4751994a534119e94134 : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":2},{"id":"30b08e323a3aba97cfe87ce2262dac71f9eb81e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-02 15:31:00.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, Andreas.\n\nWe are having mirror issues atm but it looks good.","accounts_in_message":[],"_revision_number":2},{"id":"7db164f346410478a65bca683477335fab93bc95","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-03 16:32:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c3ff0918fadb7724bb3c5df472f21744ba90ba70","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-03 16:57:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c01c5ce2816a18d1a72ae053fcb5b442e23fcf87","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-03 17:07:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4444559c07b0eb386c9b8057e7c1f3c209999928","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 17:18:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c401601c3646454892ffdd21e69b8157 : SUCCESS in 45m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0727559e0cd61e7f9a7d093c7d78c1e1e2ddf18f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 18:24:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/88fc7df82f654dc784230816f3a5a8ef : SUCCESS in 26m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c24a7639efcb47f2b9ab17db7e2f13df : SUCCESS in 43m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a0226e6eb2b448a5a92c0a61a2c45c4b : SUCCESS in 1h 17m 01s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5d402ef80efa4ac7a859df987656c2da : SUCCESS in 1h 17m 46s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/719b252dc3d042b58e9fc5c5729f6168 : SUCCESS in 1h 31m 50s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e9a0d6d8370c4ea689edfcda49f1457d : SUCCESS in 57m 18s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/90779928141e426d8989d570db20b19f : SUCCESS in 24m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b7fad07eb949440f9e3f7e63b56b269b : SUCCESS in 21m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/181b63f460f6417483752b921b88be49 : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02dc486e98fb43fc8a219513ac7592de : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d15124bdb14ff8872ab68e795448a7 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbe3733d326c4610a8e46b4d761f011f : SUCCESS in 6m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b12d7252962d4a289e800015c44724e9 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1caadc3088d74a49b6c76b0e9e00d91f : SUCCESS in 5m 52s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/90db368b3e1c418e88a9091950f6630d : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b52aef849cf4d84b473dd7bf3496baa : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8990340043f94b10ac776300b8afc2ac : SUCCESS in 8m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e74738b973414cf2a51b85125e5ee23b : SUCCESS in 26m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/66509ade2a5b441f8147c9bd7fe39b5c : SUCCESS in 34m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c3ff0006ef26422fb3f881cb2c03c38b : SUCCESS in 1h 13m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cec9b98f35384cffa9aa6324a4f12b4c : SUCCESS in 1h 02m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d990ae11fdcb42d8addb693917eae4e8 : SUCCESS in 1h 08m 32s","accounts_in_message":[],"_revision_number":2},{"id":"6c3b4af8e533477df8829fe2e1b7e38ed1e98d7c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-03 18:26:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"2694776a75e480edac2ff60d807b36f4aff4d715","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 18:27:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"efc346dd9edab5ce62d7d61560e5ab27360552a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 20:03:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1f9b4a202fef4358b4253389046d02e45abff8df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 20:03:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/54b71d879d764d0f92cad2bfcb3ad5a7 : SUCCESS in 55m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/77d1efa091e246a8be0298c22ad34967 : SUCCESS in 1h 29m 38s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ed06a314c46c4d20bf6c8fc2f4ae3679 : SUCCESS in 1h 32m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8f5d243a586423a8afac4eb57bbae98 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18a0e7edbc04574b96392909340fa8b : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3a8ebdbd1f045038318e7f2aa63e29f : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bdcc688591b7430faeba16699282a795 : SUCCESS in 5m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/02881b427a1e489c881a1623f020f424 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd50f8b2f8b433ca78326f6cd58b4ed : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9ad634a59744253a649dc77eea67009 : SUCCESS in 6m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fdc861c8fcef4bcb82ad04057b94d9d5 : SUCCESS in 26m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ce8f9b7604f4a98b6e5735409cc4e16 : SUCCESS in 1h 04m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/122bf760e5c84b8ca1a331913ebb1b98 : SUCCESS in 1h 05m 06s","accounts_in_message":[],"_revision_number":2},{"id":"5020ec3ae09b9cf9bd0d098c2d42c124a4f39bc8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 20:05:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff2f3eeb5d04be8a80cf46b95f25bca : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7d8f4f7f4274dc39265d55e3225ea9f : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e099ced659585ff598517e97b099b271d61baa26","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-04 11:04:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"651b2d48f8414d6821cd48b2ef58478277a30a73","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-04 11:55:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1456d8b6f9f86d974459685091f143f0a0b4b345","revisions":{"7214ed3f9230451083a3c04f7c25b51e9da6c101":{"kind":"REWORK","_number":1,"created":"2020-04-01 17:53:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/91/716691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/716691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/716691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/716691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/716691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/716691/1"}}},"commit":{"parents":[{"commit":"69f74554671a49c75bebeeed9d83816ffbd7efec","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69f74554671a49c75bebeeed9d83816ffbd7efec"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 17:47:23.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 17:53:18.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found by updated hacking version.\n\nUpdate local hacking checks for new flake8.\n\nRemove hacking and friends from lower-constraints, they are not\nneeded in installations.\n\nChange-Id: I67e5ca24c550163c218fb06692de0b7c528d8318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7214ed3f9230451083a3c04f7c25b51e9da6c101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7214ed3f9230451083a3c04f7c25b51e9da6c101"}]},"branch":"refs/heads/master"},"1456d8b6f9f86d974459685091f143f0a0b4b345":{"kind":"REWORK","_number":2,"created":"2020-04-01 19:42:49.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/91/716691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/716691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/716691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/716691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/716691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/716691/2"}}},"commit":{"parents":[{"commit":"69f74554671a49c75bebeeed9d83816ffbd7efec","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69f74554671a49c75bebeeed9d83816ffbd7efec"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 17:47:23.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-01 19:42:41.000000000","tz":120},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFix problems found by updated hacking version.\n\nUpdate local hacking checks for new flake8.\n\nRemove hacking and friends from lower-constraints, they are not\nneeded in installations.\n\nChange-Id: I67e5ca24c550163c218fb06692de0b7c528d8318\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1456d8b6f9f86d974459685091f143f0a0b4b345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1456d8b6f9f86d974459685091f143f0a0b4b345"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
