)]}'
{"id":"openstack%2Fkolla~717024","triplet_id":"openstack%2Fkolla~stable%2Frocky~I8c2da6437333210a7926f2deb926ab5d1c93c3c6","project":"openstack/kolla","branch":"stable/rocky","topic":"bug/1870349","hashtags":[],"change_id":"I8c2da6437333210a7926f2deb926ab5d1c93c3c6","subject":"Fix multiple CI issues","status":"MERGED","created":"2020-04-02 14:20:03.000000000","updated":"2020-04-03 15:38:39.000000000","submitted":"2020-04-03 15:38:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717024-1585928319992-6894be15","meta_rev_id":"795cb16113014ffefaa509599ac0d47487d536c0","_number":717024,"virtual_id_number":717024,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 15:38:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-03 12:01:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-03 12:01:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-03 12:00:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 11:45:05.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-03 12:01:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-03 15:38:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c806a048763f47d00df28e8b436090e9ed6a5ec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 14:20:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff46406f3fd8181d26e68fa2fed56c3a517a1e6b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-02 15:37:07.000000000","message":"Patch Set 1: Code-Review-1\n\nwill need \" around pins just like stein\n\notherwise CI fubar atm","accounts_in_message":[],"_revision_number":1},{"id":"96908730e4ebfd0dc5fb2f3f083234fef1ec5714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 16:15:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/090f64d1b99648d9b45480f45f4932d1 : FAILURE in 58m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/072622e4f0df4130ab08e394b4438957 : FAILURE in 1h 04m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/75f940929eef471e9e3b6dbc40a88d56 : FAILURE in 45m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2dcf82f4cfd146579da3b8e8d1a61909 : FAILURE in 32m 11s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/7ad136ed1b254246825884164be22d37 : FAILURE in 51m 12s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f89c205e51be4fc59731b7976c2712b4 : FAILURE in 49m 19s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7a2a0f3d5ca34afc9c18b7599fd8c260 : FAILURE in 39m 22s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/2e88b1ce1b3d43d4adeeab8972c61414 : FAILURE in 29m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f088a5146cde46f58468a479a0ab3a55 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b5c5aab3bd45c4aefd8d11749b045d : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f47692987016461eb9f9e41df532f333 : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7ed503c60aa1446987751333e648624b : SUCCESS in 6m 20s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1ffb79c6973f43a0a2fee9643ab50c74 : SUCCESS in 6m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1733555aefb45208bcfc1fdbf6eebf4 : FAILURE in 1h 17m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8870b33e964049d79d4d45303590d2b3 : FAILURE in 41m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1955d772db2141a68324e45ca805f9b2 : FAILURE in 30m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e04d51c4212343a6be184aecfde70629 : FAILURE in 1h 26m 55s (non-voting)\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"b1a0e243a0c1136e47c1c5c6899318196440fc0d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-02 16:23:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e194037c8b1d354e6958d4ac7ad95c2c7eba618a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-02 16:24:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da0e36242208dfcb6e0a138fb9b9c1d186bf8af9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-02 16:25:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6a40a592ccda2320e8bfad66e888b065eca444b0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-02 17:58:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9c5d4f36781aa7d77e47afda191a036ffcd9fd27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 20:33:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/db7482458b6140daad8cc83e311e0656 : FAILURE in 52m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/07c50d5e866445a89a628c771e0fef25 : FAILURE in 53m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5eda98c7617542f79b0471b122db8960 : FAILURE in 1h 01m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/327e395e813d4a3d96cddae9e7bf0234 : FAILURE in 47m 44s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/83d552f1411642249a5a0072033d2d25 : FAILURE in 54m 51s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c679e09a1d734f3b89e4992a352248e3 : FAILURE in 57m 41s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/da796f74c55c4359a51a28f6ba1cef0e : FAILURE in 56m 56s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/f6cdca920b754cf6a039b53992b5585b : FAILURE in 49m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acaaed23f1534ba39549f330d2c4aa76 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a38720615f4a31a3414df0af7c2263 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf0eccb4d16b4bc4a01b8b004c94c1d2 : SUCCESS in 6m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b2c2544b3fd44e3b8515ce45bc79d20b : SUCCESS in 6m 24s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8135caf3417f49e5bf6c333a232c3c3b : SUCCESS in 6m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/724fe1ed2ed747e1b9971c3da042fb0f : FAILURE in 1h 32m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/40d6e06e93084faeb8cb9d5a5ff69f91 : FAILURE in 35m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e82057b1dfd44368506c1c8b17a0612 : FAILURE in 43m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/09a4dd380b52488782a538583d81da65 : FAILURE in 1h 47m 50s (non-voting)\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"39637332cb78fe632e2c69406a300d20636f9f63","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-03 09:12:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6e00866c5fe06b506fbfd9c1e58efeafad201acd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-03 09:13:17.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"7e8dfe37f3b0e00e2be5bbfd30f3270445e00215","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-03 09:35:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7afcd283236db17527a9b69af31d82ef9e3162ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 11:31:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a96c6654dcb74ae0850c065a2a020b5c : SUCCESS in 46m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e7e043631df24249bad18317d2d69289 : SUCCESS in 44m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/94085508fb41495986944719bb3d5e8d : POST_FAILURE in 7m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d40cc358697746bbb666de207d97bc38 : POST_FAILURE in 15m 20s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ea0f7c7d68a3409da753e44fd43b9c82 : SUCCESS in 47m 47s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/fce032edaae44822af55ca999295d620 : SUCCESS in 51m 47s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/47c9998c840142ad9094307f7be967e3 : POST_FAILURE in 9m 21s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4d8cf4674a0742f78064d3409bd05654 : POST_FAILURE in 8m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5005f715981f4986a2dd07f9e4aaa08b : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8796c328b01457ba5562b0a17e91727 : SUCCESS in 7m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/816a468f173b420d86d48dcd7cc04157 : SUCCESS in 10m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/adf45ae7a0204cf194f0b6ec63e33e59 : SUCCESS in 8m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/118e92195d8249c6bada0849fb44051a : SUCCESS in 7m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3d416fc97154192860af403ebe5e265 : SUCCESS in 1h 33m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe84a65db0624587aaa3709d9339f6f4 : SUCCESS in 32m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a69b16e1ab2047ea9d4993f5e017d3ed : SUCCESS in 1h 00m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3cc32f8811ad481a960951e652b84094 : SUCCESS in 1h 56m 57s (non-voting)\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"af5cff5061555e480a9b1161a9855420c5d6d8ee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-03 11:41:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6cd2c12004a039b8c4d90f2293ba0c0997ebd202","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-03 11:44:33.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b8150b2c5b64c67c2c6086f998c36e8ebbcb33e0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-03 11:45:05.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"eb6bcb7377404176c0bff8150da68ead7722114b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-03 11:49:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"68323e7a8557e8139013f275630a7eb2d826724a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-03 11:51:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"498be99bffa859e55c7006e03b8484a32b95c06d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-03 12:00:25.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"ea3afd7cbdf37d96ede0a824047a29604570de82","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-03 12:00:51.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"6712137bef387b9ea41c707b5e78d33c2cbf1100","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-03 12:01:04.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"dae1d1156216c5770a1052d6a3c4b1757ded9d0f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-03 12:41:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3f5a0770be794e6eb0832c1598fba8289bc79f82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 13:47:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ecb91261ba7e42d2ac821dff5210a4be : SUCCESS in 46m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/02fdb349b7e94398b7a97eccc81da723 : SUCCESS in 1h 21m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8d717adace0a428b90ac0f881885310b : FAILURE in 1h 07m 36s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2f4b7d28ee794827a5e1a50d4785f965 : FAILURE in 52m 42s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/046ddf2b57f0413dac42a220bd1c1cd0 : SUCCESS in 54m 30s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/fe46bfdf0ad24a1082d0d7d0b92cb6b4 : SUCCESS in 50m 58s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3228a9b869cb4e5aa68f7ab1cdac00d8 : FAILURE in 1h 02m 44s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/c9b4acdb140d47d5bde25a990c14df20 : FAILURE in 52m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5d8117a0d8448b699982c5cf04b33c0 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49955af48ac24f4885a853eadbf4216f : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbecc09e5c4c4cecb819e3cd7e129bdf : SUCCESS in 6m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0baf2a93694f4ec18de0f8340a58bac2 : SUCCESS in 4m 57s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3bc65bb06d624e9298cdda82ebdb6619 : SUCCESS in 7m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/34d44b8ee88543c6b91ae61105a01407 : SUCCESS in 1h 22m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a643076eca67429ab3a47b0183453d1b : SUCCESS in 43m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/62feec29f77847fea06f7681ff67e088 : FAILURE in 43m 55s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dec0c5e063c147f29a0c7938ff7c0488 : FAILURE in 1h 40m 38s (non-voting)\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":8},{"id":"274343d2b95768969223ccac54f50172568a70d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 13:50:33.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c702b296a9a2defee9c0446fd6508673b36049b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 15:38:39.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6b07216da07c4d7682bbd8b5caccccc6 : SUCCESS in 1h 46m 55s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/9d146fc3ffb8420a85d6fb96fd96ab62 : SUCCESS in 1h 45m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/673ccee7dac542ce9859e623584d2fdb : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7fdf87f0f246cba47e20bac2eec3e8 : SUCCESS in 7m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7b56f40609d4ce797237d57f23481d9 : SUCCESS in 7m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5b9d91cebad34db3ac16d4694dfd8449 : SUCCESS in 8m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6aea8f1021334a949f76b94199f90f31 : SUCCESS in 8m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/96f0ffabf0204471be58a16de22ca44e : SUCCESS in 39m 47s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: Could not switch to \u0027/home/zuul/src/\u0027: No such file or directory\u0027\n  Comments left for invalid file /home/zuul/src/opendev.org/openstack/kolla/docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":8},{"id":"22df4e9341ea03cf4e857c0c0831e5634f724e47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 15:38:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"7f4baeafdd80b5c88b57c1f5e00ae6e119fe2343","revisions":{"d200c9d7da47b5954d9db81307d530ab5d6667de":{"kind":"REWORK","_number":1,"created":"2020-04-02 14:20:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/24/717024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/717024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/717024/1"}}},"commit":{"parents":[{"commit":"61f9fdf5585a9f1f41a832564d9a330eab054461","subject":"Bump service versions (Rocky)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f9fdf5585a9f1f41a832564d9a330eab054461"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 13:34:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 14:19:32.000000000","tz":60},"subject":"Pin openstacksdk and client libraries in kolla-toolbox","message":"Pin openstacksdk and client libraries in kolla-toolbox\n\nStein and below only.\n\nBackport note: adapted versions for Rocky.\n\nopenstacksdk released 0.45.0 (and possibly 0.44.0 as well) which dropped\npy2 compat but did not advertise the fact. It was amended later but it\nwill now always affect kolla images until pinned to older.\n\nThe observable error is:\n\"ImportError: No module named queue\"\n\nSince Train we pin u-c, but not before that.\n\nThis change introduces a version constraint for openstacksdk and client\nlibs:\n\n- python-novaclient\n- python-cinderclient\n- python-keystoneclient\n- python-glanceclient\n\nChange-Id: I8c2da6437333210a7926f2deb926ab5d1c93c3c6\nCloses-Bug: #1870349\n(cherry picked from commit 535c240b9db9cccb792dd399f9de18787c33e03f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d200c9d7da47b5954d9db81307d530ab5d6667de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d200c9d7da47b5954d9db81307d530ab5d6667de"}]},"branch":"refs/heads/stable/rocky"},"41f1207d6a8eae232cab49fde2a6d50de9da1f33":{"kind":"REWORK","_number":2,"created":"2020-04-02 16:23:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/24/717024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/717024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/717024/2"}}},"commit":{"parents":[{"commit":"61f9fdf5585a9f1f41a832564d9a330eab054461","subject":"Bump service versions (Rocky)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f9fdf5585a9f1f41a832564d9a330eab054461"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 13:34:36.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-02 16:22:49.000000000","tz":120},"subject":"Pin openstacksdk and client libraries in kolla-toolbox","message":"Pin openstacksdk and client libraries in kolla-toolbox\n\nStein and below only.\n\nBackport note: adapted versions for Rocky.\n\nopenstacksdk released 0.45.0 (and possibly 0.44.0 as well) which dropped\npy2 compat but did not advertise the fact. It was amended later but it\nwill now always affect kolla images until pinned to older.\n\nThe observable error is:\n\"ImportError: No module named queue\"\n\nSince Train we pin u-c, but not before that.\n\nThis change introduces a version constraint for openstacksdk and client\nlibs:\n\n- python-novaclient\n- python-cinderclient\n- python-keystoneclient\n- python-glanceclient\n\nChange-Id: I8c2da6437333210a7926f2deb926ab5d1c93c3c6\nCloses-Bug: #1870349\n(cherry picked from commit 535c240b9db9cccb792dd399f9de18787c33e03f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41f1207d6a8eae232cab49fde2a6d50de9da1f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41f1207d6a8eae232cab49fde2a6d50de9da1f33"}]},"branch":"refs/heads/stable/rocky"},"46310f15cce87892892e8d3d5cd7d224e35c85bb":{"kind":"REWORK","_number":3,"created":"2020-04-02 16:24:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/24/717024/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/717024/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/717024/3"}}},"commit":{"parents":[{"commit":"61f9fdf5585a9f1f41a832564d9a330eab054461","subject":"Bump service versions (Rocky)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f9fdf5585a9f1f41a832564d9a330eab054461"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 13:34:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 16:24:01.000000000","tz":60},"subject":"Fix multiple CI issues","message":"Fix multiple CI issues\n\n1. Pin openstacksdk and client libraries in kolla-toolbox\n\nStein and below only.\n\nopenstacksdk released 0.45.0 (and possibly 0.44.0 as well) which dropped\npy2 compat but did not advertise the fact. It was amended later but it\nwill now always affect kolla images until pinned to older.\n\nThe observable error is:\n\"ImportError: No module named queue\"\n\nSince Train we pin u-c, but not before that.\n\nThis change introduces a version constraint for openstacksdk and client\nlibs:\n\n- python-novaclient\n- python-cinderclient\n- python-keystoneclient\n- python-glanceclient\n\n2. CI: Add https match for mirror.centos.org to template-overrides\n\nSome CentOS repos (currently one - Ceph NFS Ganesha) use\nhttps://mirror.centos.org instead of http:// and we break\ntheir definition by removing mirrorlist\u003d entry and not inserting baseurl\nentry due to a missing match.\n\nOriginal Change-Id: I7b29f8c6be80820042023d2b85162d5cc9a65f08\n\nChange-Id: I8c2da6437333210a7926f2deb926ab5d1c93c3c6\nCloses-Bug: #1870349\n(cherry picked from commit 4d11cd647a12ec900059f1cb53b30ad2b0fb22ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46310f15cce87892892e8d3d5cd7d224e35c85bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46310f15cce87892892e8d3d5cd7d224e35c85bb"}]},"branch":"refs/heads/stable/rocky"},"4525ade71fe819771829076a460e462c0f7e736a":{"kind":"REWORK","_number":4,"created":"2020-04-03 09:12:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/24/717024/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/717024/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/717024/4"}}},"commit":{"parents":[{"commit":"61f9fdf5585a9f1f41a832564d9a330eab054461","subject":"Bump service versions (Rocky)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f9fdf5585a9f1f41a832564d9a330eab054461"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 13:34:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-03 09:11:55.000000000","tz":60},"subject":"Fix multiple CI issues","message":"Fix multiple CI issues\n\n1. Pin openstacksdk and client libraries in kolla-toolbox\n\nStein and below only.\n\nopenstacksdk released 0.45.0 (and possibly 0.44.0 as well) which dropped\npy2 compat but did not advertise the fact. It was amended later but it\nwill now always affect kolla images until pinned to older.\n\nThe observable error is:\n\"ImportError: No module named queue\"\n\nSince Train we pin u-c, but not before that.\n\nThis change introduces a version constraint for openstacksdk and client\nlibs:\n\n- python-novaclient\n- python-cinderclient\n- python-keystoneclient\n- python-glanceclient\n\n2. CI: Add https match for mirror.centos.org to template-overrides\n\nSome CentOS repos (currently one - Ceph NFS Ganesha) use\nhttps://mirror.centos.org instead of http:// and we break\ntheir definition by removing mirrorlist\u003d entry and not inserting baseurl\nentry due to a missing match.\n\nOriginal Change-Id: I7b29f8c6be80820042023d2b85162d5cc9a65f08\n\nDepends-On: https://review.opendev.org/717262\n\nChange-Id: I8c2da6437333210a7926f2deb926ab5d1c93c3c6\nCloses-Bug: #1870349\n(cherry picked from commit 4d11cd647a12ec900059f1cb53b30ad2b0fb22ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4525ade71fe819771829076a460e462c0f7e736a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4525ade71fe819771829076a460e462c0f7e736a"}]},"branch":"refs/heads/stable/rocky"},"f2d2979b148698c351f6c241b1ea51691f518db7":{"kind":"REWORK","_number":5,"created":"2020-04-03 11:41:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/24/717024/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/717024/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/717024/5"}}},"commit":{"parents":[{"commit":"61f9fdf5585a9f1f41a832564d9a330eab054461","subject":"Bump service versions (Rocky)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f9fdf5585a9f1f41a832564d9a330eab054461"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 13:34:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-03 11:40:44.000000000","tz":60},"subject":"Fix multiple CI issues","message":"Fix multiple CI issues\n\n1. Pin openstacksdk and client libraries in kolla-toolbox\n\nStein and below only.\n\nopenstacksdk released 0.45.0 (and possibly 0.44.0 as well) which dropped\npy2 compat but did not advertise the fact. It was amended later but it\nwill now always affect kolla images until pinned to older.\n\nThe observable error is:\n\"ImportError: No module named queue\"\n\nSince Train we pin u-c, but not before that.\n\nThis change introduces a version constraint for openstacksdk and client\nlibs:\n\n- python-novaclient\n- python-cinderclient\n- python-keystoneclient\n- python-glanceclient\n\n2. Make deploy jobs non-voting\n\nFor similar reasons to 1., the deploy jobs fail when using the openstack\nclient for testing. Due to a dependency loop between these two projects,\nwe must make these jobs non-voting to fix the image build. This should\nbe reverted ASAP.\n\n3. CI: Add https match for mirror.centos.org to template-overrides\n\nSome CentOS repos (currently one - Ceph NFS Ganesha) use\nhttps://mirror.centos.org instead of http:// and we break\ntheir definition by removing mirrorlist\u003d entry and not inserting baseurl\nentry due to a missing match.\n\nOriginal Change-Id: I7b29f8c6be80820042023d2b85162d5cc9a65f08\n\nChange-Id: I8c2da6437333210a7926f2deb926ab5d1c93c3c6\nCloses-Bug: #1870349\n(cherry picked from commit 4d11cd647a12ec900059f1cb53b30ad2b0fb22ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2d2979b148698c351f6c241b1ea51691f518db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2d2979b148698c351f6c241b1ea51691f518db7"}]},"branch":"refs/heads/stable/rocky"},"8742a716de2d248557264fafcd8ffad5fc7c2fe5":{"kind":"REWORK","_number":6,"created":"2020-04-03 11:51:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/24/717024/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/717024/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/717024/6"}}},"commit":{"parents":[{"commit":"61f9fdf5585a9f1f41a832564d9a330eab054461","subject":"Bump service versions (Rocky)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f9fdf5585a9f1f41a832564d9a330eab054461"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 13:34:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-03 11:51:27.000000000","tz":60},"subject":"Fix multiple CI issues","message":"Fix multiple CI issues\n\n1. Pin openstacksdk and client libraries in kolla-toolbox\n\nStein and below only.\n\nopenstacksdk released 0.45.0 (and possibly 0.44.0 as well) which dropped\npy2 compat but did not advertise the fact. It was amended later but it\nwill now always affect kolla images until pinned to older.\n\nThe observable error is:\n\"ImportError: No module named queue\"\n\nSince Train we pin u-c, but not before that.\n\nThis change introduces a version constraint for openstacksdk and client\nlibs:\n\n- python-novaclient\n- python-cinderclient\n- python-keystoneclient\n- python-glanceclient\n\n2. Make deploy jobs non-voting\n\nFor similar reasons to 1., the deploy jobs fail when using the openstack\nclient for testing. Due to a dependency loop between these two projects,\nwe must make these jobs non-voting to fix the image build. This should\nbe reverted ASAP.\n\n3. CI: Add https match for mirror.centos.org to template-overrides\n\nSome CentOS repos (currently one - Ceph NFS Ganesha) use\nhttps://mirror.centos.org instead of http:// and we break\ntheir definition by removing mirrorlist\u003d entry and not inserting baseurl\nentry due to a missing match.\n\nOriginal Change-Id: I7b29f8c6be80820042023d2b85162d5cc9a65f08\n\nChange-Id: I8c2da6437333210a7926f2deb926ab5d1c93c3c6\nCloses-Bug: #1870349\n(cherry picked from commit 4d11cd647a12ec900059f1cb53b30ad2b0fb22ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8742a716de2d248557264fafcd8ffad5fc7c2fe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8742a716de2d248557264fafcd8ffad5fc7c2fe5"}]},"branch":"refs/heads/stable/rocky"},"7496e96ed1d59762f1623d03bd80a5222742f591":{"kind":"REWORK","_number":7,"created":"2020-04-03 12:00:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/24/717024/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/717024/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/717024/7"}}},"commit":{"parents":[{"commit":"61f9fdf5585a9f1f41a832564d9a330eab054461","subject":"Bump service versions (Rocky)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f9fdf5585a9f1f41a832564d9a330eab054461"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 13:34:36.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-03 12:00:23.000000000","tz":0},"subject":"Fix multiple CI issues","message":"Fix multiple CI issues\n\n1. Pin openstacksdk and client libraries in kolla-toolbox\n\nStein and below only.\n\nopenstacksdk released 0.45.0 (and possibly 0.44.0 as well) which dropped\npy2 compat but did not advertise the fact. It was amended later but it\nwill now always affect kolla images until pinned to older.\n\nThe observable error is:\n\"ImportError: No module named queue\"\n\nSince Train we pin u-c, but not before that.\n\nThis change introduces a version constraint for openstacksdk and client\nlibs:\n\n- python-novaclient\n- python-cinderclient\n- python-keystoneclient\n- python-glanceclient\n\n2. Make deploy jobs non-voting\n\nFor similar reasons to 1., the deploy jobs fail when using the openstack\nclient for testing. Due to a dependency loop between these two projects,\nwe must make these jobs non-voting to fix the image build. This should\nbe reverted ASAP.\n\n3. CI: Add https match for mirror.centos.org to template-overrides\n\nSome CentOS repos (currently one - Ceph NFS Ganesha) use\nhttps://mirror.centos.org instead of http:// and we break\ntheir definition by removing mirrorlist\u003d entry and not inserting baseurl\nentry due to a missing match.\n\nOriginal Change-Id: I7b29f8c6be80820042023d2b85162d5cc9a65f08\n\nChange-Id: I8c2da6437333210a7926f2deb926ab5d1c93c3c6\nCloses-Bug: #1870349\n(cherry picked from commit 4d11cd647a12ec900059f1cb53b30ad2b0fb22ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7496e96ed1d59762f1623d03bd80a5222742f591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7496e96ed1d59762f1623d03bd80a5222742f591"}]},"branch":"refs/heads/stable/rocky"},"7f4baeafdd80b5c88b57c1f5e00ae6e119fe2343":{"kind":"REWORK","_number":8,"created":"2020-04-03 12:00:51.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/24/717024/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/717024/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/717024/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/717024/8"}}},"commit":{"parents":[{"commit":"61f9fdf5585a9f1f41a832564d9a330eab054461","subject":"Bump service versions (Rocky)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61f9fdf5585a9f1f41a832564d9a330eab054461"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-02 13:34:36.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-03 12:00:49.000000000","tz":0},"subject":"Fix multiple CI issues","message":"Fix multiple CI issues\n\n1. Pin openstacksdk and client libraries in kolla-toolbox\n\nStein and below only.\n\nopenstacksdk released 0.45.0 (and possibly 0.44.0 as well) which dropped\npy2 compat but did not advertise the fact. It was amended later but it\nwill now always affect kolla images until pinned to older.\n\nThe observable error is:\n\"ImportError: No module named queue\"\n\nSince Train we pin u-c, but not before that.\n\nThis change introduces a version constraint for openstacksdk and client\nlibs:\n\n- python-novaclient\n- python-cinderclient\n- python-keystoneclient\n- python-glanceclient\n\n2. Make deploy jobs non-voting\n\nFor similar reasons to 1., the deploy jobs fail when using the openstack\nclient for testing. Due to a dependency loop between these two projects,\nwe must make these jobs non-voting to fix the image build. This should\nbe reverted ASAP.\n\n3. CI: Add https match for mirror.centos.org to template-overrides\n\nSome CentOS repos (currently one - Ceph NFS Ganesha) use\nhttps://mirror.centos.org instead of http:// and we break\ntheir definition by removing mirrorlist\u003d entry and not inserting baseurl\nentry due to a missing match.\n\nOriginal Change-Id: I7b29f8c6be80820042023d2b85162d5cc9a65f08\n\nChange-Id: I8c2da6437333210a7926f2deb926ab5d1c93c3c6\nCloses-Bug: #1870349\n(cherry picked from commit 4d11cd647a12ec900059f1cb53b30ad2b0fb22ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f4baeafdd80b5c88b57c1f5e00ae6e119fe2343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f4baeafdd80b5c88b57c1f5e00ae6e119fe2343"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
