)]}'
{"id":"openstack%2Fkolla~717456","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I72f80f1d869f904fc369b8575303f8448d9572f5","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I72f80f1d869f904fc369b8575303f8448d9572f5","subject":"Bump versions (Train)","status":"MERGED","created":"2020-04-04 12:41:07.000000000","updated":"2020-04-08 00:00:53.000000000","submitted":"2020-04-07 23:59:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717456-1586303961093-75094fdf","meta_rev_id":"127453735a937394bbc0ef86204fdf473bda1598","_number":717456,"virtual_id_number":717456,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 23:59:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-07 20:43:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-04-07 20:13:14.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-07 20:13:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-04-07 20:13:19.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-04 13:33:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-07 20:13:19.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-07 20:43:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-07 23:59:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"806d27ee6844fd1888cf2ab7c6112864d13a29e8","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-04 12:41:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b79508569279d658228592932fadc3761411dde8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-04 13:08:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/630a4cfc08f54b2c937f38f943deddbc : SUCCESS in 26m 15s","accounts_in_message":[],"_revision_number":1},{"id":"af81c214e1f48ad01eee16df33a2555ba62840ac","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-04 13:33:22.000000000","message":"Patch Set 1: Code-Review+2\n\nCI still broken","accounts_in_message":[],"_revision_number":1},{"id":"1ec14de87206e65018844482a0f4504c229877b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 14:01:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c13d8a7f55f74391bb9458d63dde7a86 : FAILURE in 6m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b14b7da710e94eae802435a18be0f389 : FAILURE in 7m 02s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4931084dbc53431db7f2b8170fc982ce : SUCCESS in 35m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/47e4916a26244872a1153dc410ee544b : SUCCESS in 38m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d5ded8fd39484df6b2c49fedc58f0ff9 : FAILURE in 10m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f2c1b27e804f4456953e05fc0eecdf38 : SUCCESS in 1h 09m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3bab93acbd0e44a393023595b1c03a90 : FAILURE in 12m 13s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/54bc53840fc14077845c83407e9d612f : FAILURE in 1h 08m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a2847754c4ca41dcaf23903ccf710f53 : FAILURE in 10m 32s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/150ac69bf6694e1d8692a8c21af73012 : SUCCESS in 1h 01m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9d067bde6be54c68aef5f42572d59707 : FAILURE in 3m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6273809d602046718238c99de98f4ebb : SUCCESS in 21m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3ac11dec809a45169571ebb59e5ef7dc : SUCCESS in 59m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/568a1b7fd4654c4b989312a4c352c4bc : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a8872a9a8d241608746e61f6a7fc19d : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4836b0a79c8f461ca1d44c4cee782b5e : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb4e3c40620d4017b8c955bc37194b2c : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/200d378353af458eb79baed6da26b664 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9998f23d952f46e2b8f3991916bae02a : SUCCESS in 6m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c7bc160d2e04352b95a4e8138e33953 : SUCCESS in 43m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1ee72f2c77b4ee983a7bc3d29b02e22 : SUCCESS in 41m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef39439ba6114e15a073cc25c775a6dc : SUCCESS in 1h 03m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d3539de2a9a542a18df5e5cd4fdb9fb8 : SUCCESS in 48m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/44f2a156aaed4dfd989119e9a56475fa : FAILURE in 48m 30s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"14808bf2ef226c43327c01ac486f6fc7f5dc0453","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-04 19:39:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e19546d1d17f37548e4c8033e7c96d03d6ae2269","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-05 07:28:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44f287a9e8958ca988b07e48597d2c111bdea300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 10:23:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6b69c138254341dda6606a4098410623 : FAILURE in 7m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8ee1d7f5529a486e966f457a8e60f002 : FAILURE in 7m 57s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/08f6110d80b643818311e8f5cabd690d : SUCCESS in 2h 35m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cd247aba604f45a7a6904d922ef31a87 : SUCCESS in 36m 37s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2a7faa94b2a4415a9fe7b3fe88f75035 : FAILURE in 11m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/695a0c00fa61437aaedaf4cceafdd766 : SUCCESS in 1h 11m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f4a15414f96049eb88d9d69b2c24cfa1 : FAILURE in 14m 40s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4d46843bc877419e922b70a5751a9621 : FAILURE in 1h 38m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/694864ea76b74c2da92721ac5f9670e1 : FAILURE in 17m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0d6a1fde6ed54a279cc4fa56b204e536 : SUCCESS in 1h 33m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0f530902e0c4411e87a4c0e25a2159cb : SUCCESS in 1h 06m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a1b7d79bb8b04cad8ad70b4b2e5e52a8 : SUCCESS in 1h 13m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b5fa70dd0d0d4b2fa826d1e5792e05d1 : TIMED_OUT in 2h 03m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dfc8608fd9543b8962f8f6eeeedb0c0 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/480b10487c6c4713b6c9d185eac2c734 : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93b60e85d27e485798d126b1f95228d6 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a8f4d801003489ea57a57d17890158c : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03642e50f929452f97d5af754bd0881d : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21bf0f67296346fcb03fbb81f560b1e0 : SUCCESS in 9m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/645e09aee17348aa92106f6b7a885edf : SUCCESS in 2h 31m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/371795cdd89d44598b48637ae86cf63d : SUCCESS in 2h 27m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c00fb93caea14076bbf92505a105ed7f : TIMED_OUT in 2h 02m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2009d2516f024389aff549c1fcebb89e : TIMED_OUT in 2h 03m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ff5152b82a374fdbbb06a5c9277d6c3a : SUCCESS in 2h 27m 54s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"7a2953a9031470de3cf52bdfe9b5fba504934e09","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-05 18:55:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e920d1e17b6002bc95b5b5ad5d478b0247592eae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-05 19:23:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/3fbce76e8f08405ba5ef1af9dbbcfa99 : SUCCESS in 26m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e195868cec15ae7fde5f441722cf4cd4ac89b656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 20:19:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cbc75ad33a994e27abf5995f0547a94b : FAILURE in 7m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fb7dff6693194097832d140c57043026 : FAILURE in 6m 06s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9d60465207b7412bbcd3061be7d38b33 : SUCCESS in 27m 33s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8dbab5800fe44ebb9aed67147cb74161 : SUCCESS in 42m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/71ad234e896047dcb110e54addc159d8 : FAILURE in 12m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f643fc525196417eac91e962223f7d65 : SUCCESS in 1h 23m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0ef29646bb1142568e6e0b1436e1f73b : FAILURE in 13m 25s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0eb45a5cf6b849009a85aa1a72e1f781 : FAILURE in 57m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/154e5e6e00c24ea1a90d86cdf4986cb0 : FAILURE in 11m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b7f43d305d6f4905a5c8e3ec9feb4afa : SUCCESS in 1h 01m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2012548e7b634335bdfa915c0d5c586b : SUCCESS in 25m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/423103564c2e4d4c859f9c7db9ebff7b : SUCCESS in 22m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/434a2e8060284bfdab1bfd5a73746c8f : SUCCESS in 59m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4674d390ba542d7929d04b861dc6889 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f431655e124a47598f7fa9896b527225 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7060ab1a9d4a40db98258d9d2b8d07cc : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1b6b83bc5a0495daf953e4481c45ecd : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d4b85725fae496fa488289ad148af9f : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b05afd820f4cf9a7547ac8ffc4fc51 : SUCCESS in 9m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d6fbbb412208417ebc6a95e9081e56ef : SUCCESS in 34m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a660985197e40a9a463f7d51020cf9e : SUCCESS in 31m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/212e0d612d794162be4ac6063e013f1b : SUCCESS in 1h 01m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/63105c57ca324675b03ae8d308bbcd5c : SUCCESS in 59m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2b4a1a4cb91643649ca6697b7aa3ae86 : FAILURE in 45m 29s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"8598d73926d20c5cbf29e020d073874bae9dc7a6","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-07 09:04:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"902a1fa95358584c163be31544c94dd6940321fd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 09:32:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/18f05f2a414246608059155d7ba86581 : SUCCESS in 26m 39s","accounts_in_message":[],"_revision_number":1},{"id":"05b99132cab5d5097bceadadd04ed7399b7d3df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 12:17:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/46e01e1294714ba4aa6393a752979c1b : FAILURE in 6m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e9c4bb22c83945fca1dd5ef1817b0690 : FAILURE in 8m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a76905844b1049cb9d86443789bcf2ef : SUCCESS in 28m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b2bbe50ca021428eb47b622a7f11a97e : SUCCESS in 36m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/800179f517dd44549993c7abb993f19c : FAILURE in 13m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4b5e6044f2964f71902f3b2454385150 : SUCCESS in 1h 10m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ee168f077d3c4c09805466e9b7c8fffe : FAILURE in 14m 39s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b6923646bff847fa93f43de4a3cc003c : FAILURE in 1h 13m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d6705251e36e4d69a00ece9f863d0b71 : FAILURE in 11m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ab057043a09844328ea7fdb6c162982e : SUCCESS in 1h 57m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fbb4d3fdcaa14d54b61cf49b8ee3b42d : SUCCESS in 34m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d96d04b209a41dca4beef7461370423 : SUCCESS in 1h 05m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/40c775d9b5ef466a92dbb2d9e21ec372 : SUCCESS in 1h 03m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99b81a257f5d4551927045f4cfb2267e : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343e52feaa6244ac92ba9adb63844111 : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d84bf936d5664e8b931a0e979fdb8e8f : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69c383ecd9b748e99396f9530c821e7d : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b4ee59f8b134f9a8a624407ef951836 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a420c671a9294ff48a74ba936594cd89 : SUCCESS in 6m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9f8b9923e3e42ae825f57509e09907d : TIMED_OUT in 3h 01m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4e4ccedd9b9462ba81e3ad815389957 : SUCCESS in 1h 13m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d777dec051a4da19a57ad4d94fc1d7b : SUCCESS in 56m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/796e00aa87b04b72a968b51fec9ef7a1 : SUCCESS in 1h 02m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9fdc2cb2365a4e14ac3d672f8e584553 : SUCCESS in 1h 12m 46s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"754bef32a0fa8d30084aff8eb4a2145ff606c4ad","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-07 20:13:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"88a87b21f8831fed225a2ba4e4d994fc2549b5b7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-07 20:13:19.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"70bc09fca81eae1913160d471062bfd158664ff0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 20:43:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/a24c7cf2c5a14e2bb2ccbf6b64c480e0 : SUCCESS in 27m 39s","accounts_in_message":[],"_revision_number":2},{"id":"73aa7c52b23a24f4fa5c822d4362bec052a1c34d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 22:13:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6f0562cf2911436c9e45b0db50df86a1 : SUCCESS in 1h 08m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/836c4e805c444c7db34f3a8172425dc5 : SUCCESS in 1h 15m 05s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c375d7d0d92c4b44813c41e3c57753ff : SUCCESS in 41m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/94410b04ed6e4705917b63bde01de417 : SUCCESS in 42m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f6969888c7274118a903798232c30bec : SUCCESS in 1h 30m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e9d23fcc60ff4a9989c34a9f38152fbc : SUCCESS in 1h 19m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8efd49bc9634419ab7cbd0456ad00302 : SUCCESS in 1h 29m 12s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a184c3c2a8cb4073952073d55a2e9f2d : FAILURE in 1h 24m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aae430d2cb424ba69414bacc70290c00 : SUCCESS in 1h 47m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ef00749ca6864f87a7d3c36d493e1f65 : SUCCESS in 1h 22m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1f15c055d52c4625ab74b9a7dabe9acc : SUCCESS in 28m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6c657f3320f643f8808052acd22f64f8 : SUCCESS in 17m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/282de0dcf3ae44928a8bb850dfc8676e : SUCCESS in 1h 21m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/963448fd61e74a22953e10ad3c5678a0 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343079b528d84549bccdb88e69b2de52 : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39d6962e6c5747e28cd1badd57e2e9d9 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea586820de75471e9275d0a8a5cf5691 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88e320de79ca49c4a1715fc122adf669 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc84481b56fa42bb8d4e577cb22458b1 : SUCCESS in 9m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bb7ca58ff1ca4923967cc5023f7d284f : SUCCESS in 48m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4329453dc624899b6e916ae582a1ea6 : SUCCESS in 50m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3192b51552e34ff2aa8f8e336e79e49b : SUCCESS in 1h 12m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/92af3072b94b4831bd41bb560398a5e2 : SUCCESS in 58m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a5aaf6f87cf94fcbac20fbc605d187a1 : SUCCESS in 1h 31m 27s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"4a6daddc83adec71c70f6b5122777cada4b5f352","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 22:14:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9f00b3a6aca0d6c86884e58e77e96c26de8d475c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:59:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/36c018706e5c40d5bc169195bf5fcc7d : SUCCESS in 49m 45s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/747dac99b2fc4187a56b9ba49118b869 : SUCCESS in 37m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c44588a1e2e44c3d9c98d9b512b3b718 : SUCCESS in 1h 22m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/30a29a3f54fe4e8fa79eb02bfdc8e77c : SUCCESS in 1h 24m 56s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3e0b668b25e54889ac60c76d20b86ec7 : SUCCESS in 1h 39m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6d1c2be3e466423a9e313c391865832d : SUCCESS in 54m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00e018e0e42041bf9168accb4b431143 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a40dd5567bb74fa287be8dfce098fb42 : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e554c6bd6f747e099455eebb1f43c82 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60481a15185e4fecbe9c8e44aaa57bfe : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d85293527d7444eab28031f4b14a958d : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668e7c3149eb4be5bfcb9a023bd2d934 : SUCCESS in 9m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a87c6834ca73476a9fa0d8f16babea2a : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ed250c1f6024c53aa91611f94f5c9e1 : SUCCESS in 58m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2ed7f9e28cb049658261c4820649aec4 : SUCCESS in 1h 16m 36s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"a17b52290c121d2180c1814a1ef89c889bd4c888","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:59:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"127453735a937394bbc0ef86204fdf473bda1598","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:00:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0289bb3c26a84315ace920e19a359716 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5a8e4e2d890ffd818c7748187244d2ea027d35c","revisions":{"84e1f526f884e7b40f89f9aa9bc3b3481f76ebbc":{"kind":"REWORK","_number":1,"created":"2020-04-04 12:41:07.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/56/717456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/717456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/717456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/717456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/717456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/717456/1"}}},"commit":{"parents":[{"commit":"b912f94c5d69203d5f53291c2c96fb67a5a9aab0","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b912f94c5d69203d5f53291c2c96fb67a5a9aab0"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-04-04 12:40:42.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-04-04 12:40:42.000000000","tz":180},"subject":"Bump versions (Train)","message":"Bump versions (Train)\n\nChange-Id: I72f80f1d869f904fc369b8575303f8448d9572f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84e1f526f884e7b40f89f9aa9bc3b3481f76ebbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84e1f526f884e7b40f89f9aa9bc3b3481f76ebbc"}]},"branch":"refs/heads/stable/train"},"a5a8e4e2d890ffd818c7748187244d2ea027d35c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-07 20:13:14.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/56/717456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/717456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/717456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/717456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/717456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/717456/2"}}},"commit":{"parents":[{"commit":"97c44232b2a2861d70893bd065d3f1956938d567","subject":"Merge \"CI: Use network mode \u003d host\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97c44232b2a2861d70893bd065d3f1956938d567"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-04-04 12:40:42.000000000","tz":180},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-07 20:13:14.000000000","tz":0},"subject":"Bump versions (Train)","message":"Bump versions (Train)\n\nChange-Id: I72f80f1d869f904fc369b8575303f8448d9572f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5a8e4e2d890ffd818c7748187244d2ea027d35c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5a8e4e2d890ffd818c7748187244d2ea027d35c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
