)]}'
{"id":"openstack%2Fkolla~717506","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I2f895aea0cc59d808129e9fc636af0890196af33","project":"openstack/kolla","branch":"stable/train","topic":"bug/1867614-stable/train","hashtags":[],"change_id":"I2f895aea0cc59d808129e9fc636af0890196af33","subject":"Fix --skip-existing and --skip-parents","status":"MERGED","created":"2020-04-04 18:24:54.000000000","updated":"2020-04-07 12:29:23.000000000","submitted":"2020-04-07 12:26:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717506-1586262393396-86ddf8e2","meta_rev_id":"64ced35b2448ddd47f7b54463f977ae3f5f8ee41","_number":717506,"virtual_id_number":717506,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-04-07 08:40:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 12:26:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-07 09:09:25.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-04-05 07:27:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-04 18:53:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-04-05 07:27:33.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-04 18:53:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-05 07:27:33.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-07 09:09:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-07 12:26:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"135c743ffadd218abb3a7e2e172c757740cf96b3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-04 18:24:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ec928ee61f9bba6947ada5ab1bf925220fdfad73","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-04 18:51:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/4014c21c790c408f8641ba66514cd4f4 : SUCCESS in 25m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2f8c8359cff88d5e12494d1bbe6af254b494ffc5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-04 18:53:44.000000000","message":"Patch Set 1: Code-Review+2\n\nlooks clean","accounts_in_message":[],"_revision_number":1},{"id":"7767b8ffc82ae5fd971e8d872dba48f13ef63a79","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-04 18:53:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"31cfb9a006295c89d43dcf0023761cc26ea66ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 18:53:54.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d68c86e6fa47e2c40118d6e6faa8bb14ede7ba70","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-04 19:20:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/b5e03fc08ee5401f9a2a103e69b8e3cd : SUCCESS in 26m 02s","accounts_in_message":[],"_revision_number":2},{"id":"989283c6c04d49991cb3600565a27e5de2fee360","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 20:31:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/526481e31dcd40d8b5cfdd014f9417c6 : SUCCESS in 36m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f38e895e089f4154af33362eee6132c5 : SUCCESS in 47m 19s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7e68dd0a630d4049a3565838f98e8979 : SUCCESS in 30m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/32a7da3ecbd3495f9ae9a4b41497dc35 : SUCCESS in 39m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ffaae4b125864e79b3ad3c6617ea4707 : SUCCESS in 1h 33m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f589f7d20a3042aa87e4e091dc151a24 : SUCCESS in 1h 06m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/91e4490bb49f43eb968e85b40869e2fc : SUCCESS in 1h 14m 14s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/27ca4c26964942da97bed4a024e3f05e : FAILURE in 48m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cbe36038479340959da817bcf2341554 : FAILURE in 58m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1201bae348d445d5b3ea874196d0c02b : SUCCESS in 56m 12s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/261c8ca1a7fa445b84447df31548139b : SUCCESS in 21m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a4ef2512294343738af79481f3e8b26e : SUCCESS in 19m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/debf5e4e77e14b21af86517cee6acff8 : SUCCESS in 54m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca3d9a8d52954aecb4a86374b4e52e7f : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b95436bdacbc4418aef0f278893d1eb0 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/759098ffc03d4bec847619f91bb08249 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e65fc94d010548d886b98f4929812302 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa74f2ef39d7451294c35815fb5dac07 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a033584047f444daef6dc9351b0545a : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91e15650d32e411980bbedd58c9da00e : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fad2c3f228f8405bb5c7330bbb2c45f4 : SUCCESS in 23m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ccf3d4db1764efa9bec88cbbee5dafa : SUCCESS in 31m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d6bdb2a18224bd38e8923ebf10469f5 : SUCCESS in 54m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/339a664018a34b39aee94ee3756af7af : SUCCESS in 51m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cf34c0674c364b04afb310ad57ac2d70 : FAILURE in 41m 38s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"d8c09567b6bc6bf48c4e442743d3ea9f9b733f06","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-05 07:27:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"be9c0acbaff3c395c0b2b27d5960b8d34fa66231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 09:43:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/935a2687c2304c3e9fa320278478ba87 : SUCCESS in 46m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2b17975e38e04ef0a3ad74505f826d8c : SUCCESS in 54m 44s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e44b3a0d48a04410a628fdbafe112d46 : SUCCESS in 27m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ddeea5ab0f0c47098fd14cc5198ffd9d : SUCCESS in 32m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b546a62970d944c9bd761e0c7584e89f : SUCCESS in 1h 54m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f7166609449b40fbb8a6ff03fef5eeb6 : SUCCESS in 1h 09m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/70ddb70c2b484d7a84b72311250554fc : SUCCESS in 1h 20m 23s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/58829f759459438ea8dcd92278dcbfe7 : FAILURE in 50m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/423266ab021b4d568ccd4e12701d09f7 : FAILURE in 1h 11m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/054e99e08b5f4571b548eeb6c0353e72 : SUCCESS in 52m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a567c63ef3c241e3b09e63b16ecec175 : SUCCESS in 25m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/63a1724011944d759f633aa3748aa2f3 : SUCCESS in 19m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/97d0c86d965047dba5a06616eee4d1f5 : SUCCESS in 56m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4ef793d9fba485a83c977f15a2f137c : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/425fa2bb79cc41ad8f76f8a4818cf0dc : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79aa81036e254f35aaf797b28cf8124f : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8faeb2cc066d422a8e1efb0a0f740ef0 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9594357038634c128169193d32d169e5 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c3b35fb82249efafafdc2eafcc11e6 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8962f1f3750a4f77b841f18a3901ebcf : SUCCESS in 4m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43cbe682919943e18550d506e587a59e : SUCCESS in 1h 32m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/877d00fdf57b49579b1b1478099b5019 : SUCCESS in 1h 33m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f36f36ed85ea4d3192f443d3018e2f5f : SUCCESS in 1h 38m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06d33ec8620b462385f92616369d83e8 : SUCCESS in 1h 24m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f3c13cbc39144986bbe5318d58ffecef : SUCCESS in 1h 47m 41s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"98b5ce47ef20b3894d1a5b4b22d57a7ae0fc9007","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-06 19:58:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"58601abef3ed4d19432140ed33581af4ae790436","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-06 20:27:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/fae5ccf9c6e74700baeef517f2dae2cf : SUCCESS in 27m 33s","accounts_in_message":[],"_revision_number":2},{"id":"37ce5ffcc28b0c6d383233cf62d43520df1a714e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 23:03:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/dbbf6671e4d447c9af11e471ea32fade : SUCCESS in 41m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ac7f7c4011094316a27cf2e194cfe936 : TIMED_OUT in 3h 02m 52s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9e209caca9ea4a1cb0ce072ebd350c56 : SUCCESS in 36m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d253e493056b4e54912ad25c821e03f3 : SUCCESS in 37m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5c4bbff69f234e269571057702b254b3 : SUCCESS in 1h 26m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c4277686886146b4ae3c2ac17917d066 : SUCCESS in 1h 18m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c0d24208f1044e33b979ff9d968e20d6 : SUCCESS in 1h 14m 29s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a7612187efdc49689462490be2a8bfe5 : FAILURE in 50m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2e84e689a7a048b3939019a6645b94f5 : SUCCESS in 1h 28m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/aacb6729a4114d5585f891d13f9270d0 : SUCCESS in 52m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ad86698b89e7427bb5821a56e67d231f : TIMED_OUT in 3h 01m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/49147df8815c4506b7ab1e2966b57e78 : SUCCESS in 19m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/37e7707657e84789b229930b6017f8b3 : SUCCESS in 59m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ab55dc8501a40028588a82761f0cc13 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b0058099524b3d96b4637d2eef1a18 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6d29db4c55641ca93c1270215391fc2 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb78245b1eb144499591fbd3ab9389b3 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c898aab620d4805a6e173ca4c6a8d5a : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8362f441e3fd4d348a7a2ea684df9f6a : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bb239135f804b88a81af62b0b8b880c : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0e92140b704e43198fa2e5773f6fff81 : SUCCESS in 22m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cb2423d658b54c4b8ee07d38155f0e64 : SUCCESS in 32m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ae98e5d52654da3a91dd55e80796086 : SUCCESS in 1h 04m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/69c4b3a43bad4d8b8dd7aba8a3c72148 : SUCCESS in 55m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9b7a0ccc98b84ecebe5e95c800b85ec8 : SUCCESS in 1h 05m 28s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"f29d3d244edb4d23b401b571537e5703ce813613","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-07 08:40:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"86fa6d2e456cab1c249f647e4cf595576058c33b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 09:09:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/136afbd47f74425f897448cbaefe2b86 : SUCCESS in 27m 07s","accounts_in_message":[],"_revision_number":2},{"id":"be832880587cba5fde9793139fbc104d82301eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 10:35:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/de69501e117e4818b75bb4b344329a84 : SUCCESS in 1h 22m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/25c0adb37df04066b0efdd99c65d395d : SUCCESS in 1h 36m 52s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/425e7393d8434fcebc802540bb711952 : SUCCESS in 39m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1d749a7b4c2a478c9d8a2ed85b8c1ee5 : SUCCESS in 35m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9c79eead151743deb838346763b9f364 : SUCCESS in 1h 18m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e16527b0d3164241b5ff5f51948198b2 : SUCCESS in 1h 25m 07s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3dce848217e746bd82ec61d0860a57fd : SUCCESS in 1h 19m 18s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5724633a627a4e11833fee8161430a97 : FAILURE in 51m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e551a5d5ed114e3498ac69e9355bd406 : SUCCESS in 1h 47m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2e0ab225390e43c8b6ccbbc85bdcf06b : SUCCESS in 1h 07m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/27308dd593664d9d82c6273a4ea6de6d : SUCCESS in 39m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/17b23ab657474dee96b4722539f1559a : SUCCESS in 37m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8cc9322d61084fac81137b50ed7531fe : SUCCESS in 1h 02m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d3c5ab8662f44b5a6add57a4397260f : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3429ba3ea71744a3903bf6a0c40e36d9 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5de40c065b8c4c379e87b4b712419af9 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2bfa10bf5f94c9ebc313533501e6aae : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a6b367a50c547369240525de7574a10 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/764bba74ec034df98523036c29987817 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d30bf9a44b1046da9d5a2cf8fcdf6855 : SUCCESS in 5m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7cfaf508537e4ba59f4770b18c9bc861 : SUCCESS in 23m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a77c592527cc44acb009f9205ba81436 : SUCCESS in 27m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/01c93ad241314ee6b94ae3c8a9d78b76 : SUCCESS in 58m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/edfa32424c4846c3ab0db20e453602dd : SUCCESS in 57m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4fc233c89fb44af7a9acd642269aeb8f : SUCCESS in 1h 05m 13s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"a821ed6c4451464d6159c7e761c0282f26769892","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 10:35:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"992327460dfabb2abea66953f616994a2fd7aaf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 12:26:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"77531d9f99f4825316c60691918aed9606af4416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 12:26:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4d7d844fb5c9412485c6c18d1398658d : SUCCESS in 50m 31s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e69e736f396b4e88afe540ee0031c309 : SUCCESS in 37m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2ea9b570e88242559b6c1b88deb1b52f : SUCCESS in 1h 33m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fd7d615e23f7415eb9eb7a2c25a6dcfe : SUCCESS in 1h 31m 09s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fa8b59d0e3de4dd78fbb10911cf102f5 : SUCCESS in 1h 48m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/12b2f2742ac847a9bcdcf55a7718da48 : SUCCESS in 58m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02492149166c4f1daa8945386a8893f6 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae7fc970a2da466eb5b8a2709cd7cce7 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17e9ba8077e14afc9abb8d8013af4b9a : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/366f713f66ec48c99cc8956e7f984dd2 : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e56e437f579432eb725b8781d988ed7 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a87bcf74c84515bfd1661c36fedadd : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1be3cbb02eb4e1ea8fad979c51bac75 : SUCCESS in 5m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b6d9a77a7a1a42978c634da099a61d1f : SUCCESS in 1h 03m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e82b7e6abaff4eb8a2d8a514d3b2fff3 : SUCCESS in 1h 07m 59s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f2219285a43d42e0a66e2bdd6118fbb8 : SUCCESS in 1h 05m 53s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"e11fdcab924ea2f959f6a75406623b16b2e4cdef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 12:29:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221e778c9d274cdca72cad4faf648edc : SUCCESS in 2m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c0957d358544a1582b92d5beb0ea813 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c59dc07e9031eaf127b4222a66efc06db29454d2","revisions":{"2a49b3dc67df9dd43330cbddf5892296971487b6":{"kind":"REWORK","_number":1,"created":"2020-04-04 18:24:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/717506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/717506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/717506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/717506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/717506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/717506/1"}}},"commit":{"parents":[{"commit":"b912f94c5d69203d5f53291c2c96fb67a5a9aab0","subject":"Merge \"CentOS 8: Drop trickle install to build freezer\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b912f94c5d69203d5f53291c2c96fb67a5a9aab0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 12:54:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-04 18:24:54.000000000","tz":0},"subject":"Fix --skip-existing and --skip-parents","message":"Fix --skip-existing and --skip-parents\n\nThere are currently a few corner cases with the --skip-existing and\n--skip-parents flags. Examples:\n\n1. kolla-build --type source --skip-parents monasca-grafana\n\nThis does not try to build the monasca-grafana image. This appears to be\nan issue when the image is a grandchild.\n\n2. kolla-build --type source --skip-parents\n\nThis will build all images. Expect it to skip parents.\n\n3. kolla-build --type source --skip-existing\n\nThis will build all images. Expect it to skip existing images.\n\nThe filter_images method does quite a lot, including handling\nwhether images are buildable, regex/profile matching, and which images\nto skip. The matching and skipping parts are done in a single pass,\nwhich can lead to some weird effects due to dependencies between the\nimages and their statuses. Also, skipping is only currently applied when\nthere is a regex/profile filter.\n\nThis change splits out the matching and skipping into two separate\npasses. In the first pass, we mark all buildable images that match the\nfilter as matched. In the second, we iterate over matched images,\napplying the skip existing and skip parents rules.\n\nChange-Id: I2f895aea0cc59d808129e9fc636af0890196af33\nCloses-Bug: #1867614\nRelated-Bug: #1810979\n(cherry picked from commit dfddcce3a5bc98cbd177c984fd9d0a8bab51d40c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a49b3dc67df9dd43330cbddf5892296971487b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a49b3dc67df9dd43330cbddf5892296971487b6"}]},"branch":"refs/heads/stable/train"},"c59dc07e9031eaf127b4222a66efc06db29454d2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-04 18:53:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/06/717506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/717506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/717506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/717506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/717506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/717506/2"}}},"commit":{"parents":[{"commit":"169b3aa7c3569194b33747403fee4535044885ff","subject":"CI: Add https match for mirror.centos.org to template-overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/169b3aa7c3569194b33747403fee4535044885ff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-16 12:54:11.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-04 18:53:49.000000000","tz":0},"subject":"Fix --skip-existing and --skip-parents","message":"Fix --skip-existing and --skip-parents\n\nThere are currently a few corner cases with the --skip-existing and\n--skip-parents flags. Examples:\n\n1. kolla-build --type source --skip-parents monasca-grafana\n\nThis does not try to build the monasca-grafana image. This appears to be\nan issue when the image is a grandchild.\n\n2. kolla-build --type source --skip-parents\n\nThis will build all images. Expect it to skip parents.\n\n3. kolla-build --type source --skip-existing\n\nThis will build all images. Expect it to skip existing images.\n\nThe filter_images method does quite a lot, including handling\nwhether images are buildable, regex/profile matching, and which images\nto skip. The matching and skipping parts are done in a single pass,\nwhich can lead to some weird effects due to dependencies between the\nimages and their statuses. Also, skipping is only currently applied when\nthere is a regex/profile filter.\n\nThis change splits out the matching and skipping into two separate\npasses. In the first pass, we mark all buildable images that match the\nfilter as matched. In the second, we iterate over matched images,\napplying the skip existing and skip parents rules.\n\nChange-Id: I2f895aea0cc59d808129e9fc636af0890196af33\nCloses-Bug: #1867614\nRelated-Bug: #1810979\n(cherry picked from commit dfddcce3a5bc98cbd177c984fd9d0a8bab51d40c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c59dc07e9031eaf127b4222a66efc06db29454d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c59dc07e9031eaf127b4222a66efc06db29454d2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
