)]}'
{"id":"openstack%2Fkolla~717891","triplet_id":"openstack%2Fkolla~master~Id5e143636f9a81e7294b775f3d8b9134bee58054","project":"openstack/kolla","branch":"master","topic":"bug/1869319","hashtags":[],"change_id":"Id5e143636f9a81e7294b775f3d8b9134bee58054","subject":"Add vitrage-persistor image","status":"MERGED","created":"2020-04-07 01:41:43.000000000","updated":"2020-04-11 20:23:05.000000000","submitted":"2020-04-11 12:51:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717891-1586609477143-577aa814","meta_rev_id":"8a21f888b55bdcaede98ef90819026ab1ef887a8","_number":717891,"virtual_id_number":717891,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-11 12:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-11 09:39:59.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2020-04-11 09:12:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":1,"date":"2020-04-11 09:12:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 01:00:12.000000000","updated_by":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"reviewer":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"state":"REVIEWER"},{"updated":"2020-04-10 10:38:31.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-11 08:51:20.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-04-11 09:01:37.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-04-11 09:12:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-11 09:39:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-11 12:51:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c3278068b5e8eea8f826db58384bd426bbee9e4","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-07 01:41:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d574409f9088c426507529dc7146dc23fa07b6b","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-07 01:46:53.000000000","message":"Topic changed from dev to bug/1869319","accounts_in_message":[],"_revision_number":1},{"id":"ddd783ec31e938323bfa832cefee2e8b826f202f","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-07 01:47:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4cc077b2b2474330e082f92390598900f33be999","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-07 01:47:22.000000000","message":"Topic changed from dev to bug/1869319","accounts_in_message":[],"_revision_number":2},{"id":"7b9ff1e3e9dc397b3ca59c47d05b889a504324ce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 02:17:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/7151d63d03bc46ed96732ba2f2063ea8 : SUCCESS in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d65cb9f932b8af52bb4dd44cd067a41d91647b6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 02:30:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c3fb74d7fda642ffb74a9b60007cfadb : FAILURE in 43m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea05bc31a7f54d0ad9df807e4f87a3bfd4cb777c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 03:10:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1ad0863217b54b5383e2fdbc6f414321 : SUCCESS in 20m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/25ee4713ef45488fbe7d9df1189bd001 : SUCCESS in 26m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7ce5d51d7c6e4e9199a88aea80909d26 : SUCCESS in 1h 12m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0a1e81801c4d492ba3cad0983367d850 : SUCCESS in 1h 10m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/434f064a91254b10b84049a0a07a304c : SUCCESS in 1h 17m 51s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7b1612184f1d4ac48a339bc1415cec8c : SUCCESS in 44m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cb49205d347d45c9869d6f1ca67da63a : SUCCESS in 19m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4cbc22ee9569451daa51f63138c296e9 : SUCCESS in 16m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/444ad190dde64e0e9f63ef115d97e9b5 : SUCCESS in 1h 00m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e3e047b08624197ae2c6231520019a8 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d3a2026d24b45c49ef34e86a8283bdf : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/098e0c98978d42b08bc15221c2b2a5db : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba4e48f0fd5646ffb8476ce291f1e40f : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/707bd25ba6e34ff9875fc5da859a4591 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/979dd9c8ec6640b5b2ff306e4ca34416 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/96e5acd62b654401b35aa4840a9ac2f2 : SUCCESS in 21m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef5640cf34fd407ea5bbf4274acdd723 : SUCCESS in 23m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c4ba58d3c004a16a3203d060ea38efe : SUCCESS in 55m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1cc15e49869147c79c6da263d4d067d8 : SUCCESS in 56m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b931baf42e774cadb614ffc7520b7d8a : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":2},{"id":"86ac8fe55e0654731dad31a8776d14e7ca675306","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-07 06:18:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"85374689ff8e25b342c3fed5beb627d7187bba20","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2020-04-08 01:00:12.000000000","message":"Patch Set 2: Code-Review+1\n\nI just modified a little comment message.","accounts_in_message":[],"_revision_number":2},{"id":"d1c8dbc5e3f3e3ec64023a63fba18f6e35b7ad69","author":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"date":"2020-04-08 01:01:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3cf7135d7f33984810ba221c47df7f396e2718a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 01:06:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"52d81ff093fc318eed034d674b0a276957798043","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 01:38:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/895827b14ae24bd69c8fd007d17b38a9 : SUCCESS in 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b1fffc5455839482fba2d59f9c96eb9178d54073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 03:02:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9aefeadfdeca421896e62b5fda4f1690 : FAILURE in 55m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b8c520060133429583651096123f7e68 : SUCCESS in 48m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c7b6d716eb4e4ff9847fc376b104e7a2 : SUCCESS in 1h 40m 04s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3d7d9352c6ac4f63a15faa00ad7a65e5 : FAILURE in 22m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9dc385bdf6de455f9fe94301a6cc1d98 : SUCCESS in 1h 24m 30s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/219b703f925f40de9fc825177b068083 : SUCCESS in 45m 19s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/524246e86b3640d5a5d4e90e152c74b5 : SUCCESS in 21m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f9db983b6a7f4fbda94dc230880c66b8 : SUCCESS in 21m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2e0909750a5749629f639652f7cd3e0f : SUCCESS in 1h 00m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11f77fb9d6994adc98f729018f80a3d2 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/852f861aae7b4fca8825a4862547c4ad : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea157bb66997466a946e4e1af48e9bce : SUCCESS in 7m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6cb38a0097242d48e46e5f431650a10 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62f9602987414559a343cf1f268f9689 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce37f98c8d0417eb810f52bf827cdcc : SUCCESS in 5m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8b4c5689c17d4adf81af75b08ce6f41e : SUCCESS in 22m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/770c0048438347e5a08b39995f800468 : SUCCESS in 36m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5b042dceb9cb4a539bcb4a20874f627f : SUCCESS in 1h 25m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7de93c241711491289b24713904f67b7 : SUCCESS in 57m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ea890bff982446dd88be9b87e3c8b1f6 : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":3},{"id":"dc9964cb595aa35ab22d88f3ff14d047ed44de75","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 07:15:33.000000000","message":"Patch Set 3:\n\nHey, thanks for the patch. Is this component required since Ussuri? Or since earlier releases as well? Asking because we will need a proper reno for this.","accounts_in_message":[],"_revision_number":3},{"id":"1ccf9be403f951b70132ef5183619512040591b3","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-08 07:21:33.000000000","message":"Patch Set 3:\n\n\u003e Hey, thanks for the patch. Is this component required since Ussuri?\n \u003e Or since earlier releases as well? Asking because we will need a\n \u003e proper reno for this.\n\nThis is a must-have service. Without this service, data cannot be stored in the database.","accounts_in_message":[],"_revision_number":3},{"id":"fd064a9a008e4020982bd34ae431fb51ecfcf8c7","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-08 07:39:10.000000000","message":"Patch Set 3: Code-Review+1\n\n\u003e Hey, thanks for the patch. Is this component required since Ussuri?\n \u003e Or since earlier releases as well? Asking because we will need a\n \u003e proper reno for this.\n\nHi its needed for all version of vitrage","accounts_in_message":[],"_revision_number":3},{"id":"e43a465573dd80a41c68e56d684f096fe8a35125","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-09 02:34:26.000000000","message":"Patch Set 3: Code-Review+1\n\ngood catch","accounts_in_message":[],"_revision_number":3},{"id":"945bb0dd6619c54617b4f6c781f6ceac1a567fd8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-09 18:47:58.000000000","message":"Patch Set 3: Code-Review-1\n\nthanks, needs a bugfix reno, but otherwise lgtm","accounts_in_message":[],"_revision_number":3},{"id":"0545cb898fb678a5210e25cfc11ead6c8009e9a4","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-10 02:09:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"453bcc94ecd5f147bfc859f08415bdd86e44daa1","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-10 02:13:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e9e3ff6e602e9043e31ea0e043b85e4bb92ac5b6","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-10 02:34:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1ae74dfa263bf47ac1ef2fbff561790073db9bc9","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2020-04-10 02:47:53.000000000","message":"Patch Set 6:\n\n\u003e thanks, needs a bugfix reno, but otherwise lgtm\n\n Added follow your reply","accounts_in_message":[],"_revision_number":6},{"id":"f4dae8bd4d303d6287f19f59f16a074dd29cb2b0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-10 03:01:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/cf4e37b9c5e94a00af6bc122f48c30ca : SUCCESS in 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7d16ecc92c664496b99ed003309458975e568078","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 03:37:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d01ede1349c46a680cee3722dcbeaaa : SUCCESS in 47m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6defc58aeae4442797ccfd5e3f5d713d : FAILURE in 55m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2d44853afede3086d878505430b2288c95449f52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 04:19:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/15ab83878bf444eda21c4627d1f65b66 : SUCCESS in 25m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/75fa417a1a4a43d58c546abb07077c8e : SUCCESS in 27m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/94a32b2c5d9d4f89976a1377f20c27af : SUCCESS in 1h 14m 40s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9a27e1e5ad294422aac49a7da8a040d9 : FAILURE in 42m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7a2aa10b45344e98850443fc03906036 : SUCCESS in 1h 27m 09s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/5115c09d315946d2ba31778febd1c8c5 : SUCCESS in 42m 14s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/022fc59e632a456b88276064bb3cf905 : SUCCESS in 28m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d968d0a668224d71bd40a78e7c0a3801 : SUCCESS in 30m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/25215d5f12394832b6cffd716b32ea93 : SUCCESS in 59m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ddadb9b60774adba49adf09984ea15a : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1dd68fdd954166817480da578c5487 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a388c919d0094013ad1315863edd5156 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c0d3fa942d14cd988650a1e93dccd04 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/addd6f4b031d4b0b9881059d12663229 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef8cda68f372476eaecef8225192a1a3 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/521d57bb7ee84533aa8a94f5b16829ab : SUCCESS in 6m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/561b8a9af7e44d849098eeb4ab446703 : SUCCESS in 1h 38m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc90171a22b34ab2b05090549cf8a774 : SUCCESS in 23m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/be4f83e83e6f409cbc4476c03341427f : SUCCESS in 1h 39m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4cca83232cf440e198320b027444d0d1 : SUCCESS in 53m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a2e0579e605040db96651974de1bcec1 : SUCCESS in 1h 15m 20s","accounts_in_message":[],"_revision_number":6},{"id":"f27380209dcfd7aa39ba3c1485075940ce1f2a51","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-10 10:38:31.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e8479078d967559815f91accdf7754353e114817","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-10 10:43:06.000000000","message":"Topic changed from dev to bug/1869319","accounts_in_message":[],"_revision_number":6},{"id":"834424ded01d0d6ca312adf2ff10171279cda086","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-04-11 08:51:20.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":6},{"id":"1d087e08f917b2d487b3ff5599a0f7ad7b52bad6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-11 09:01:37.000000000","message":"Patch Set 6: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"9c156aff0617f3278e6137bc96b9a27dd43e6316","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-11 09:03:54.000000000","message":"Patch Set 6:\n\nlooking good overally, I will fix language and merge this, thank you","accounts_in_message":[],"_revision_number":6},{"id":"02571c9f11dcd34d727d9626e2b9d0198861ed78","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-11 09:12:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f35d132fa71a5a051b3f53af7623aefd7526697a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-11 09:12:54.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nafter it merges, we can try backporting to Train and Stein","accounts_in_message":[],"_revision_number":7},{"id":"9073fda620def87f47d84f43b541a66fb323d986","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-11 09:39:59.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/301bf1986f3c4c399a6690a2d9767155 : SUCCESS in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"75cecb355aa451d6e9f536ed481bbb48ec0b8126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 11:19:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bb266a6df2d4479aa8f01ae594e8fb0a : SUCCESS in 2h 03m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d81381f9d8774578a6834ee85a2a432f : SUCCESS in 2h 03m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a3fba7583a634257b6432bf4f6ae5430 : SUCCESS in 1h 17m 23s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b3c870757b514d26ab76aa357f51ee05 : FAILURE in 1h 45m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/81953fedce9f43cc8d2d2a91c891a971 : SUCCESS in 1h 35m 46s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2f89809a1ffd49939f014226e5f699e8 : FAILURE in 1h 57m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/76c7b1e67e8b4e07b5496c4b306d55b4 : SUCCESS in 17m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8dd0e54895754b459eb04b8df50d3116 : SUCCESS in 1h 36m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/48b8c080c6f54258a7be78f3e6fe0d74 : SUCCESS in 52m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bae14a1f45748c886c3e3b017d2f00c : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d661d1b4bce433a875e6644ee27d722 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d1ba2aeb34342e28cadaff19c461b01 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2e7d46582664fea88a0a0d4075e114e : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/503f1412150e41c78dc56110dd7cef1b : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b83169111da468bbd8d64f1742d9b13 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd4c8d16c60342edbec38a41d30f448e : SUCCESS in 4m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/56a691e6583f4c5eb3db9265c57b916c : SUCCESS in 28m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/88272a17c20f40f7a82df350d8e9785d : SUCCESS in 44m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/41e6c48ba7154c1c92054680fa1d62cf : SUCCESS in 55m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6a4c1891ab78472f9c08ccf4419a89b4 : SUCCESS in 1h 09m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f8d49262a844898bee7f36daad6feaa : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":7},{"id":"76413c7372b5c1e2873d02e03c224437ef09720b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 11:19:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b40fc7c14deb0c95c3074a3a1583f78f9372fdfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 12:51:16.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/68816b8189a64561890193529a86b044 : SUCCESS in 31m 07s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c6a0648260b74f3c97a45532b4571e31 : SUCCESS in 1h 27m 24s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6118ec260b8b42f68351b75662ffd8b5 : SUCCESS in 1h 30m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8a28191eecc4a1bb59838b0b911bb93 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37c4c92710a647a98f889478056b1442 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10c76ca934f844f28d1541c3992505b4 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/128f417e78c84dd09753dc9800cacf57 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c22e15696c84309977408304e4145cf : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c887f9e3d77d48aeb720870ef74d3f1f : SUCCESS in 4m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/93ff62e5758f402bbb1f0e1f56abcdbc : SUCCESS in 22m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b52442bb7aa4983b4925b7d8c7db921 : SUCCESS in 54m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/64b0423814af42eaa2bc0eac3a6c1ca5 : SUCCESS in 1h 15m 13s","accounts_in_message":[],"_revision_number":7},{"id":"185170b61017da8d8554ce757b82ac650563de54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 12:51:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"48b7e41269bd44799eab6b487f31b1980b0e7e83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 12:52:48.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0539155f54c0461da86201ef3faedf3b : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9ab95faa8bf4606adba68622955538c : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":7},{"id":"8e65697c230e13eb26e129810796ea0fb48ecfb5","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-04-11 13:27:29.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit c1fd2f9516408efe98eacc11f9508aa89cf40cdf","accounts_in_message":[],"_revision_number":7},{"id":"8f7e8b5782692720305f98a8aaef7bdeddf45cb8","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-04-11 13:28:43.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit f7cf4df5ada5553ce4cf2433ebf40276fe59a803","accounts_in_message":[],"_revision_number":7},{"id":"8a21f888b55bdcaede98ef90819026ab1ef887a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 20:23:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1625f346bd824149ba9eb176521506ff : SUCCESS in 46m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a3c00c667c344b73b70de56a0d742081 : FAILURE in 44m 48s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ad06096448a585a16ad73b610b0adc81b597ccec","revisions":{"56ebbba748a9b6f9441761c83e7db5cb57660828":{"kind":"REWORK","_number":1,"created":"2020-04-07 01:41:43.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/91/717891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/717891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/717891/1"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-07 01:30:42.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-07 01:40:32.000000000","tz":480},"subject":"Add vitrage-persistor dockerfile","message":"Add vitrage-persistor dockerfile\n\nKolla should contain vitrage-persistor image.\nSo i add vitrage-persistor dockerfile  in kolla.\n\nChange-Id: Id5e143636f9a81e7294b775f3d8b9134bee58054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56ebbba748a9b6f9441761c83e7db5cb57660828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56ebbba748a9b6f9441761c83e7db5cb57660828"}]},"branch":"refs/heads/master"},"31e2821de5f67f46ee6c3775dd5f7f17ffa2a80c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-07 01:47:01.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/91/717891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/717891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/717891/2"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-07 01:30:42.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-07 01:42:12.000000000","tz":480},"subject":"Add vitrage-persistor dockerfile","message":"Add vitrage-persistor dockerfile\n\nBecause vitrage-persistor service is an independent service running on\nan independent container. So Kolla should contain vitrage-persistor\nimage.And I add vitrage-persistor dockerfile in kolla.\n\nChange-Id: Id5e143636f9a81e7294b775f3d8b9134bee58054\nCloses-Bug: #1869319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31e2821de5f67f46ee6c3775dd5f7f17ffa2a80c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31e2821de5f67f46ee6c3775dd5f7f17ffa2a80c"}]},"branch":"refs/heads/master"},"3baba21af283e0606c381113cc1f62615c6c0c85":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-08 01:01:13.000000000","uploader":{"_account_id":30562,"name":"Ziyu Bai","email":"baiziyu@inspur.com","username":"baiziyu-inspur"},"ref":"refs/changes/91/717891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/717891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/717891/3"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-07 01:30:42.000000000","tz":480},"committer":{"name":"Ziyu Bai","email":"baiziyu@inspur.com","date":"2020-04-08 00:58:39.000000000","tz":0},"subject":"Add vitrage-persistor dockerfile","message":"Add vitrage-persistor dockerfile\n\nBecause vitrage-persistor service is an independent service running on\nan independent container, kolla should contain vitrage-persistor\nimage. This commit added vitrage-persistor dockerfile into kolla.\n\nChange-Id: Id5e143636f9a81e7294b775f3d8b9134bee58054\nCloses-Bug: #1869319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3baba21af283e0606c381113cc1f62615c6c0c85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3baba21af283e0606c381113cc1f62615c6c0c85"}]},"branch":"refs/heads/master"},"fc0a27e8c00aad01eff15071e0aaadc516d1b66e":{"kind":"REWORK","_number":4,"created":"2020-04-10 02:09:59.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/91/717891/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/717891/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/717891/4"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-07 01:30:42.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-10 02:09:21.000000000","tz":480},"subject":"Add vitrage-persistor dockerfile","message":"Add vitrage-persistor dockerfile\n\nBecause vitrage-persistor service is an independent service running on\nan independent container. So Kolla should contain vitrage-persistor\nimage.And I add vitrage-persistor dockerfile in kolla.\n\nChange-Id: Id5e143636f9a81e7294b775f3d8b9134bee58054\nCloses-Bug: #1869319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc0a27e8c00aad01eff15071e0aaadc516d1b66e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc0a27e8c00aad01eff15071e0aaadc516d1b66e"}]},"branch":"refs/heads/master"},"e40e4d128a6f968a7cc383c755ef4fa32bfc3390":{"kind":"REWORK","_number":5,"created":"2020-04-10 02:13:37.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/91/717891/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/717891/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/717891/5"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-07 01:30:42.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-10 02:12:31.000000000","tz":480},"subject":"Add vitrage-persistor dockerfile","message":"Add vitrage-persistor dockerfile\n\nBecause vitrage-persistor service is an independent service running on\nan independent container. So Kolla should contain vitrage-persistor\nimage.And I add vitrage-persistor dockerfile in kolla.\n\nChange-Id: Id5e143636f9a81e7294b775f3d8b9134bee58054\nCloses-Bug: #1869319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e40e4d128a6f968a7cc383c755ef4fa32bfc3390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e40e4d128a6f968a7cc383c755ef4fa32bfc3390"}]},"branch":"refs/heads/master"},"5b72029f5f7a611f1fa6fa4577ad40e535481c86":{"kind":"REWORK","_number":6,"created":"2020-04-10 02:34:35.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/91/717891/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/717891/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/717891/6"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-07 01:30:42.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-10 02:34:16.000000000","tz":480},"subject":"Add vitrage-persistor dockerfile","message":"Add vitrage-persistor dockerfile\n\nBecause vitrage-persistor service is an independent service running on\nan independent container. So Kolla should contain vitrage-persistor\nimage.And I add vitrage-persistor dockerfile in kolla.\n\nChange-Id: Id5e143636f9a81e7294b775f3d8b9134bee58054\nCloses-Bug: #1869319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b72029f5f7a611f1fa6fa4577ad40e535481c86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b72029f5f7a611f1fa6fa4577ad40e535481c86"}]},"branch":"refs/heads/master"},"ad06096448a585a16ad73b610b0adc81b597ccec":{"kind":"REWORK","_number":7,"created":"2020-04-11 09:12:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/91/717891/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/717891/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/717891/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/717891/7"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2020-04-07 01:30:42.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-11 09:10:36.000000000","tz":120},"subject":"Add vitrage-persistor image","message":"Add vitrage-persistor image\n\nKolla was missing vitrage-persistor image required by Vitrage\nfor data storage.\n\nChange-Id: Id5e143636f9a81e7294b775f3d8b9134bee58054\nCloses-Bug: #1869319\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad06096448a585a16ad73b610b0adc81b597ccec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad06096448a585a16ad73b610b0adc81b597ccec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
