)]}'
{"id":"openstack%2Fkolla~717990","triplet_id":"openstack%2Fkolla~stable%2Fstein~I498aa50a0d59d090486d60c2b34b4f2fd077d401","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I498aa50a0d59d090486d60c2b34b4f2fd077d401","subject":"Make deploy jobs voting again","status":"MERGED","created":"2020-04-07 08:56:17.000000000","updated":"2020-04-07 12:45:53.000000000","submitted":"2020-04-07 12:39:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717990-1586263153768-86ddf8e2","meta_rev_id":"d73b4bce2a1f58fff41b41818b136f17ab124c93","_number":717990,"virtual_id_number":717990,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 12:39:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-07 11:20:59.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-07 11:20:59.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 12:39:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78ee97cc33b95b29a4b35e8f3fd5502877ac2128","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-07 08:56:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"828b85c66ca01f9fda0ec165fac6c4c9943800c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 11:09:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/822fb31ff5b24e1eaa236291ee574fe6 : SUCCESS in 1h 09m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c523d5ad44044f57a279c3f95573870d : SUCCESS in 55m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/90d8a54bdfa84158b5141e021b8e774c : SUCCESS in 1h 26m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ecb0c2a5dea846f9a8c2ec36c93110ec : SUCCESS in 1h 19m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a3c95ba995c94f058a65955659180c76 : SUCCESS in 1h 36m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5dd3695a722e4f6cbbf282fc6e5f270b : TIMED_OUT in 2h 09m 13s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4230e487dcc743878463499bf6d599a4 : FAILURE in 50m 01s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/41702e42aeae4937a1e109cdda867213 : SUCCESS in 50m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e62ca8de9b264ac98f36c4a69a0830a6 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e25239383240407e85f71afb394b43ab : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a924473f896e4fa986208a812ea3a390 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e3e8aad6699482594cc4ee6d24a5b7d : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd7a5ab0796d4de69aa35d36d1604211 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29d9443f40964ec8b301409e2805d9b3 : SUCCESS in 4m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/234784900b4943dcb285ca4d47f39d44 : SUCCESS in 21m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcf7a2667d97417c9ffbc8e32f429156 : SUCCESS in 34m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/63fb2925c7944c89a0434f8de9ba3057 : SUCCESS in 1h 11m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9de23af5648408c895807e97fe4a172 : FAILURE in 42m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aba80e800f944402aba29d7bd8e0c342 : SUCCESS in 1h 07m 09s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"2b32d56375655f8a1a89b079e67b46519bba7bf4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-07 11:20:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5f87dc6b0e016f5a450638cc893c380148d362de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 11:21:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4eb548b08eb7e1643841548017bfdb998b18cbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 12:39:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/263f890047754c41920d4521240a3e46 : SUCCESS in 55m 00s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/b469fc81c2254619932c6a704d6e072d : SUCCESS in 52m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6d2e93e268da4ee9b0c369aec8f130a6 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba7e52b08f9493d89ba74356e7a245a : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3af9c699b3824d4e8af38ff0047ed9b6 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be9b0b78c62746d9abf2f40221cb47b6 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b0e83a91ee14837999f1aa5d2cb042a : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeed6f171d584a7ba85acce101df1d57 : SUCCESS in 4m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d905e04545a4422b2a99fe3ac60f5a0 : SUCCESS in 28m 36s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"7bdc96d8e419cda93323305303eb61c97d2cb5d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 12:39:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d73b4bce2a1f58fff41b41818b136f17ab124c93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 12:45:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad756ba5becd4691a8b1236be42f1c59 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc0967b88212c2626c1469e74223515677785c6b","revisions":{"fc0967b88212c2626c1469e74223515677785c6b":{"kind":"REWORK","_number":1,"created":"2020-04-07 08:56:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/90/717990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/717990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/717990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/717990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/717990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/717990/1"}}},"commit":{"parents":[{"commit":"4696fc0f322bdc4327b0ac4ec58f2ddc9d014ec1","subject":"Fix invalid test in cinder data migration script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4696fc0f322bdc4327b0ac4ec58f2ddc9d014ec1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-06 11:51:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-07 08:55:48.000000000","tz":60},"subject":"Make deploy jobs voting again","message":"Make deploy jobs voting again\n\nThese were temporarily made non-voting in\nI8c2da6437333210a7926f2deb926ab5d1c93c3c6 to allow us to fix CI.\n\nChange-Id: I498aa50a0d59d090486d60c2b34b4f2fd077d401\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc0967b88212c2626c1469e74223515677785c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc0967b88212c2626c1469e74223515677785c6b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
