)]}'
{"id":"openstack%2Fkolla~718081","triplet_id":"openstack%2Fkolla~master~I8ed1b6c1d0985b2a73206bd9249a5664cd80c912","project":"openstack/kolla","branch":"master","topic":"bug/1870455","hashtags":[],"change_id":"I8ed1b6c1d0985b2a73206bd9249a5664cd80c912","subject":"nova-compute: add ndctl to expose NVDIMMs to guests","status":"MERGED","created":"2020-04-07 12:20:38.000000000","updated":"2020-04-15 15:34:27.000000000","submitted":"2020-04-15 15:30:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718081-1586964652676-61dee02a","meta_rev_id":"2e0361ad4f30bc1e4aedf2cc6800447e3506008d","_number":718081,"virtual_id_number":718081,"owner":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 15:30:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-14 15:09:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-04-15 12:33:40.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2020-04-14 14:30:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2020-04-15 13:13:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":1,"date":"2020-04-15 13:13:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-04-14 20:30:50.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":0,"date":"2020-04-14 20:31:13.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-04-14 20:30:43.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 12:21:14.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2020-04-07 14:47:29.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-04-09 18:44:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-14 14:30:47.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-04-14 15:09:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-14 20:30:43.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-14 20:30:50.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2020-04-14 20:31:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-15 12:33:40.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-15 13:13:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-15 15:30:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83230abb2a79bbcc025fbc7c36a1976febd2f055","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-07 12:20:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54357f857538d25cc5dbd38b4162dda7c0b43a66","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-07 12:23:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)\n\ncommit message should be like this:\n\nwhere: what and why\n\nlonger description","accounts_in_message":[],"_revision_number":1},{"id":"0da1a5b209f0ede91c87020714fdfa9eef8f17cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 12:24:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e2326d963f125938ad1d942616024bc019fd3e10","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 12:55:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/3d161e425b894a7085418d09b8d14cdf : SUCCESS in 33m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9537cc00d6b8b70b18c7e80940bbafbe927b15f","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-07 13:13:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"46ed471b8a47ece0c97a3af6960591aa9c037730","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-07 13:14:17.000000000","message":"Topic set to bug/1870455","accounts_in_message":[],"_revision_number":2},{"id":"f57cdd3406930cef5fd021e588bdee32fe28fbc3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-07 13:21:05.000000000","message":"Patch Set 2: Code-Review+1\n\nHave to check how it goes on non-x86 in centos/ubuntu.\n\nDebian has ndctl package for all architectures.","accounts_in_message":[],"_revision_number":2},{"id":"e6c06ba8666ecf27f6560f32757d0b54e751bb18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 13:32:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"07248a38b0156067bae7a2f5714d0e9094f7a8d0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 13:41:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/275e424065444379a3893a12845e8bb9 : SUCCESS in 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34f3270cc7e96b4d62a81a3276314371b83b4a73","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-04-07 14:47:29.000000000","message":"Patch Set 2: Code-Review+1\n\nhttps://logserver.rdoproject.org/81/718081/2/openstack-check/tripleo-build-containers-centos-8-master/275e424/logs/buildah-builds/kolla-builds/5b95e488-0338-4b3c-b645-db37d11c1fdd/docker/nova/nova-compute/Dockerfile.j2","accounts_in_message":[],"_revision_number":2},{"id":"a3f7dbb13b429b4c3040f9522c1b7f588bbbb0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 14:53:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a914d1e523624781b02ddd3f4f73056d : SUCCESS in 1h 01m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/95a8f6c7491c49cfb71440b8a317a5aa : SUCCESS in 30m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ffb63de5f43344308a3e568e87bee9d6 : SUCCESS in 1h 15m 43s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4b573022069c41a7abb90df435ede934 : SUCCESS in 1h 11m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cb4311bc2f184c31abd0fea81ec50798 : SUCCESS in 1h 23m 12s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e8665c500441462c87c11060440ca036 : SUCCESS in 46m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/31b809f807864cf8bfe58e1fcea22409 : SUCCESS in 22m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/88fd40a2fd4f4c56a5c0cb706a50d85d : SUCCESS in 26m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/60c2e2352d2e42f782605e96f7928b2d : SUCCESS in 1h 08m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed83acf15c474d97824a13aa5ca409c3 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8cf011c0123462080ad21e2290bd43b : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0f74dec506045a79f5e820905bc1193 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff453c71c1ed472c94c2a76fe9e5b375 : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86d7ce4c472f4993929689b83af59d85 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c801411d4ad747e8bad1c4ced412ed54 : SUCCESS in 7m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e574f4bb0e5246a4ad67ec1bfa45bdb3 : SUCCESS in 24m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca6c572e078a4d5da316203f8f4c6ee9 : SUCCESS in 28m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/16601c8de2404d2c8029e7c8b6dddb37 : SUCCESS in 56m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/81bf991969354a7dae7c2857492460e0 : SUCCESS in 57m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e4c95e92746540e8b5afa1de56759159 : SUCCESS in 1h 11m 56s","accounts_in_message":[],"_revision_number":2},{"id":"77054f5b3020eeee674ec709da812062e555ab10","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-07 20:20:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7f55e604f69e5ca357af60b8dc1b682707ed3097","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:26:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"99e7e941b94fd785047e003f8c034cddfaf35849","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-07 20:44:41.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"883bd961d8c0ac18757e9896a2dbaeb73b83eef5","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-07 20:54:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aab717311d6153fb2709d68187f74f1047d62a0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:57:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7db0dbccbd4c3d695808a4c9fb6d1b974f9f75ae","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-07 21:05:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"91874202b19a469285017428364556b368933d0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 21:09:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a184cbaa284da6d2954285ac2ee8971efe05f3fe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 21:16:25.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/5230f9bbb2954ab7ba02cf2f1300ab6e : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd13cc06d48f2c108b0916318460beb2e0ca4927","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-07 21:22:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b5a3e1f9b07e9f915de5b684e8fc5385bea7423c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 21:26:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9a922b3f806a98d1418c0b9be2e373bcd85603ab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 21:35:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e753931c355e42058863160a67863dbe : FAILURE in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"85fcd4f5806e7a394a49e0bbc0044eb0dd6cebd5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-07 21:37:36.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b4974b4f83dfef5049b88f9e20a1da554c944a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 21:49:54.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/337750f82f584811bcd3a1003aa186d2 : FAILURE in 7m 37s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/85619beef16d430db90b792b0ccd4365 : FAILURE in 7m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/30f68389bd9f4f30982c9a00584f7747 : FAILURE in 13m 12s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/992edf05af294274982d2188787f8896 : FAILURE in 15m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/12a611375e3f454dbf2e0a5154d8a755 : FAILURE in 10m 42s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/5809ffc7b2274c6bb796ea74a61be271 : FAILURE in 23m 07s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/56b69619874f47f99a4488af599f8ee6 : FAILURE in 4m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/41f5488996b64b5ea7848286d7d11111 : FAILURE in 6m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/12cad56b64d14af6be44b250e0062461 : FAILURE in 11m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ba8ff2bb53e45f3bb7418c67f662af4 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0888be16487648b298c8beaa3dd301d0 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5af8dfd69e6418981cd76b3c3e3314e : SUCCESS in 8m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8376c90d132f40a5a01f8d6dbcbbfb58 : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed7e2a8cc4e040909b3f1bef761193cb : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20acd9ff541d4ef99eedc865521e553a : SUCCESS in 8m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c7cd0dc49dfe43278dda4d22682c8c4d : FAILURE in 5m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eff30480cafc40d78044876f07937aab : FAILURE in 6m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5593889ac6c498b94a3beb61c7864e1 : FAILURE in 10m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e75c0479d95c4f1c9174abac36a70b1f : FAILURE in 11m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c7009298190940869e8701e22a09cf3b : FAILURE in 11m 22s","accounts_in_message":[],"_revision_number":6},{"id":"4f598eccff4bd17b1d58412f7f1e5f33990497b2","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-07 22:23:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0fb5f7ee4ffea8919824c55d1be09420952526bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 22:49:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"599a20f0e7d46c187fb493e5ac5f251ac4900275","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 22:57:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/25be6bced05146c2a53db02b8c15e039 : SUCCESS in 31m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7590adab6593f9530ef97a83b8b14e5291d777f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 23:54:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8817d8c596ce49eea9e15ff0fc0f9ddc : FAILURE in 23m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2f05f664ce804ab2974fba9985159040 : SUCCESS in 28m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f22cac159aba4e94916e867f863ac88d : SUCCESS in 1h 18m 24s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/710e619db72642a187aa72e150c288dd : FAILURE in 20m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/42e92397c9c540e18a26d42b2c77b938 : SUCCESS in 1h 17m 03s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/041910b752114063803949a0230a4f9a : SUCCESS in 1h 00m 19s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3a521a11c2294990a7c817f3e63c827e : SUCCESS in 21m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f4b0fdffe55649e3b60781264206ac5b : SUCCESS in 28m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e847256978054744ae6c726633411d05 : SUCCESS in 1h 02m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f32a2958c9642c4a87054271e157f05 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7c5cbe416e649729e7c33d61a3a908c : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a73d57ae99c4754ab8cc92d92b2913f : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/05f3c42d81ed45e197f87b66b500f790 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/076bc5d1c20c42beb12481b8f6564284 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46299fa196454f269a862ba2a1de4b7e : SUCCESS in 7m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/527ffd60988c45bab45c923c242847ff : SUCCESS in 23m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ee070455b06477498c1744b7dfdd25f : SUCCESS in 26m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/af7ee8832d1a4a86a926906f23a03806 : SUCCESS in 54m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f38f1b54ff5f4da6875051a8c2fea666 : SUCCESS in 57m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/871fd34dc0664af99a442c9789dbf119 : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":7},{"id":"fe9db2ae9c92c4b8f2c21fe6def1242fdd935c20","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 06:50:35.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"17aa13de1fc508a211512900f6eb770e53a9b2df","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-08 15:15:54.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a0c0fc504e93aac221db40a47515a3772b00b587","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-08 15:16:16.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"43106411bee4b39006462576caa7736bc117d6a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 16:07:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/af3d66fadbc74a1dac270e2771a481d2 : FAILURE in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"47c8385927063dd2c24f4c7831a56b6582204afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 16:51:58.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/afc39f69e1524bfeaf67d0550c995d21 : SUCCESS in 30m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2080e0eaddcc485eb3bbca39afdd3748 : SUCCESS in 39m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/677d4cd8d07c4d9c9d24a651d0118eb4 : SUCCESS in 1h 20m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/edd18c8070da4e4581d366b19bc848cb : SUCCESS in 1h 19m 39s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3ffeb429e0a9417fb87095c393e25e25 : SUCCESS in 1h 23m 34s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/8e6af44e5ea34a7d854a5c15d4a4e34a : SUCCESS in 44m 37s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5704846e749245b1a02ad91af0ccfbd2 : SUCCESS in 28m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/85d11129449b4dd0b52b8b430b5a859d : SUCCESS in 25m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/09af28a95e98421f9071586611d1348e : SUCCESS in 1h 01m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f18ebf7345f4b5b982467ed08977af5 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83ec8c835adb43d994be349bef9e2d67 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3660cf9115748ff88d84c4fd3fafb8b : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f7dc1f37d7640dcb0452c310ca6f24f : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29cc1a5f823f4dce9a231b08f7f3d6f7 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5271eb5a45647988e5ff622d238612a : SUCCESS in 9m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4bc38edb547c49a49895264ba2a6371c : FAILURE in 29m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4176f342923d45a39a3ab0a9d18754ce : FAILURE in 30m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1aedb08b603f4f379501c1d929a3eb76 : SUCCESS in 1h 08m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/432e27dffe0e4f9f863b1ac1c056123b : SUCCESS in 1h 08m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eaeb49a3bb3e4b409e1f834611aa38ef : SUCCESS in 1h 17m 17s","accounts_in_message":[],"_revision_number":8},{"id":"3f581a74c19f88f981780326551e760e832c6ca5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 18:02:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/7ea4c9f6a2b6406ea5c5ab2331c1d021 : SUCCESS in 28m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a8431738c24ca6e54102f070290143569a5b5c4c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-09 05:28:39.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"a8623f1fcf03a7f7c931acb46c0366d634a54644","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-09 06:05:13.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/f0ae8b57a4334edb99c45120ae7c6a5d : SUCCESS in 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"13a84e34795021bf2ba3035a517bd17b4c13b408","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 06:17:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b20c5f3fdf1445398632e2f6048860a : FAILURE in 48m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1cbf32f5d90556a9f9e11775ebb68cb7de6e5805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 07:05:20.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b4b605b801414ab680ea37168db9898b : SUCCESS in 25m 30s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4eeeb6d4c7244394824d4dc78e078112 : SUCCESS in 34m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/22a86b52e3234a19bcd0eda68ba82a5c : SUCCESS in 1h 22m 28s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9598956ac8174444aada16bb1bae6f1c : SUCCESS in 1h 15m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e9207af68f9e400681eecf42ce7d17dd : SUCCESS in 1h 27m 22s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/391bfcf3f6cc4cba8233c1ca3485f941 : SUCCESS in 45m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f0439f5998e74e209f3445142a374d50 : SUCCESS in 22m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/036105dbd39147f3a0a728c122b21091 : SUCCESS in 26m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0a573dfc031041558019852b92c3ad30 : SUCCESS in 1h 00m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d216cbbe7abf45268f2673b1ca01eebb : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7d3cf35653b4d2f9b298f6844ca00b8 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0609db1e8da43789bac6fbadae6ffc9 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/626393b2977247dc8d7f229688419cdf : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a1a208677374fde8fd1f11e0f5495e5 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77ca0dd62bd44328e89db3afdc98295 : SUCCESS in 5m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d67ca500fee46e7a3191325c082eb70 : SUCCESS in 21m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c494a46c61046a8b4e45ac9487bba37 : SUCCESS in 26m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca1a1b4366154db7bc76ff6fd87dd673 : SUCCESS in 1h 20m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aa875d524ae143729f42df1faf206b90 : SUCCESS in 1h 03m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c07f8f76b15e4fbcbfe4974c226e1a6c : SUCCESS in 1h 31m 15s","accounts_in_message":[],"_revision_number":9},{"id":"a91d6805311849fe967ad84a66f0b239e0dec19e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-09 18:44:02.000000000","message":"Patch Set 9: Code-Review-1\n\nthanks, just needs a reno","accounts_in_message":[],"_revision_number":9},{"id":"2259fd8d0ccd1b9fdfa85b39c39063fc3acf706e","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-13 20:42:37.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7c38471163b671d7541e680b32898a9c28b64dc7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-13 21:13:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/506e415369824571a58a1a3e7617197b : SUCCESS in 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7d58fafafd77d263f20d06da677a924e4dc9e8be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 21:33:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea3f21defd2340509adacf211f2fd1cd : SUCCESS in 50m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/69c0119acc244af281859acd88bbbef3 : FAILURE in 48m 52s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0d339b0632ad1ea05631e880922cf9c16dd42673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 23:20:19.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/118848e718854ed3ada595adb8fe5f08 : SUCCESS in 39m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/116c8d06ad234c50ae1361334951f303 : SUCCESS in 1h 20m 22s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f8aba0e19eab4cf2a2dacc88d2708878 : SUCCESS in 1h 14m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/224260c4e05d40689587b8f6343c701e : SUCCESS in 1h 04m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/79e17a54b680401baaee18d76c2b588e : TIMED_OUT in 2h 35m 24s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/4904697832c64547a6c0eba2583cedd0 : SUCCESS in 47m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe776db45fa9405d94c95542010192c1 : SUCCESS in 20m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/07735c662ace404490bfe8c1aa20b94a : SUCCESS in 19m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2878df0a23484278bb392b4cf74a5ab7 : TIMED_OUT in 2h 04m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/289b6d21cbcc4bcaa3e6c36671861271 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6941db9907e643058ddf96831c4582d4 : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56f792f0fbc44a729be401f3416d27fb : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d6cb82f08944229976e5d6804dc82e3 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cea5696693d47fe936f3fcd4333be0a : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83f48a4a2a7d4911b130b713fc6f0bc6 : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9be29875b8364f609dfead03afce8693 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b98d6a2fe48e472fa547ab1dbf406f98 : SUCCESS in 28m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2575ff2a9c91439aba573ced6ffb985c : SUCCESS in 2h 14m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/01651d1e54b546249b2c6a899764b391 : SUCCESS in 55m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/23c901f31d4a47b4a9dd6eea9b080c0f : TIMED_OUT in 2h 03m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/feadb2a57b814b37b1f3d234a4521f3f : SUCCESS in 2h 30m 24s","accounts_in_message":[],"_revision_number":10},{"id":"ce7f55f53f92c121741573ab9822768512507159","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-04-14 13:46:24.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"2c51096bb98ce906f5d596cb10807d5dd3209a4c","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-04-14 14:30:47.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"a8b9a25e38cbd56837fa226c3536802898689555","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-14 15:09:13.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/3dbb0d186fc5402da1de0e6c47bc4d09 : SUCCESS in 55m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c4bd8672e00aa2916ffc09c43728aae6ddc8033b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 17:07:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/59bcdd9e9b7d4cc4b2825a6109bba8aa : SUCCESS in 25m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a64c04e015304f7eac76698f433c9e58 : SUCCESS in 1h 20m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/45c9e72ad40c48f68d6a1a3065073391 : SUCCESS in 1h 23m 37s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d11967abf2a64826a64d727839f5c42d : SUCCESS in 1h 18m 03s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/92d6d09e153f4185a9ca42602483e3d7 : SUCCESS in 1h 20m 22s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/8e3d05bb796c49fab6462999936a9df5 : SUCCESS in 43m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e4c2baff28754332b98fe7adf6ee103d : SUCCESS in 20m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/258b4b05d48a4ccd8e88bdf0a811bc4d : SUCCESS in 53m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e1b69366fe7d4f6f8c9229ecd13bb5c7 : SUCCESS in 1h 22m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5dee1ad26064488a1343f30808e83cc : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd28d99d76140df98431f519b78a418 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e01be536a25647859436b45a17d2cfed : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c31281d0adf460ab16a201a73589cba : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44ae65a90d7c4a068f156866dfc4ba96 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0aada8102d840329535f6a42fdfe16d : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76989f7105d44bbcb3ff48c6b1a722b4 : SUCCESS in 5m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6fc34134fc5d43d2a39dc05d76bda7b8 : SUCCESS in 35m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/49910cbebf3b491fba0859133f1d4054 : SUCCESS in 36m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d28507663bc545f6a788a972875e7989 : SUCCESS in 1h 09m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/84f3c0d1c09440d39f6b66b3dea85639 : SUCCESS in 1h 04m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/46c00a39517249a1b57f755da960f0c2 : SUCCESS in 1h 41m 16s","accounts_in_message":[],"_revision_number":11},{"id":"521c58117c9c8594ad3f2b59a0da39ead67b4949","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 18:09:03.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/610e6f758ecc4474b5735eaaba6049ca : SUCCESS in 51m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b84a5744bd334e1d8a5c999a0748fecf : FAILURE in 49m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"bdd12ad4f0681bf4638f6a3ae79e8a0b729d801d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-15 12:33:40.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"8cc84ee37657700b1de88b0605b35a9954bf8b44","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 13:13:52.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"0930070af1c9882a0693e792d846953db581a2bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 13:14:11.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"26a97d616aa6e2973441c0dfcc51769ed5f28dd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 15:30:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"c49904de9cf510b087fd07c44a2088a02ae64011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 15:30:52.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3b4af6de05924fd4be42daeff56dc005 : SUCCESS in 35m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7078866b929b41adafb89adef0c0564c : SUCCESS in 1h 34m 51s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bebdff2f91cb45a3a10dfbb7e3f53ab8 : SUCCESS in 2h 14m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b2683e7db2c4fa6a52672fd2cd312cb : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f401239050440287989488b3fd3a4f : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e7563a2968340c29a384e76718a1ee4 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/591465020759404980edb10126bfac9e : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ffb24ceeeee464f85f42b94b66c6697 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb6dc85bf8d9459ea15a1e847c43fe76 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/011493a917da49cea511b44759da8d7f : SUCCESS in 33m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6ca9fe38a01b44dfaa64a77c44855b88 : SUCCESS in 1h 03m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bfcf698f11c144ee9dd23167de46bc51 : SUCCESS in 1h 09m 04s","accounts_in_message":[],"_revision_number":11},{"id":"2e0361ad4f30bc1e4aedf2cc6800447e3506008d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 15:34:27.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace6a78af8194b3d96e69df16c54fa53 : SUCCESS in 2m 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/873a68223d594290ad58fdb7618749b7 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"f9ea23d940ec4ce35d503ac62f202daaa9db502d","revisions":{"2f893eeecfe955e6ab17cc7adf04f2572633d367":{"kind":"REWORK","_number":1,"created":"2020-04-07 12:20:38.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/81/718081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/1"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:20:35.000000000","tz":120},"subject":"Add `ndctl` package","message":"Add `ndctl` package\n\nAdding `ndctl` package to nova-compute container.\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for exposing PMEM as vPMEM to guests.\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f893eeecfe955e6ab17cc7adf04f2572633d367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f893eeecfe955e6ab17cc7adf04f2572633d367"}]},"branch":"refs/heads/master"},"629daaa82f7ab568e4f38585a9dbd246c636cbb3":{"kind":"REWORK","_number":2,"created":"2020-04-07 13:13:58.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/81/718081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/2"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 13:12:35.000000000","tz":120},"subject":"nova-compute: add ndctl to expose NVDIMMs to guests","message":"nova-compute: add ndctl to expose NVDIMMs to guests\n\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for attaching PMEM(persistent memory) namespaces to guests [0].\n\n[0]: https://docs.openstack.org/nova/latest/admin/virtual-persistent-memory.html\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/629daaa82f7ab568e4f38585a9dbd246c636cbb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/629daaa82f7ab568e4f38585a9dbd246c636cbb3"}]},"branch":"refs/heads/master"},"ae4daa73b9c76e1cf1f5d0e7353601991365d0d8":{"kind":"REWORK","_number":3,"created":"2020-04-07 20:20:40.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/81/718081/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/3"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 20:20:27.000000000","tz":120},"subject":"nova-compute: add ndctl to expose NVDIMMs to guests","message":"nova-compute: add ndctl to expose NVDIMMs to guests\n\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for attaching PMEM(persistent memory) namespaces to guests [0].\n\n[0]: https://docs.openstack.org/nova/latest/admin/virtual-persistent-memory.html\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae4daa73b9c76e1cf1f5d0e7353601991365d0d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae4daa73b9c76e1cf1f5d0e7353601991365d0d8"}]},"branch":"refs/heads/master"},"cdc109b1d08670038892b180755a33c4ac9a79b0":{"kind":"REWORK","_number":4,"created":"2020-04-07 20:54:51.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/81/718081/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/4"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 20:54:35.000000000","tz":120},"subject":"nova-compute: add ndctl to expose NVDIMMs to guests","message":"nova-compute: add ndctl to expose NVDIMMs to guests\n\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for attaching PMEM(persistent memory) namespaces to guests [0].\n\n[0]: https://docs.openstack.org/nova/latest/admin/virtual-persistent-memory.html\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdc109b1d08670038892b180755a33c4ac9a79b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdc109b1d08670038892b180755a33c4ac9a79b0"}]},"branch":"refs/heads/master"},"c6c8704e7c07df50b977aa11ad36a5979a1755c3":{"kind":"REWORK","_number":5,"created":"2020-04-07 21:05:27.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/81/718081/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/5"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 21:05:13.000000000","tz":120},"subject":"nova-compute: add ndctl to expose NVDIMMs to guests","message":"nova-compute: add ndctl to expose NVDIMMs to guests\n\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for attaching PMEM(persistent memory) namespaces to guests [0].\n\n[0]: https://docs.openstack.org/nova/latest/admin/virtual-persistent-memory.html\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6c8704e7c07df50b977aa11ad36a5979a1755c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6c8704e7c07df50b977aa11ad36a5979a1755c3"}]},"branch":"refs/heads/master"},"6b9a7936a252ee8ade3923e8c8fee67619e37e7f":{"kind":"REWORK","_number":6,"created":"2020-04-07 21:22:49.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/81/718081/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/6"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 21:22:36.000000000","tz":120},"subject":"nova-compute: add ndctl to expose NVDIMMs to guests","message":"nova-compute: add ndctl to expose NVDIMMs to guests\n\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for attaching PMEM(persistent memory) namespaces to guests [0].\n\n[0]: https://docs.openstack.org/nova/latest/admin/virtual-persistent-memory.html\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b9a7936a252ee8ade3923e8c8fee67619e37e7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b9a7936a252ee8ade3923e8c8fee67619e37e7f"}]},"branch":"refs/heads/master"},"728415cfbf67006f77e71a972977e7eb75f159f9":{"kind":"REWORK","_number":7,"created":"2020-04-07 22:23:25.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/81/718081/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/7"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 22:23:10.000000000","tz":120},"subject":"nova-compute: add ndctl to expose NVDIMMs to guests","message":"nova-compute: add ndctl to expose NVDIMMs to guests\n\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for attaching PMEM(persistent memory) namespaces to guests [0].\n\n[0]: https://docs.openstack.org/nova/latest/admin/virtual-persistent-memory.html\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/728415cfbf67006f77e71a972977e7eb75f159f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/728415cfbf67006f77e71a972977e7eb75f159f9"}]},"branch":"refs/heads/master"},"928e1a9f3ec4adfc3f5693dda0b8115dee5e14e2":{"kind":"REWORK","_number":8,"created":"2020-04-08 15:15:54.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/81/718081/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/8"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-08 15:15:41.000000000","tz":120},"subject":"nova-compute: add ndctl to expose NVDIMMs to guests","message":"nova-compute: add ndctl to expose NVDIMMs to guests\n\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for attaching PMEM(persistent memory) namespaces to guests [0].\n\n[0]: https://docs.openstack.org/nova/latest/admin/virtual-persistent-memory.html\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/928e1a9f3ec4adfc3f5693dda0b8115dee5e14e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/928e1a9f3ec4adfc3f5693dda0b8115dee5e14e2"}]},"branch":"refs/heads/master"},"702a3553559c003c23313f3548613a0a25ede241":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-04-09 05:28:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/81/718081/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/9"}}},"commit":{"parents":[{"commit":"5d98d3cae09b89c02f306fa52636d78b41fbd03e","subject":"Merge \"openstack-base: PyNaCl needs \u0027make\u0027 to build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d98d3cae09b89c02f306fa52636d78b41fbd03e"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-09 05:28:39.000000000","tz":0},"subject":"nova-compute: add ndctl to expose NVDIMMs to guests","message":"nova-compute: add ndctl to expose NVDIMMs to guests\n\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for attaching PMEM(persistent memory) namespaces to guests [0].\n\n[0]: https://docs.openstack.org/nova/latest/admin/virtual-persistent-memory.html\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/702a3553559c003c23313f3548613a0a25ede241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/702a3553559c003c23313f3548613a0a25ede241"}]},"branch":"refs/heads/master"},"148eba95d0ce44b5c48b0d1a514b5c24ed17a896":{"kind":"REWORK","_number":10,"created":"2020-04-13 20:42:37.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/81/718081/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/10"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-13 20:42:19.000000000","tz":120},"subject":"nova-compute: add ndctl to expose NVDIMMs to guests","message":"nova-compute: add ndctl to expose NVDIMMs to guests\n\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for attaching PMEM(persistent memory) namespaces to guests [0].\n\n[0]: https://docs.openstack.org/nova/latest/admin/virtual-persistent-memory.html\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/148eba95d0ce44b5c48b0d1a514b5c24ed17a896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/148eba95d0ce44b5c48b0d1a514b5c24ed17a896"}]},"branch":"refs/heads/master"},"f9ea23d940ec4ce35d503ac62f202daaa9db502d":{"kind":"REWORK","_number":11,"created":"2020-04-14 13:46:24.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/81/718081/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/718081/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/718081/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/718081/11"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-07 12:17:56.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-04-14 13:46:12.000000000","tz":120},"subject":"nova-compute: add ndctl to expose NVDIMMs to guests","message":"nova-compute: add ndctl to expose NVDIMMs to guests\n\n`ndctl` is a utility for managing the nvdimm subsystem required\nby Nova for attaching PMEM(persistent memory) namespaces to guests [0].\n\n[0]: https://docs.openstack.org/nova/latest/admin/virtual-persistent-memory.html\n\nChange-Id: I8ed1b6c1d0985b2a73206bd9249a5664cd80c912\nCloses-Bug: #1870455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9ea23d940ec4ce35d503ac62f202daaa9db502d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9ea23d940ec4ce35d503ac62f202daaa9db502d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
