)]}'
{"id":"openstack%2Fkolla~718236","triplet_id":"openstack%2Fkolla~master~I98147939ccfeeb7b3e86b5e33c427372663ecf9b","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I98147939ccfeeb7b3e86b5e33c427372663ecf9b","subject":"base/centos: influxdb repo is only on x86-64","status":"MERGED","created":"2020-04-07 22:14:58.000000000","updated":"2020-04-09 01:55:03.000000000","submitted":"2020-04-09 01:53:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718236-1586397190027-fe1485d3","meta_rev_id":"454b803556091907941140f1f01b8c7bd7886d97","_number":718236,"virtual_id_number":718236,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 01:53:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-08 21:39:34.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-08 21:10:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-08 21:10:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-04-08 21:10:51.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 06:37:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-08 09:12:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-08 10:54:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-08 21:39:34.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-09 01:53:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c10f9b9467c7d6f2ef4df67edb330528c02c43c0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-07 22:14:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"099218c1854ca7af6a234569398fa8644100f16e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 22:48:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/46e87e567afa4e60a9b3908bf8c7c1dd : SUCCESS in 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82e2a9681a466d4248520cafcc160b9762a79df0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 22:49:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1f5b47cd995be6f18c99cee2b78b9bd81f38529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:50:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/36704e194e5a4f15b42218692d51a8d1 : FAILURE in 23m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/74e133d8adbb49d28bc5a12de76ff864 : SUCCESS in 54m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3970f0ee5499495eaf50b4adfa47d11d : SUCCESS in 1h 51m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c0373b38f5e04433abba947234327272 : FAILURE in 16m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5aacfb7d09ca4db5b6890ea2cf307347 : SUCCESS in 1h 19m 45s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/aeba345380c64225aefc16185e8862b3 : SUCCESS in 57m 37s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/d9a494112ac044b78a54d191738f2505 : SUCCESS in 31m 38s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ae1bf91d300a46049ea6e99a05da398a : SUCCESS in 1h 49m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/bdc153e92d2b4de690ffc4d61402f3a8 : SUCCESS in 2h 07m 41s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/e28331aee4ed4a10b9e03eca57fb240a : SUCCESS in 1h 49m 49s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5611f7414e9748a087ad836be8faa85f : SUCCESS in 1h 43m 34s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/8c81cffc68ff4a7bb0f143033b046bd1 : SUCCESS in 1h 41m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ef527efb1f694cf7913a14a50ad953de : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/244d60be7e6c43a2ba177bdb076f5e71 : SUCCESS in 38m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0da4961ca2b84236a1642cb1caa67298 : SUCCESS in 56m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9fd177db23c4fe9b10e86247e010d7c : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddfaa40c57a54ecbbf8f32a8105be35c : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/557029c26e4b4e47a71b2aac45d641b8 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c99685ea407c414daea38f287712bd5c : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e340241aad848499e5c1d294ceae146 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/395b78ff7c5f479cb45ac6b02ee4dd2d : SUCCESS in 10m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b52d699c81844d5795a3c497918dd2ff : SUCCESS in 1h 20m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f9705a3ee8641cdb3351e967ab4b372 : SUCCESS in 27m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9668e727d3df4b74b79b44cf396ace8f : SUCCESS in 59m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9cdadb9bbac24ef39f3d4d0c2af862d0 : SUCCESS in 1h 01m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6736e63651d547c08b7fdfec343ecf9c : SUCCESS in 1h 05m 05s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d9940e39cf8141c58aaa1b9242f3f965 : SUCCESS in 1h 33m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/df3baf6e44d2495db941ebcc114b5b89 : SUCCESS in 2h 30m 38s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/578d79e77898487b9f577c1139b95023 : SUCCESS in 1h 39m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/21183c7c67c14f6986c68bd744f5e078 : SUCCESS in 1h 31m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38bf60697a47e8a86fb738cce93be75be08c8293","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 06:37:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"622124ccd6e8e5828f6028d950be4ad78b5fd937","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 09:12:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32e811274a5327ed415b4008c7ae056cca3b8f2b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-04-08 10:54:07.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"247e1a940f867d616ad099453b1f028f0ff87ef6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 10:54:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9532a46763338396c1189d03db76dc011224ffbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 13:10:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2f0482580b48408bb9f383f2f17ea02e : SUCCESS in 1h 25m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7dd5bf7f499941cfa6ae0aaa4f770180 : SUCCESS in 2h 06m 57s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/15e3aab0f93049ae8d4f54d6425eee0b : SUCCESS in 2h 13m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e7cdc5d62fe425991357028b815eab6 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa50c881a8ae45d3acefc7d0d9e8cbdf : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f30187f5f3c445b82af49d096184fb9 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e500416394d741c8912f0695a02970d3 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03ef5ce28b0845e0bcebfa0c92bbe699 : SUCCESS in 7m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b1b68cb2463041659b1461422e5a968d : FAILURE in 39m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0450773e34c9486880dae876bd4c7c09 : SUCCESS in 1h 05m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d404eb9147e4389bb0f169c08b3cb46 : SUCCESS in 1h 08m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5b176f3f0730b6aa16cecc2aef5911dae2ef9e27","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 15:46:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e9011fd57b394ee6eac608d07ad883700dc7906c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 16:56:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/305ea0d073f440e38a19fb637e41ba4d : FAILURE in 55m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"750ceb97ce94022642b9dddc6d42b9892557f220","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 18:19:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e932a03052274ddfad708cace42da71c : SUCCESS in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2858a6898596cd8f5405f375b46700018f1ff16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:02:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a80385add5764c2494131e0a4af9be41 : SUCCESS in 36m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ee89849231c44ceeb36417c43483cfb0 : SUCCESS in 54m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4d650746d70b4757a93855050a63cbb8 : SUCCESS in 1h 15m 00s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5e75f656bde241f7a6b940b3fec8b94c : SUCCESS in 1h 14m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f60be135cf1549b7987d66ff2c7713b1 : SUCCESS in 1h 21m 16s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/670e857ad83f4f278d8d9124ea8c584b : SUCCESS in 41m 46s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/e30dabc5f1494f638382553a0d113938 : SUCCESS in 34m 04s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0cb4f3b161454736a81f3d077f868a71 : SUCCESS in 1h 50m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/bd611376fc184634a3fe336aa9635fa7 : FAILURE in 1h 08m 29s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/b57cbdfaa5404378a0559127d4d8c285 : SUCCESS in 2h 01m 02s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/0f8bdd648b3544d49f477a0decd640b0 : SUCCESS in 1h 58m 15s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/90825394d48044fc93ac33c93e7f3fa7 : SUCCESS in 1h 32m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/672d7d503d924948bd5b2df83ff816d0 : SUCCESS in 38m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0c98b462c93d46029d1d2175bd446c1a : FAILURE in 10m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/740104da923e4d0497efeebcdd5791e1 : SUCCESS in 55m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a00d6bc3112348b598ecc5e018adf914 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6d2649fc88842ac8e707cce8981a4c1 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d944865d9fe244f186d7462ed47c38e7 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56ec45d1d7dc4dfb82bf26d20a4fbc7e : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44e986aee5a64022a542a6637d624066 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b038834fbd4fdd87b68076be47386c : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7e797b9408dd4e35a1c62b3abeeb5b1c : FAILURE in 1h 04m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0ad19c6152ad4814b8f133107dcb165c : FAILURE in 26m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36c5187c1b454606b60412afe88c2c72 : SUCCESS in 1h 02m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/037d3956e42a44479e530cfff02c2f2a : SUCCESS in 59m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5213a98654e041fe9fdff6fe2f306f26 : SUCCESS in 1h 07m 48s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/bc3c4d0d8e0d46e7a235cdbcb812a12d : SUCCESS in 1h 29m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/6a4591e9643b49bb92427ca7f09f6dba : SUCCESS in 2h 25m 27s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/69d130697bfc4e9986740ea43a8f1f3e : SUCCESS in 2h 11m 28s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/f7e4dfb69e6c41c78a14eea6e79189ee : SUCCESS in 1h 08m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e782b55a2c2b6a2c73ed9c419accca20910f7ce8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 21:10:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2c06b78e1ad925ca022b42008d9787fe8409b9e7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 21:10:51.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"160baf6285bebe9f55395b0078a940fa34374d31","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 21:39:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/778089bd70e74cd289f5e6ca934ba912 : SUCCESS in 27m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b2f91ff8dc6ccb5436dd531f48bb2e2c31af58f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 22:18:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/160f2a995f69441b8ae01029649f01b1 : FAILURE in 51m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d899a96e6539a6ea956faaa10bcfb6520fe3cad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 00:07:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/439a410daf3042f8be3c57e0b58cfdde : SUCCESS in 24m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/04ac008ac9f243e6b71ce6d5cde4f92e : SUCCESS in 33m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e5412ca4ae0345c4999ea428fde8233f : SUCCESS in 1h 35m 20s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/30afcc0f5ddd426e823bc5a380027622 : SUCCESS in 1h 18m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1f8d7e31d0c3432db9217ccf6c49001b : SUCCESS in 1h 27m 34s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/23e5f87136ac49b9a67bdd4404d619a0 : SUCCESS in 46m 06s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/f3def4f59fdb4c1fbfa7b3df9d793465 : SUCCESS in 33m 01s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/05a05c27e267451eaf778e249d93c73a : SUCCESS in 1h 54m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/29b354b4829d40599e38cbc5bb28daae : TIMED_OUT in 2h 40m 03s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/546b2a0b4f174fafb1a62883aaa82876 : SUCCESS in 2h 04m 29s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/3f11be2ae6274df5afb2fc2c7259b7d5 : SUCCESS in 2h 05m 25s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/e624c848725d43af9bc7d22a6871c357 : SUCCESS in 1h 22m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0fe15c70985740e185a73d5b27bb1016 : SUCCESS in 20m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7a44951a5fc4457eab3f4d74fe3e55d9 : SUCCESS in 21m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7254d6d26136461691b8f6ee77a02115 : SUCCESS in 1h 02m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be639d78d1c94009954e215c01409dfc : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35611a362aaf4f898aea12a1afa5a825 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64ec0c620b464411bdb1dd21b9d79581 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16a6ccf48b25400e9802be7f4cab777c : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64718a7ede524b0c84f645cc3c23a62e : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9509a904e18848f3b4f4e663c420ec69 : SUCCESS in 7m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8d54dd319e6403c9c918ac0bc19f9c1 : SUCCESS in 23m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/551c20c2037544959fa066d68d701400 : SUCCESS in 29m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e307a37978924d55ad71d7f65caa43bf : SUCCESS in 59m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a13202f6f89849539faa08ec2238a8a7 : SUCCESS in 59m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cc4413b5efb244af85a74a493e2dc5b3 : SUCCESS in 2h 04m 29s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0cfafc790fd04dd58880fbca4f719b8a : SUCCESS in 2h 02m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/625b86c8725e4f8997f803bd083b383e : SUCCESS in 1h 51m 23s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/e55e56b2e46a4eebb9a553f145988976 : SUCCESS in 2h 04m 27s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d394153108b44a22ad07126fe74c7475 : SUCCESS in 1h 06m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"35528a7a895449af47217ee17e16cf7ff73476b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 00:07:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"50914655172405c4c9af4497b1d1ddb262328515","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 01:53:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ebcb40630a7044f28824614ffc3613e3 : SUCCESS in 31m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6b565539c23947c8a9bb178fc59ca59b : SUCCESS in 1h 35m 35s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9ccd66714f98499998002fdba8acc17a : SUCCESS in 1h 42m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/252a25bc8c624259bdb3a6226c307a08 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d7cca53ccc4833b9dfa208884beb21 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/444b5a4f4d05469aabbf6bc8521f5cc8 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8b2b4c2e63941c99ea2651692fbe6fa : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a222e2376b4aba86578b09ef33eb8d : SUCCESS in 6m 19s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7cff35115f9e458cbc4ba4e4a5681931 : SUCCESS in 27m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef78bdd9eb754145b9ba7efe5cd5c856 : SUCCESS in 53m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7bf130d56a94402f92b824defd4e7378 : SUCCESS in 1h 14m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2750a479677bf2d4d81bd912eb7c68e2e2e22e2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 01:53:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"454b803556091907941140f1f01b8c7bd7886d97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 01:55:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b02aa6880f430ba827eebc1200ff75 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfc336ded269f5a83c75693486b4302b3e29fc30","revisions":{"6e362d5fc250123d54d1db733cf013a8464b5085":{"kind":"REWORK","_number":1,"created":"2020-04-07 22:14:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/36/718236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/718236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/718236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/718236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/718236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/718236/1"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-07 22:13:49.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-07 22:14:50.000000000","tz":0},"subject":"base/centos: influxdb repo is only on x86-64","message":"base/centos: influxdb repo is only on x86-64\n\nChange-Id: I98147939ccfeeb7b3e86b5e33c427372663ecf9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e362d5fc250123d54d1db733cf013a8464b5085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e362d5fc250123d54d1db733cf013a8464b5085"}]},"branch":"refs/heads/master"},"dfc336ded269f5a83c75693486b4302b3e29fc30":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-08 21:10:43.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/36/718236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/718236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/718236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/718236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/718236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/718236/2"}}},"commit":{"parents":[{"commit":"2d1017be3fd658011fc5a86bd38b7265da306ea3","subject":"Sensu: Remove ceph and process-checks plugins from client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d1017be3fd658011fc5a86bd38b7265da306ea3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-07 22:13:49.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 21:10:43.000000000","tz":0},"subject":"base/centos: influxdb repo is only on x86-64","message":"base/centos: influxdb repo is only on x86-64\n\nChange-Id: I98147939ccfeeb7b3e86b5e33c427372663ecf9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfc336ded269f5a83c75693486b4302b3e29fc30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dfc336ded269f5a83c75693486b4302b3e29fc30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
