)]}'
{"id":"openstack%2Fkolla~718238","triplet_id":"openstack%2Fkolla~master~Id8be6cc8f98207bc9ebf546c55d34cf9c6360cd7","project":"openstack/kolla","branch":"master","topic":"Id8be6cc8f98207bc9ebf546c55d34cf9c6360cd7","hashtags":[],"change_id":"Id8be6cc8f98207bc9ebf546c55d34cf9c6360cd7","subject":"openstack-base: PyNaCl needs \u0027make\u0027 to build","status":"MERGED","created":"2020-04-07 22:15:13.000000000","updated":"2020-04-25 21:06:53.000000000","submitted":"2020-04-09 03:41:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718238-1586403680935-85e1605a","meta_rev_id":"d68411d191fe61116c03311ce5e95e5977b6322e","_number":718238,"virtual_id_number":718238,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 03:41:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-08 21:40:49.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-08 21:11:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-08 21:11:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-04-08 21:11:14.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 12:52:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-08 13:34:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-08 21:40:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-09 03:41:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"737ee93168894319b42b1d4548603e122fc86c86","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-07 22:15:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3408c59d2d553623637276189de11998500fc980","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 22:49:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bccb9f6aae0249ebd1dfbdf0ce8f78c0d083af93","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-07 22:54:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/8df8d42169f34f699e038f8e94d986e4 : SUCCESS in 38m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0be40463eaa71b4d3c2afdf86e47a44f19a686b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 00:34:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a14c421c0d374bb3b905221674cb2682 : FAILURE in 31m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/022d8131232b4701842da4217e049237 : SUCCESS in 37m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6fe57c79861f4524b5a94fd5a9c51bcf : SUCCESS in 1h 17m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a33d83b6281a42739d6b182c97e7c9ce : FAILURE in 20m 20s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b254281cc95b46439ad0c8aebc0c64cd : SUCCESS in 1h 29m 51s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3b9f2609763e49678b746ed5af9cbf83 : SUCCESS in 1h 00m 56s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/bc03c5eb0fb34e24b20715c3f5ad84c3 : SUCCESS in 32m 14s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/94a23f8135e04237ab834a6ce56fbd5f : SUCCESS in 1h 56m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/4e82045493a1445a99d021edca1a96f5 : POST_FAILURE in 1h 03m 43s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f09049918fad423f969baf0092c48dd9 : SUCCESS in 1h 52m 24s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d120fd699f304ece9f0b0db92d1aee71 : SUCCESS in 1h 47m 06s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/dbb931850fab4b48b21ebe12480c0e06 : SUCCESS in 1h 14m 58s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/05452553e1844376971497df3e875b3d : SUCCESS in 30m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/17d56086816440c8ad1130c1775da63a : SUCCESS in 1h 02m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b1e6250d94bd45d1ba05e77be16e3795 : SUCCESS in 1h 12m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5f3921df39e40f48c88caa06951c9c5 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24eb671d034f41328bdf4281af4c7d6f : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d69d1a03fdd648ceb4c4c5995880518b : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c29780197e34d84b5d37d42f8ea3552 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a71205d17504237921db749e573e372 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b6104139204e7c9bfdddf0f7b8fcd8 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/82ff6fd41dc74def939dc436505e3d77 : SUCCESS in 24m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/19deaf473fde4e188dfb5d58941e61f2 : SUCCESS in 27m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/348e44d13a6d48ba956ee4a768abc021 : SUCCESS in 1h 20m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b4f34a6f65c8456abf6e2491c818a99b : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b32a239d5bd044f695872a590183aef1 : SUCCESS in 1h 10m 16s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ed3b0895e32f435480871663f0f4e509 : SUCCESS in 2h 04m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/02d78c9b5a094ebb8b3a4237c262eed2 : SUCCESS in 1h 44m 46s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/aff496080d7046f68b2bdb9b05b40d20 : SUCCESS in 2h 10m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/6e9e0c58c65341e094d72648fa34a98b : SUCCESS in 1h 09m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5adb94d4d0efecb01a78931aa8fd44b8573d8d7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 06:38:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06d9c5d952e95dd4ccfdd5d222225020ead283f6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 09:10:26.000000000","message":"Patch Set 1: Code-Review-1\n\nCould use a bit more info about what happens if we don\u0027t have make.","accounts_in_message":[],"_revision_number":1},{"id":"4f824c1d38a881d7e5277d5a21b02114bbb26901","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 12:18:16.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c2eaf9440f04f54b86f101e2be74083914aeaece","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 12:49:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/08db76e8fa044f80b31d9195a800e3f8 : SUCCESS in 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"69352500a5301d1e9da196f774bac5a0c4a86a82","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 12:52:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"86b6fb08513c9e129c3837c5592cfb52d2e9e540","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 13:33:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/70204b8bad77439196f33f931c961635 : FAILURE in 49m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15e6d0cd12f7b58bfa64957cfc5b673316aad5ef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 13:34:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nhrw explained this is to fix non-x86 builds.","accounts_in_message":[],"_revision_number":2},{"id":"4f24bbd510433aad02f927e6c8cdd2c58b50850b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 15:17:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2f8a8899e1034bbf9cceeef5eabce77e : SUCCESS in 59m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7334307f0e2648c59b2657caba66c654 : SUCCESS in 44m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ba0bebe8a930482e9a9d0067ffe02833 : SUCCESS in 1h 37m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d3b7515943d546cb8646ea6b1b238dee : SUCCESS in 1h 18m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e666e48c7fa44f6080a65548d6be8920 : SUCCESS in 1h 37m 36s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/bb1e7c94fd0141c2a6f938dacf7e3767 : SUCCESS in 55m 32s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/ff766c75d0574ef491fbdbde0fce9850 : SUCCESS in 39m 58s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5709682cc3e749cdbbd163e55ecd6bda : SUCCESS in 2h 22m 04s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/dfe40d3563494e39a662f0bac485be2a : FAILURE in 1h 07m 02s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/9543ab9bea9c4c33ba65860d33261ba1 : SUCCESS in 1h 57m 37s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d498eeba8b8847209f58a7464ba0a818 : FAILURE in 1h 30m 19s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/89846afcaa2843279faafad1ad45fb62 : SUCCESS in 1h 17m 48s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/970600b52938444f95539b12242872ad : SUCCESS in 24m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c0d269d0ea924e9fa0a248592ac95912 : SUCCESS in 20m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/81e5ce553791471f9d1cc7575e4c0673 : SUCCESS in 1h 00m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/170ed5125ac44535ae93b44cc46f78e0 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647db0d6da5440dcb4cbf21dc5694a19 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ef95d92a02c4e65895ed97526b89764 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec718eb169de4c408dd9d05259da9876 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbc59e1fb4cf4b279a4664e225fc87aa : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69929db75af34551be757aeda203eda8 : SUCCESS in 8m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca9a3deb2ffe4b809972cbf57e6f72b1 : FAILURE in 26m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/294780e21322401fafb766c727b49c5a : FAILURE in 28m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d25a719312354cc2a21b1d47fa6d6aa5 : SUCCESS in 1h 29m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c9295102d5b4750a56bfc96628a8b82 : SUCCESS in 1h 03m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ae2cdf75995b4ae797d2a36d288e0684 : SUCCESS in 1h 45m 29s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/57fd32a6b7324496b7918f06e342168f : SUCCESS in 1h 37m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/0c90c10e830b4e179b6040ca208d6b56 : SUCCESS in 1h 43m 01s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/60995e6e69e045bc840ae742da1242bc : SUCCESS in 1h 59m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/f9351f0b6c744b018739ccab661f2a57 : SUCCESS in 1h 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a89b93d7e6dc01e75ecf2a01eafe430f70c2d2c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 15:46:21.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"27e0d733c26be16d87a8542aba1f9af8976962a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 16:54:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c71a95d2268f44e19fd5e9e558c9b8b4 : FAILURE in 54m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac524499b5b03273b4f970aa48956dec730d559d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 18:18:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/da9af8427b3445e79df93cbe3b53465a : SUCCESS in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4d4e1947e262c7ff7aa4397f0b7f9abeadc71ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:07:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3ed84d9388b74daeb6b0f5fd042f389d : SUCCESS in 30m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1f2d0a7024ee4a5ca916cd52b3b064a1 : SUCCESS in 36m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/104cc523ad85485d93e22257cf15d439 : SUCCESS in 1h 32m 19s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/069567f9d9124e1ab4358ccc073c6321 : SUCCESS in 1h 19m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0cd5de2ec7a442da8af59238d76be681 : SUCCESS in 1h 31m 16s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b2fad84a9c824fe2ac1a9becf309afd6 : SUCCESS in 47m 05s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/276cd7f1e8d545ca812395663ccaf96b : SUCCESS in 33m 44s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b73a8f20733743cd8454b6bd257da67e : SUCCESS in 1h 48m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/29bd723476b041808057fbe7072e7959 : SUCCESS in 2h 09m 38s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/884601eb5e324feb901714e0ef6ae3ff : SUCCESS in 1h 58m 39s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/26821f16d69e48c68f0507eaae43201f : SUCCESS in 1h 49m 56s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/a4637ddfb14a4a8299f936261b1e1591 : SUCCESS in 1h 17m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6099393a7bf04a088652148cdcabfa30 : SUCCESS in 25m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/87d8d854772b4e9886df8dea487b2475 : SUCCESS in 20m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/56bb4876c55e408f941c3d062eefd625 : SUCCESS in 57m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d1bbeab2b6d4dd7a94ac959e27e604d : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1498ba9ec094ae9b3cd5712ee5ad800 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e883b2f7a5724f258fefe2f2c169751b : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c083f0483f5e4c5ca17c1f57c8e1d434 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/174e51382b90457da7f325d334f8036b : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7cca89d2934d979b8a4ad1d33ae3ed : SUCCESS in 8m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b3dab995d04f460c8ab6dafcb056ecbe : FAILURE in 53m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e07a5a0f93304aac9b8e139cc835d44f : FAILURE in 28m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7d9f24b51894fd884c8aa7b7171f60c : SUCCESS in 1h 21m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d00bb7fa8fc345ff992c4ae9f668d9bf : SUCCESS in 57m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c96aaf5861de4fe391e1c78ca5e13615 : SUCCESS in 1h 41m 35s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f4a5ea324ae74843a2ad231377cd570d : SUCCESS in 1h 28m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8292a48376984121b2bcf6d77f3a46f2 : SUCCESS in 2h 06m 56s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6ff1aac4f33d4ac5921b2e5c512bc590 : SUCCESS in 1h 29m 20s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/58e0e233b59d417cb23ea90c1e3f2c46 : SUCCESS in 1h 01m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f37d1ff1dc0bb017a8a061577f3a621c4183e58","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 21:11:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7126ebb1b37c7a7916794f0422206bccf10bc9c5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 21:11:14.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4d8a4d92f00d2dbe7f2ba182c494c25a1d2a92de","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 21:40:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/5fa049cc0cba4011be5762dc95026e61 : SUCCESS in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a47c7fb6d4645e0ac18657d3ad4c4664cb30f5dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 22:19:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c6c9e973a0d94fa0bd113d61b301f85b : FAILURE in 52m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7fd5848788229a7ef96164d297165479911ed77d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 02:09:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e25f2d82ccab4cc596aa52be4a10fcb7 : SUCCESS in 31m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/01d7215512494b0da3bd12f41880b8e4 : SUCCESS in 29m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c9dcdd3d974a4f819424f7634e28f244 : SUCCESS in 1h 17m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7f90f5ffdbf14a3fa2cab6d9483e84d4 : SUCCESS in 1h 14m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/67a0f14cdfa14beaa877e6372b4e0b84 : SUCCESS in 1h 21m 10s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/96a872c037824e36a290f6e7e8d19a73 : SUCCESS in 54m 23s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/72b623cb77e848959bb347ffe6a35cc8 : SUCCESS in 43m 27s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/3401a636fefe4f38b9ee4334685758d1 : SUCCESS in 2h 05m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/993156e4d92f4ab89f5a2fc016c23522 : SUCCESS in 2h 10m 34s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/5877f2f85834417b8b111f75e6a9de9e : SUCCESS in 1h 57m 06s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4a6f3f74684d4c5c9435da57989d9fb1 : SUCCESS in 2h 06m 36s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/8a863175f2254897ab8fd1411f8ad674 : SUCCESS in 1h 20m 14s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/af7720bd6d1d4e41911424be6238c869 : SUCCESS in 22m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6fb3876d642a4194a085adddc87c14f0 : SUCCESS in 54m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b2b0400fa2ce4ec7874a854c89ea3a3c : SUCCESS in 1h 00m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0cd424078124b90b8d116af488fa023 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc08dff1182248d58106f8f03d8c3743 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e3d6b204be548d09d9da3a72d7ca3f9 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b29d6f17047440696e36a36c785e107 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7ff249e92624238a6241108181135bd : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ec6b6da37c49be8274e509bd605743 : SUCCESS in 6m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae2e9e5b1e004617b51837c17b7be30b : SUCCESS in 32m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1756de331d6b417eb1b9e52786e1e8da : SUCCESS in 1h 02m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c530ec5cd8fe446c948621bf1ac5b287 : SUCCESS in 1h 20m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9121076e8f984ccc810f2d3cda755021 : SUCCESS in 53m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/82c9ba8d7e804dbfb41005a7a3ed316b : SUCCESS in 1h 07m 59s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d83e229d755f451ab2b93f405eb33ceb : SUCCESS in 1h 33m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/733646b6dfc54654a47caaf0afef9edd : SUCCESS in 2h 25m 10s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/cc953cd504d84de294cabcbcb86822be : SUCCESS in 1h 33m 51s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/1e903fd6ce1f449b882328e28e09e06e : SUCCESS in 1h 09m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"659233c5234891aa6cf2252ade02109d20920c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 02:10:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"602f2c459e9a27896ac292eb2a7e438d69402885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 03:41:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/be6dc56dbe5d4f568eaaf84c55d2d59f : SUCCESS in 36m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/884d9f08a8034b448fa02197f540c472 : SUCCESS in 1h 15m 43s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3ad0db3044a042e7989f75297dd0278c : SUCCESS in 1h 29m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cb8743825f74839a6afa4c1a1efd05b : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8af0ea5fc67456db663c0b42b306b1a : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/085f1e7842814436980777c8cb0f1554 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/358b8e63625f40469f93c62ca520d21b : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02d2c263d574b92b45cf8e1ba9046a6 : SUCCESS in 7m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1dce20884d004ff690ec3d00ee214436 : SUCCESS in 29m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd2eb47a2edc461d9b1bfa71f96357e0 : SUCCESS in 52m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ea9c93606ccc4b90bb97699dd94266a5 : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":3},{"id":"9bfe8adcd825f59cef1780a6052f96a77daf4222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 03:41:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dc27944f6cea11c97e62d3caf765b71ded51b0de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 03:44:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee74551e4ec4ea385663392082ad9c7 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":3},{"id":"859c5f2d0d41cd6b43092e8eb822b34303a61bbd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-25 09:41:27.000000000","message":"Patch Set 3:\n\nIs this required for stable/train?","accounts_in_message":[],"_revision_number":3},{"id":"ee7db48b04c93f45e96fc27c2c444a60b1fb32d6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-25 21:06:53.000000000","message":"Patch Set 3:\n\n\u003e Is this required for stable/train?\n\nNo","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2749e460195b2fed4d9afb87b1a7bda49e6b817d","revisions":{"c0293c7faf91d79e8424dd199669062f6345e018":{"kind":"REWORK","_number":1,"created":"2020-04-07 22:15:13.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/718238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/718238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/718238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/718238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/718238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/718238/1"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-07 22:14:28.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-07 22:15:06.000000000","tz":0},"subject":"openstack-base: PyNaCl needs \u0027make\u0027 to build","message":"openstack-base: PyNaCl needs \u0027make\u0027 to build\n\nChange-Id: Id8be6cc8f98207bc9ebf546c55d34cf9c6360cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0293c7faf91d79e8424dd199669062f6345e018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0293c7faf91d79e8424dd199669062f6345e018"}]},"branch":"refs/heads/master"},"ac5dade05f761992d34f6f3b57ce7898fb7ef48d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-08 12:18:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/718238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/718238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/718238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/718238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/718238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/718238/2"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-07 22:14:28.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 10:56:25.000000000","tz":120},"subject":"openstack-base: PyNaCl needs \u0027make\u0027 to build","message":"openstack-base: PyNaCl needs \u0027make\u0027 to build\n\nINFO:kolla.common.utils.openstack-base:    File \"setup.py\", line 159, in run\nINFO:kolla.common.utils.openstack-base:      raise Exception(\"ERROR: The \u0027make\u0027 utility is missing from PATH\")\nINFO:kolla.common.utils.openstack-base:  Exception: ERROR: The \u0027make\u0027 utility is missing from PATH\nINFO:kolla.common.utils.openstack-base:  ----------------------------------------\nINFO:kolla.common.utils.openstack-base:\nINFO:kolla.common.utils.openstack-base:  ERROR: Failed building wheel for PyNaCl\n\nChange-Id: Id8be6cc8f98207bc9ebf546c55d34cf9c6360cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac5dade05f761992d34f6f3b57ce7898fb7ef48d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac5dade05f761992d34f6f3b57ce7898fb7ef48d"}]},"branch":"refs/heads/master"},"2749e460195b2fed4d9afb87b1a7bda49e6b817d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-08 21:11:09.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/718238/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/718238/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/718238/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/718238/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/718238/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/718238/3"}}},"commit":{"parents":[{"commit":"2d1017be3fd658011fc5a86bd38b7265da306ea3","subject":"Sensu: Remove ceph and process-checks plugins from client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d1017be3fd658011fc5a86bd38b7265da306ea3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-07 22:14:28.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 21:11:09.000000000","tz":0},"subject":"openstack-base: PyNaCl needs \u0027make\u0027 to build","message":"openstack-base: PyNaCl needs \u0027make\u0027 to build\n\nINFO:kolla.common.utils.openstack-base:    File \"setup.py\", line 159, in run\nINFO:kolla.common.utils.openstack-base:      raise Exception(\"ERROR: The \u0027make\u0027 utility is missing from PATH\")\nINFO:kolla.common.utils.openstack-base:  Exception: ERROR: The \u0027make\u0027 utility is missing from PATH\nINFO:kolla.common.utils.openstack-base:  ----------------------------------------\nINFO:kolla.common.utils.openstack-base:\nINFO:kolla.common.utils.openstack-base:  ERROR: Failed building wheel for PyNaCl\n\nChange-Id: Id8be6cc8f98207bc9ebf546c55d34cf9c6360cd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2749e460195b2fed4d9afb87b1a7bda49e6b817d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2749e460195b2fed4d9afb87b1a7bda49e6b817d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
