)]}'
{"id":"openstack%2Fkolla~718363","triplet_id":"openstack%2Fkolla~master~I26bbd081f70ac76bcf45ab4348068e463a313d88","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I26bbd081f70ac76bcf45ab4348068e463a313d88","subject":"base/debian: pin erlang 22 for rabbitmq","status":"MERGED","created":"2020-04-08 09:49:48.000000000","updated":"2020-04-09 14:04:14.000000000","submitted":"2020-04-09 14:00:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718363-1586440835701-5d3090cb","meta_rev_id":"2cbc7c55df206b8eb9622539d02234258df78d54","_number":718363,"virtual_id_number":718363,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 14:00:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-08 21:40:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-09 12:35:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-09 06:11:50.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-04-09 12:35:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-04-09 12:32:30.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 12:50:36.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-08 21:40:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-09 06:11:50.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-09 12:32:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-09 12:35:51.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-09 14:00:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61b76abdfce5b5c6caa8d6bb382d645f9b6fab9a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 09:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f3211c5c92394e3a483fafde6d5dde326b81491","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 10:40:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2efc1bcaa01e4fa9b6631e96784ad83d : SUCCESS in 49m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"640b6c79d04cfbab6949b227a2d6f502728d26f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 11:03:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/96c6d2ac30e7445a9712c002102d15b0 : SUCCESS in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99e6d0e23539bf4d6e487532422dc87d4222a76f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 12:35:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/08dae1b45504483180edd0fed69a7253 : SUCCESS in 23m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b3765d1f81014bd482b39f7afaee71bb : SUCCESS in 31m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/42e7fab4d78844b881036603ca1026e5 : SUCCESS in 1h 19m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a368db48ded34fc3838ed473f7d89e11 : SUCCESS in 1h 13m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/427c40a5a35a47108bd78c5bf94d6206 : SUCCESS in 1h 22m 55s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e8243a88bfb348209ae048712419115b : SUCCESS in 54m 25s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/d58d608abadf4e6989a2093026854309 : SUCCESS in 43m 32s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b849aaf003704aca975f0811e699d065 : TIMED_OUT in 2h 41m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/60a1303a36554f55a2c069bf1c2d5fd0 : SUCCESS in 2h 08m 45s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/2374c39e9b84446d987714977ff7523c : SUCCESS in 1h 54m 51s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/776c267617a04d30ace05209f2c2b5b1 : SUCCESS in 1h 46m 45s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/55187925ad754b9293ed8a4bc28c0c47 : SUCCESS in 1h 33m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b6d9caf97e7a4996b8f4161f6685887b : SUCCESS in 38m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/20e68217b55641959d93ccb1fed68537 : SUCCESS in 24m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bdca1c96b27947849a0fec8a6a7ddfe4 : SUCCESS in 1h 04m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df8f9dbf04824e58896a087339abe0d2 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1bcaa8dee75409fa500636573229ab6 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42e397b9aba54ba78e6cc7524fe1e889 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd4ff3619b40483cbd6170b04565e598 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60ba044c38314f87b01169f668e90f70 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a30a90c30d0404d861efd3f869bebce : SUCCESS in 8m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/50ccac5be62b4b0ea9dcea79e736060e : FAILURE in 32m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf3ed941adbe4aa082df490480f12f41 : FAILURE in 52m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/132dd099c69a401e9850932b926acc92 : SUCCESS in 1h 20m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/55ad590670a3417ea3f62cdf2c102eab : SUCCESS in 58m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0473bc23102840f6b84d92e8b5772749 : SUCCESS in 1h 11m 38s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d0fc674c40aa404a9eba21c4917ebd10 : SUCCESS in 1h 45m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/fa029dd5a5b44c3c9a6a1f7e59f98171 : SUCCESS in 1h 54m 44s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/5072d3d9277f4174bf2c4ba374b916f3 : SUCCESS in 2h 02m 41s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/31ee13637aab463eb1fdc55d263bdfa2 : SUCCESS in 1h 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c4e9fbc7fc94445d5a3b11b6487d4a0452d82bc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 12:50:36.000000000","message":"Patch Set 1:\n\nIt *will* bite us as rmq moves freely between erlang major versions in their patch versions (which is mega semver lol but that is a separate topic). Could we make it aarch64 only?","accounts_in_message":[],"_revision_number":1},{"id":"d248575a68d6aa9a16b4f34b2e69b9de2094489d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 13:01:30.000000000","message":"Patch Set 1:\n\n\u003e It *will* bite us as rmq moves freely between erlang major versions\n \u003e in their patch versions (which is mega semver lol but that is a\n \u003e separate topic). Could we make it aarch64 only?\n\nOnce it move we can always fix it.","accounts_in_message":[],"_revision_number":1},{"id":"bb740a0dac1c0432fb2db18c34bd23044d34eff9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 15:46:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa0b5e7cc64363827521be0634e82de8514c2c01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 16:55:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e3a8cf2d95942818a34281e9063a63b : SUCCESS in 55m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"777fe370cfacbb677b307afc3109eb62e41eabb9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 18:24:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/4897e9ce1f21485b96e345f875840903 : SUCCESS in 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6fc24c24688cbea6fa9d9dded33ddadbfb71046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:14:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/82049e6274974d639ae1c8b67ff51c4c : SUCCESS in 32m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4308b42a1acf4934b8bd31d4b9b6107e : SUCCESS in 53m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/43e626b718804abc825def919867a450 : SUCCESS in 1h 16m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/07c215e8776d4e0f82ad1e510f4bcc70 : SUCCESS in 1h 15m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/17885f44fbd849f08a357d8f7a1c49f2 : SUCCESS in 1h 24m 02s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/9b18c43444fc4be398995070c12ebd70 : SUCCESS in 52m 32s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/c6e4d279203740ccb107bf9e0ccfd060 : SUCCESS in 33m 04s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2a46c22083674176a2d34d186d0f3509 : SUCCESS in 1h 49m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/be8596ba7e354a449a25816598561c80 : POST_FAILURE in 1h 12m 54s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/e9d86f2feb894db4a53b72cf5217dbd3 : SUCCESS in 2h 01m 03s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/60b52a1550fe4820941c0d77b45230f2 : SUCCESS in 1h 58m 30s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/97ce6307457d4cdd985ea0e11fbf186e : SUCCESS in 1h 16m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2f5b4cfef8384064a77fd5f4addc08ea : SUCCESS in 26m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/41769616a90f4bddb05b5363e3138561 : SUCCESS in 25m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/af60944ed9a14fd689d7b123d33546b1 : SUCCESS in 1h 02m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/473ed50117334a8a8e1a9928dd5ba671 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24d3242ccda443f9dc34adbbb8fed38 : SUCCESS in 9m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e025a5a3d4545ee8ad53f948acec43f : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/030311c0f70f4910839328cac01c43ad : SUCCESS in 8m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6809a5c2749b41c6bc9e19655a752052 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52abbcb0096740a7aaf8bc5f9c98add5 : SUCCESS in 8m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/731ff59dd6bc41f28a07c6f20f036c82 : FAILURE in 52m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/00808159c1914af3b782163f6cdb8eb1 : FAILURE in 52m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b375812e5be0452882f5de2241aee1a6 : SUCCESS in 1h 04m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fe2e8e70415543c3bf496aace1e7e7e1 : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9806617d980e4c8aa11e73046094f563 : SUCCESS in 1h 35m 35s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0125f2f6c0984664a2851c17de30d164 : SUCCESS in 1h 43m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a44ed7e502c34fb09e5a8eb1839732d0 : SUCCESS in 1h 50m 37s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a56be6107d574c38ba3d9066922c9dd5 : SUCCESS in 1h 48m 21s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/da9feadea8c6415b97a3f26c533cd14d : SUCCESS in 1h 04m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5538815ad50126d29f52491ef9181328ed1dd14","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 21:09:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"779b9af1baa0d39b7af594a332a688a98963c293","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 21:40:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/3d5a1bbb45d447f0a072cb56b96cce37 : SUCCESS in 28m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b327b97f7b22f130753c21f28cfaf4a426fd76bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 22:16:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/846edca608c44c97b97639bcfc0048bc : SUCCESS in 51m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c1d7b121d30c62159a73366ff7bae42f0e1674aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 00:03:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/09f9ef0c524f4631a5111daefe73ad46 : SUCCESS in 1h 28m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1728691454d04891b7aa23232a70bc39 : SUCCESS in 30m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5cc98845418c4aab8bfc7095eec8950a : SUCCESS in 1h 21m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b522cd30640e4cce94021f2b0242ff65 : SUCCESS in 1h 12m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f639c847a1ad4cfd861fc5938e849088 : SUCCESS in 1h 25m 15s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/92ecc6dae9c34b79aaf50c13bff971f3 : SUCCESS in 56m 20s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/5043b27e16104c4b84f4fd3f2c91df33 : SUCCESS in 37m 12s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1a0dde28ef2e42968979d6cf6df12446 : FAILURE in 1h 32m 39s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8d76261596e344b68b340bbc1e9b7cd6 : SUCCESS in 2h 15m 30s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/be43cb0151d641b5969630e41a2b1b52 : SUCCESS in 1h 53m 11s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6157b296ebc54977854b23e4e5cd81d4 : SUCCESS in 1h 53m 58s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/d551af32b9794a8388fc8ac1a8afbacb : SUCCESS in 1h 23m 27s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/049938ffcd6948b4901ced770d63ee36 : SUCCESS in 1h 19m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/49cc97f0b6d14da5b2f204ca6bcd193a : SUCCESS in 53m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9d733ac0ccde434a8be958e44e2b09f6 : SUCCESS in 1h 01m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/737a5908302d496fabff115f260102ba : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1e62805024541d49bf0d7b42ccc9d8a : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26409ed7cff54986982fea3df8ebdf85 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cb94bd9df5043d69e5181d3ccc466d1 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ccce83b0c4b433a9b7395b3ffbef97c : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d8b0ef5f26f46818a18edea8e52f855 : SUCCESS in 10m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d725545d9f47423eaa90c3ae16c14318 : SUCCESS in 23m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe3bfa1a4ae74653b13521c519ba590b : SUCCESS in 28m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/61db034b7508440d8398c1f4f5f546dd : SUCCESS in 1h 01m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4738118194ea4a2a9f96f1d2eb0257c2 : SUCCESS in 1h 03m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/70ccea60876046c0a0525daf6c76df4a : SUCCESS in 1h 15m 05s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b9c33bd1026d46b3858e75df142d8fb4 : SUCCESS in 1h 30m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d959b786b2f44d73a8bf1270a73d8478 : SUCCESS in 2h 37m 09s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f5ffb27dd11d4550b7d3430dc87a382c : SUCCESS in 1h 40m 05s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/94b368584ee64f5bb7ae2aaa8e52d4fa : SUCCESS in 59m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"55a75a8cc16afe9247643de92880fbb3b3ac45dc","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-09 06:11:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3f06055bceea66bcda423146517f3a69255f2fc6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 12:35:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e45c5279ae128b6023512a14e32adb99f5c6960","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 12:36:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ad9ad256a8b795225ccb2c0e3521f001a179d4ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:00:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"32fe5fddd58f597387931aa154504c43dbd7b43e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:00:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9503e6aacfbb426dabc51d119014cc2b : SUCCESS in 27m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/81b8547fa12c4df3a575d24b759c816d : SUCCESS in 1h 14m 59s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/19414985556e4d8bb5b97c875072bc71 : SUCCESS in 1h 19m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65f67ab5cfee439aa2163e5b09a91f83 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b723e031fa0c43858ce25901788cec5d : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3534652d979b4a5193b990c0e0355a95 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b151292b388742b2b6e48c447b90554f : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f88ad4dc8ad451cbd9f4893f0e8b006 : SUCCESS in 8m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8fddc9b4c507435eb830f2e346be54a0 : SUCCESS in 58m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/86c67b4739f74033929493babf96e218 : SUCCESS in 58m 59s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dda5a562e6154fccbf485812f81d5fea : SUCCESS in 1h 09m 58s","accounts_in_message":[],"_revision_number":2},{"id":"2cbc7c55df206b8eb9622539d02234258df78d54","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:04:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/437ce034e0414788bfa6cc71330ba09d : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"003e2ff3bb131e5f740cbc4787353d9caf9567a7","revisions":{"8a020dade96eae0781abf901557bc2f69cf01ba9":{"kind":"REWORK","_number":1,"created":"2020-04-08 09:49:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/718363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/718363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/718363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/718363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/718363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/718363/1"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 09:35:43.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 09:49:34.000000000","tz":0},"subject":"base/debian: pin erlang 22 for rabbitmq","message":"base/debian: pin erlang 22 for rabbitmq\n\nWithout pinning:\n\nINFO:kolla.common.utils.rabbitmq: rabbitmq-server : Depends: erlang-base (\u003e\u003d 1:21.3) but 1:21.2.6+dfsg-1 is to be installed or\nINFO:kolla.common.utils.rabbitmq:                            erlang-base-hipe (\u003e\u003d 1:21.3) but it is not installable or\nINFO:kolla.common.utils.rabbitmq:                            esl-erlang (\u003e\u003d 1:21.3) but it is not installable\n\nWith pinning image builds.\n\nChange-Id: I26bbd081f70ac76bcf45ab4348068e463a313d88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a020dade96eae0781abf901557bc2f69cf01ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a020dade96eae0781abf901557bc2f69cf01ba9"}]},"branch":"refs/heads/master"},"003e2ff3bb131e5f740cbc4787353d9caf9567a7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-08 21:09:32.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/718363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/718363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/718363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/718363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/718363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/718363/2"}}},"commit":{"parents":[{"commit":"2d1017be3fd658011fc5a86bd38b7265da306ea3","subject":"Sensu: Remove ceph and process-checks plugins from client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d1017be3fd658011fc5a86bd38b7265da306ea3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 09:35:43.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 21:09:32.000000000","tz":0},"subject":"base/debian: pin erlang 22 for rabbitmq","message":"base/debian: pin erlang 22 for rabbitmq\n\nWithout pinning:\n\nINFO:kolla.common.utils.rabbitmq: rabbitmq-server : Depends: erlang-base (\u003e\u003d 1:21.3) but 1:21.2.6+dfsg-1 is to be installed or\nINFO:kolla.common.utils.rabbitmq:                            erlang-base-hipe (\u003e\u003d 1:21.3) but it is not installable or\nINFO:kolla.common.utils.rabbitmq:                            esl-erlang (\u003e\u003d 1:21.3) but it is not installable\n\nWith pinning image builds.\n\nChange-Id: I26bbd081f70ac76bcf45ab4348068e463a313d88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/003e2ff3bb131e5f740cbc4787353d9caf9567a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/003e2ff3bb131e5f740cbc4787353d9caf9567a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
