)]}'
{"id":"openstack%2Fkolla~718395","triplet_id":"openstack%2Fkolla~master~Id1592e20bcc42893a5e9d0dba9f67cf4e896d268","project":"openstack/kolla","branch":"master","topic":"add_collectd-pmu_for_C8","hashtags":[],"change_id":"Id1592e20bcc42893a5e9d0dba9f67cf4e896d268","subject":"Add collectd-pmu package for Intel PMU (C8 only)","status":"MERGED","created":"2020-04-08 11:23:56.000000000","updated":"2020-04-09 20:44:15.000000000","submitted":"2020-04-09 20:38:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718395-1586464683203-d5dcdf52","meta_rev_id":"b6641063e84e462a3b341ae690ef942fe05d5173","_number":718395,"virtual_id_number":718395,"owner":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 20:38:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-09 09:22:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},{"value":1,"date":"2020-04-09 11:47:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2020-04-09 10:51:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-09 18:42:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-09 18:42:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 09:22:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-09 10:51:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-09 11:47:19.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-04-09 18:42:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-09 20:38:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bdd758dc221d31e46ae2e7691b7d7feee9f16f53","author":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"date":"2020-04-08 11:23:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9227fcea81a9837b5256d1db8169ca0ca7e5654","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 11:34:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0d4afa535df428b959980cedd719db9924dd2d6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 12:13:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a2ca80007ef4959bf06266518163385 : FAILURE in 48m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e10e1ef2ab0b104f8facc0376909b09410b4cc9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 12:24:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/10cabf1ba77b4cd99de9c063dcc737fb : SUCCESS in 28m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8d3d6b202c578a31dcbc2044f3a0a4e6d406e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 12:52:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8d2717ebc05646dcab79423b9a186693 : FAILURE in 23m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/99429bdadcd042248c07f260d1075b47 : FAILURE in 30m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/09c4c5795d87453a9aab6ed06967cedf : SUCCESS in 1h 17m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b825b7f703414c488f53191ae73d4e55 : SUCCESS in 1h 11m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7f5ac61a94af4b6b8d7c49fb577fb6d7 : SUCCESS in 1h 19m 45s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/1e84ab8e49104dd6912431aa674a7ae8 : SUCCESS in 39m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/552dac53db08401cbbc2c53a10dad8a4 : SUCCESS in 21m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0dfec5189e8a42ab9778b27da8d3524f : SUCCESS in 19m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fa63b604c8ca4d2cbf2876bcaa0da75d : SUCCESS in 1h 02m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccbc5deb3d4f41f6aadd55c1162a2d1a : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c49f58ef1f344e687dc1721dc04300f : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/705207877ac14cb0856c26369d5dd3dc : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/465e01f6825a423eb5f166a3ef4945be : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cd251c60ab24932a3d054b7250be4d7 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e322dc8be8048d493ecc7ea7c2c5809 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82e35bb8156d4661b8cd77a30d6e2df3 : SUCCESS in 5m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fec0a57e4fb24c2fbac7d29e371a000f : FAILURE in 24m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3246a53c8544064a71dd4ddd45d32b6 : FAILURE in 28m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/582b96cd40c34b5a82bc2cb3e6810749 : SUCCESS in 1h 00m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ebfd787e71784e098efce0c5321eae55 : SUCCESS in 57m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ee3c6b40d1b435dbdadec05ddf41cb4 : SUCCESS in 1h 10m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5cd93c9742d50ad6c9b04bd7c5e262129cdfb2c4","author":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"date":"2020-04-09 07:21:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9d2261db0c5649271066d5ecfe01c1cd2b41032","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-09 08:17:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b4aa62fa24dd4d40b98587277eb1e16c : SUCCESS in 32m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c77cacd3d07222e543a8c2f1aeda5d0c4bf01fd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 08:25:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e3d92ec9b5ec45c0987e32ad96b036f3 : FAILURE in 46m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"185ff923a66d98ebb3afd6c90015f73ca876182b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-04-09 08:29:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"848a34eb198060c3e7111025e087372e3df3f5f4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 08:36:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce8874a56f617d0ce516f72bc1d78c321aa7cd1a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 08:37:25.000000000","message":"Patch Set 1:\n\nreno comment still applies","accounts_in_message":[],"_revision_number":1},{"id":"4fce49f4939ba88f53c03573c772bd1a9eafd86c","author":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"date":"2020-04-09 08:44:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2dffd11417ab87fc024be6989d21881c4b82d735","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-09 09:22:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/9beb1ab92b1b41b29f15ef24b40ae827 : SUCCESS in 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"186fda2b430d95f6dbe013cf7b2c8dfb40af69c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:44:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/32483db18475452399afef4f119dd039 : FAILURE in 54m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c9490b8e59a8cf906ad658d02273fc046b82cd0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:46:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/54ca0b9842d143adaaddacf1182ecd13 : SUCCESS in 28m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a2cc46642cb44ec6824fbf1f89b1c3fc : SUCCESS in 37m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/40e9887748334440ab11a931bdb90ac5 : SUCCESS in 1h 45m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a55f00727737499fbc3ed80be5b13077 : FAILURE in 1h 20m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/88118dbc658542e3ae0676a718092bfa : SUCCESS in 1h 53m 26s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3f83d4c9a92e4d1594948d14d5343359 : SUCCESS in 1h 20m 14s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2bfdc592c7ac461487897efce3453acc : SUCCESS in 39m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/077b2e19fe984f818ebf914ac335328d : SUCCESS in 28m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0a48baf5aeb1463aa93566fa943c87ea : SUCCESS in 57m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f5acf05ea8a49a189007b9fc10dec17 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b31a72b64a62497ebce7d9b2add95e47 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5489055a24e14211b5f4b87b8a14fad1 : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd51c6e7f9be4275bd4f13921279285d : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae94b7b094fd440dbb71ac4f0064ebb5 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86db09967f464b3987b8003a62932dd9 : SUCCESS in 8m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/807fae2d8fc040a989e03f00ba204dd0 : SUCCESS in 56m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/098fa231bb474b39bbd570b3d1e179ed : SUCCESS in 34m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a608226bd4d945248b1aeac5cebd89df : SUCCESS in 1h 04m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cc23900d3106413cb00945013bc8d938 : SUCCESS in 58m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/751e1185a5e5417e912fa0f7455bf892 : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":2},{"id":"376671e9840d9b95dd06ad71bb535cef4bf86f96","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 10:51:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"aee4bddf9d9f387e715456dc106c3b6dfeda8ccb","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-04-09 11:47:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e47b448d8a2c42b05452f03a01b4e3879dd870b6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-09 18:42:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nwe seriously need to clean up these conditionals before ussuri gets released","accounts_in_message":[],"_revision_number":2},{"id":"6ab39291e2f7d4be9539bb87a3681a1a485cd5ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 18:42:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"293f89c00248ef6acbdfaf216e6b7d1ecdba14a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:38:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f9585f32d79f447a935629421bd067e8 : SUCCESS in 38m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/674546873bc14eea9bd7e6337f103678 : SUCCESS in 1h 19m 35s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/929d4940c05847bfa6eab36f1af7f5d5 : SUCCESS in 1h 23m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/819145808b0742bc801c14bd666227a4 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94dbd9131a94aa4a4d36f0b75cae082 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1349a22b96d24be1bd5126a952243320 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c2f7d363cec4094b117b0b121356a9c : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d70865464cbf4dd9ad017d046e735a31 : SUCCESS in 7m 40s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2ff08b6540b4ee6a0250df3cc96a32a : SUCCESS in 25m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2f650213825644698ddb7646259ebddd : SUCCESS in 59m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/56c501e99f0c487fb97c241297bb4446 : SUCCESS in 1h 16m 19s","accounts_in_message":[],"_revision_number":2},{"id":"82a4336bb8a827f21e4977e204179166f60eb9d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:38:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"76c51c24b664523030054c125a6f0d4447c05f89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:44:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6b3e7932424a27804a294ad59d3d73 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e2fe8a063aa6ec977b044094a8f056431ed60322","revisions":{"09ca53504677084657d67836fa7be728d4edf44b":{"kind":"REWORK","_number":1,"created":"2020-04-08 11:23:56.000000000","uploader":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"ref":"refs/changes/95/718395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/718395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/718395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/718395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/718395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/718395/1"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-04-08 10:30:28.000000000","tz":120},"committer":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-04-08 10:30:28.000000000","tz":120},"subject":"Add collectd-pmu package for Intel PMU (C8 only)","message":"Add collectd-pmu package for Intel PMU (C8 only)\n\nThe intel_pmu plugin collects performance counters data\non Intel CPUs using Linux perf interface.\n\nChange-Id: Id1592e20bcc42893a5e9d0dba9f67cf4e896d268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09ca53504677084657d67836fa7be728d4edf44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09ca53504677084657d67836fa7be728d4edf44b"}]},"branch":"refs/heads/master"},"e2fe8a063aa6ec977b044094a8f056431ed60322":{"kind":"REWORK","_number":2,"created":"2020-04-09 08:44:04.000000000","uploader":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"ref":"refs/changes/95/718395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/718395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/718395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/718395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/718395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/718395/2"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-04-08 10:30:28.000000000","tz":120},"committer":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-04-09 08:34:46.000000000","tz":120},"subject":"Add collectd-pmu package for Intel PMU (C8 only)","message":"Add collectd-pmu package for Intel PMU (C8 only)\n\nThe intel_pmu plugin collects performance counters data\non Intel CPUs using Linux perf interface.\n\nChange-Id: Id1592e20bcc42893a5e9d0dba9f67cf4e896d268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2fe8a063aa6ec977b044094a8f056431ed60322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2fe8a063aa6ec977b044094a8f056431ed60322"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
