)]}'
{"id":"openstack%2Fkolla~718476","triplet_id":"openstack%2Fkolla~stable%2Fstein~I9d34413fa10b5a2920aca60b08141bfaddcc7415","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1871600-stable/stein","hashtags":[],"change_id":"I9d34413fa10b5a2920aca60b08141bfaddcc7415","subject":"Sensu: Remove ceph and process-checks plugins from client","status":"MERGED","created":"2020-04-08 14:44:19.000000000","updated":"2020-04-08 21:37:37.000000000","submitted":"2020-04-08 21:35:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718476-1586381721081-887b7afd","meta_rev_id":"32b04f7f1fd738813051a2f145cc7e5899ca194c","_number":718476,"virtual_id_number":718476,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 21:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-08 15:11:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-08 19:57:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"date":"2020-04-08 14:44:46.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-04-08 14:44:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-08 14:44:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 14:44:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-08 14:44:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-08 15:11:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-08 21:35:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d7df34815bb01e96a851f0df98392554d047aad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 14:44:19.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"330c13e9203701585024c5d01085bdee0da1b730","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 14:44:39.000000000","message":"Patch Set 1: Review-Priority+2\n\nwill bp reno too","accounts_in_message":[],"_revision_number":1},{"id":"962d7075d43f5564bead2330522c823f7a91f93b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 15:11:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9217faf4b124ee59206ff9f2cb298c62cfb852d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 18:07:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b52e334f66ec4094b9211bcc98ec8671 : SUCCESS in 1h 37m 10s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5311bb83df6f4bc9b0d5310de449c647 : TIMED_OUT in 3h 03m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8d6736bf7337414ba4df0292494c5720 : SUCCESS in 1h 57m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dea698fc5c484f87a79d637345a0df82 : SUCCESS in 1h 14m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d906256c0a8b40c4aeb71832843ecb2a : SUCCESS in 1h 23m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c7fd078bfba146a0b1c6fc45a7fc2e25 : SUCCESS in 1h 02m 03s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/6dc006a1bcfc4a1f86865caa101f8b59 : FAILURE in 50m 38s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/1f3e2ec7d49d465187721da76f752370 : SUCCESS in 58m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0871a8bd0b14481fb37b4a1a758976bf : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddfffdad543e425cb2165e04df3cc4ae : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb25ef80436d48ecb3e6cc6167e51ef4 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a9ebe463a8d4fb3a94626a7ba0bcbc7 : SUCCESS in 7m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5c39e39067342d4a61f5992224b088a : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be91782d976145f1ba2128242e0bee91 : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca5319d712f04206b5057ec3e9a4c33f : SUCCESS in 33m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec1cc52d121c4fdcb69cb6a6ceb48239 : SUCCESS in 29m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce74fe21095e4794b61c27a391ad4c90 : SUCCESS in 1h 01m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc72db3ecb06444fb25b63a295631a96 : FAILURE in 39m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fab15febb709410aa3824ac2a90f9949 : SUCCESS in 1h 07m 37s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"d534bbe4b9d38635b4407c5f3e2abacbdc1cd9ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 18:19:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9c2ce9189c195b36e6f9a856eac717d3afa77edc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 19:57:57.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e781b478b53e4b2c94ded979db4c1e8954674b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:34:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4d1d943111d640ac810cf59a01d11f43 : SUCCESS in 42m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f056a357189d4e49a04f4f5f185b31f2 : SUCCESS in 46m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9ab854b8920d493c8944ff40e5b8c4a1 : SUCCESS in 1h 31m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/47771c4a2985446095ba15ad8c597fcb : SUCCESS in 1h 14m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f364e4f2b87549ccb59225d6292de128 : SUCCESS in 1h 17m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/72c400ce46e9416da6d21d8f2ad773fc : SUCCESS in 56m 14s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/aed6cb50a26e42bbbcf3f13933443032 : FAILURE in 1h 06m 51s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/934f15952c9e437b87b289ebbbe07cf0 : SUCCESS in 59m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fc1cc4ce66e49a89946de06ae55cd5f : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/226132be8cf443e4b6917994d9778118 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d4de2ed6b41407c94c8040e94457d30 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6a9d86031f84c89be09d5af1b6b5c28 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3d84b0737654e6eb7ed5b03ce9a876e : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808dc4c3662d4a6f8e9a9ca22eab2e11 : SUCCESS in 3m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/612ec68bae6e440396fe83098dde0a30 : SUCCESS in 28m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fae4b74cd064cb5ab2fafd4a4f3f7d5 : SUCCESS in 29m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f8fec7fdfd14cb69779b8cbb982eaf6 : SUCCESS in 57m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/95eee1ddffeb4a54a41dc9a65040c642 : FAILURE in 36m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9a2f97ae294549279744773ac1522f48 : SUCCESS in 1h 04m 13s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"1ff49780e3843751379986c996ea241cbce8e4f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:35:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4d2b559f65788e21bf62c1ccdea9418fc2ff5e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 21:35:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0f25b1fce6ff4d8b9842e4c777d82f80 : SUCCESS in 58m 44s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0ae3fed05f3a432ca1e384ecc3c8e138 : SUCCESS in 51m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dff1d3946d434b90bd4b30edfc2ea6dd : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b523fb95fdff43b78e022ad2b60a8770 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f849c85d7be446f8beb9581f3cd77a9b : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9de50ef3427c414b87d5a647bb52ffd8 : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65194eafd5894afca6ee13e43f9e45e6 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7132e8b211194eaa9fdcf17d0866c060 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/71fe28c7bfba4e8f9f7168ddda7e6014 : SUCCESS in 29m 16s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"0193d6c7442d3a1c4e1f97225d3dac62c4624bd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 21:35:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e38378130190d510b788d2af443873254c40995d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 21:37:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff4af883c7d04b009d0574c14ff72491 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80caffa8c81f3d0d79fcdd7682acbcffae3efdce","revisions":{"80caffa8c81f3d0d79fcdd7682acbcffae3efdce":{"kind":"REWORK","_number":1,"created":"2020-04-08 14:44:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/76/718476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/718476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/718476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/718476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/718476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/718476/1"}}},"commit":{"parents":[{"commit":"8d0af2871dae761cba50638edc650cf9a5ea2996","subject":"Bump versions (Stein)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d0af2871dae761cba50638edc650cf9a5ea2996"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-08 10:19:28.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-08 14:44:19.000000000","tz":0},"subject":"Sensu: Remove ceph and process-checks plugins from client","message":"Sensu: Remove ceph and process-checks plugins from client\n\nThe sensu image became unbuildable, failing to install the ceph plugin:\n\n    Unable to resolve dependencies: sensu-plugins-ceph requires english (\u003d 0.6.3)\n\nThis has happened because the author of the \u0027english\u0027 gem has seen fit\nto publish a new 0.7.0 release, and \u0027yank\u0027 [1] all previous releases\nfrom rubygems.\n\nThis change removes the ceph and process-checks plugins, which both\nsuffer from a dependency on the yanked english 0.6.3.\n\n[1] https://rubygems.org/gems/english/versions/0.6.3\n\nChange-Id: I9d34413fa10b5a2920aca60b08141bfaddcc7415\nCloses-Bug: #1871600\n(cherry picked from commit 818dc4d6ff12a3119be072e0c68e2d2c8da9b799)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80caffa8c81f3d0d79fcdd7682acbcffae3efdce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80caffa8c81f3d0d79fcdd7682acbcffae3efdce"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
