)]}'
{"id":"openstack%2Fkolla~718510","triplet_id":"openstack%2Fkolla~master~I428099d66715d6156a891d6f682f52e49f1b7753","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I428099d66715d6156a891d6f682f52e49f1b7753","subject":"swift: use python3 for scripts","status":"MERGED","created":"2020-04-08 16:22:37.000000000","updated":"2020-04-09 10:26:39.000000000","submitted":"2020-04-09 10:23:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718510-1586427834589-5dbef688","meta_rev_id":"f5312c822f2ebdc02b2c2f88aba9821c7d8e59f7","_number":718510,"virtual_id_number":718510,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 10:23:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-08 21:36:41.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-08 21:08:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-09 07:55:45.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-09 07:55:45.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 16:25:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-08 21:36:41.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-09 07:55:45.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-09 10:23:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa821624664da81f0df93f2e9e0dddfc3b1ba96e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 16:22:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbae7f6adf598200456bdd1148526ec6a1c3a612","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 16:25:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b91668098df284136abbef38c5e52c6f45130717","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:19:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43a9b712257c4849bfdbc1cae8a27471 : FAILURE in 53m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d438d17d02b4c905330a2a8c5cddceae62551b3c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 18:47:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/1f511937a7094bca8df12d3744aa9289 : SUCCESS in 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d5755914d7c66da7ccf527f12d687ed1e939ea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:25:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b95c314bbd1845c28b3445e69fd3e3a8 : SUCCESS in 42m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cf7abe63888540a5874dba5624d4664a : SUCCESS in 46m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e3b2814431744d2d9e47f176712ba959 : SUCCESS in 1h 21m 13s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4094b6308b7946589c48ca0389bde68a : SUCCESS in 1h 09m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f092cb26b2ff4ca3928321a869e7afd9 : SUCCESS in 1h 35m 31s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7425c285eaa241fda07098eb82c5c42b : SUCCESS in 53m 22s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c58c61e541f8442c8acd8e11abdbc772 : SUCCESS in 43m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3f60d31215a3466e89e30bc35e063bfd : SUCCESS in 21m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8de47353a617452487104c0d5cc01f63 : SUCCESS in 59m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a98ef6e03a2546729c1478032c4a73b6 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff101b454e4846808a1a52f447f59f49 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d97711563cf464fa3e1ece2a6d7d9a2 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c52ba19d40094e43af9441ce3525394d : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aebf63eea1647ecad59f60f44843e88 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f2bef0770a4d109a85544f1d7eb0bf : SUCCESS in 5m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/997511a0cb5042dcb525479e144fded9 : FAILURE in 34m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/56054ffe1f3e47379866709e862d53e5 : FAILURE in 24m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/41a114e98763408a8c673dde2c84c1a0 : SUCCESS in 1h 15m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a827c4573174798b911859a6c418375 : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/02bca167c3d34b6b9ded210d39d5fb13 : SUCCESS in 1h 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8675a79a7426b444be91ae55f166cf59a583f074","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 21:08:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5a8ad4eb100ef090659abe82531e37d105e60930","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 21:36:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/1f096252abf549bbb12a39c86fe7abe6 : SUCCESS in 27m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eac6ada591d9b7cc45c1bcb6740f9040f2c1db3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 22:14:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/29104e810df446f9aeab2290579764c4 : FAILURE in 48m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13ae9f5140a6e05d324ee4042545dae70e40d94c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:09:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a1f8b98ae5da447a920ead52b4c47a0c : SUCCESS in 26m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/711bfb75fb784280a3045750e36b9409 : SUCCESS in 31m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7c1ce0ccdc764f1aa3f0fb4af3b116af : SUCCESS in 1h 14m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/48adfef6c529414893b6d8250bee5c7e : SUCCESS in 1h 14m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/720d3628c2914e9b9835f69cb46fc0e7 : SUCCESS in 1h 40m 07s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/15beba8052c14eb78375ae1a661a2648 : SUCCESS in 56m 57s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1faaba13d985448bae835712fc7b4ebb : SUCCESS in 27m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/269d81d8cccc4e57adf4dfd7cefb3ad6 : SUCCESS in 46m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6e351c55e2a8435fbc50077658f386bf : SUCCESS in 1h 02m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64b59b720097472d9665e44f6297325f : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c27b114783d497ca98dd684b83a8293 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e3148cce7eb445c9c66168211dbc6d9 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f0af231492a470899624bb800673e3f : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e4363c1582946eab0f753245cd425ce : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c8ac39434246d9b45aeaa8f57038d3 : SUCCESS in 7m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b50b841f4ef04b88af2851e51beaaa2b : SUCCESS in 41m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9316d8fab7b4359aa8703d444729172 : SUCCESS in 38m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/656602d1a7764989a8f5ad46bd8ee17e : SUCCESS in 1h 24m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e8f63b6f16c24c19b330f4668293e97a : SUCCESS in 1h 01m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b28d5fc718db44b380d66ad6376e4f97 : SUCCESS in 1h 45m 46s","accounts_in_message":[],"_revision_number":2},{"id":"80ea3b918465aa69dfa25fe9e33cc037d103e1b8","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-09 07:55:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fee482a57e8db33c2d66c0f65fbfe00ef5d5517d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 07:55:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a0a343a4ea96f35c26719a0b9c8f1209170bbbb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:23:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/68de463959c34c2097610d0dba0b0942 : SUCCESS in 29m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a33349b922ec43d688838e99ddfafda8 : SUCCESS in 1h 48m 48s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8e6bd0340dc74ab99c6a20771b809d97 : SUCCESS in 1h 24m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f26e3f0100d84d09a56c33891e415f69 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1785e0d6f4144e60985476638ee8dae2 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7edbe93639f6433cbed452df4d0948e5 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50c933215309403c8d5d080f7f663c20 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d483d3476dfa4e4b929c6b170150b2d6 : SUCCESS in 7m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed8d6c66f615427ab7726e1d226089b6 : SUCCESS in 26m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0dbee0fae464355afe8fcca961c24d0 : SUCCESS in 1h 05m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee5756b662a7433193f716c7ceb51a13 : SUCCESS in 1h 18m 34s","accounts_in_message":[],"_revision_number":2},{"id":"8f781f85e28a0c30d4eef278a8ede0b9eb614d3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:23:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f5312c822f2ebdc02b2c2f88aba9821c7d8e59f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 10:26:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/865f5f1957d748d5b89963e003e23ac3 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f7f241d940f5e8cf51f4b30837f2dbe91035695","revisions":{"fc9e33c6d77fb831a2ffd71978319a4d07440133":{"kind":"REWORK","_number":1,"created":"2020-04-08 16:22:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/10/718510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/718510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/718510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/718510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/718510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/718510/1"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 16:21:58.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 16:21:58.000000000","tz":120},"subject":"swift: use python3 for scripts","message":"swift: use python3 for scripts\n\nWe no longer ship Python2 in images so use Python 3 in scripts.\n\nChange-Id: I428099d66715d6156a891d6f682f52e49f1b7753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc9e33c6d77fb831a2ffd71978319a4d07440133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc9e33c6d77fb831a2ffd71978319a4d07440133"}]},"branch":"refs/heads/master"},"6f7f241d940f5e8cf51f4b30837f2dbe91035695":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-08 21:08:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/10/718510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/718510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/718510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/718510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/718510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/718510/2"}}},"commit":{"parents":[{"commit":"2d1017be3fd658011fc5a86bd38b7265da306ea3","subject":"Sensu: Remove ceph and process-checks plugins from client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d1017be3fd658011fc5a86bd38b7265da306ea3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 16:21:58.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 21:08:27.000000000","tz":0},"subject":"swift: use python3 for scripts","message":"swift: use python3 for scripts\n\nWe no longer ship Python2 in images so use Python 3 in scripts.\n\nChange-Id: I428099d66715d6156a891d6f682f52e49f1b7753\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f7f241d940f5e8cf51f4b30837f2dbe91035695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f7f241d940f5e8cf51f4b30837f2dbe91035695"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
