)]}'
{"id":"openstack%2Fkolla~718511","triplet_id":"openstack%2Fkolla~master~I74069ac30684f875c26abb670812e9ff30e158f5","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I74069ac30684f875c26abb670812e9ff30e158f5","subject":"qdrouterd: drop Debian/Ubuntu part","status":"MERGED","created":"2020-04-08 16:25:43.000000000","updated":"2020-04-08 23:37:41.000000000","submitted":"2020-04-08 23:34:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718511-1586388846941-0a348a98","meta_rev_id":"bb51cdb38f29bd3b536e204b19e6cdf6467bcc9f","_number":718511,"virtual_id_number":718511,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 23:34:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-08 20:37:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-08 19:58:49.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-08 19:58:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-08 19:58:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 16:46:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-08 19:58:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-08 20:37:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-08 23:34:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18651e073e7ca9e03916accaaa318cb6618bd1cd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-08 16:25:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b20b5d75963c53cf0d231a6b3b0bbc89c219a2d1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-08 16:46:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e49b0321883668344600085bb5de70cd03021eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:22:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/70d680f7f3ba4c1ebac5ecc00e4ec2d6 : FAILURE in 53m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61bebcac776e184040410b36f9a32cfec9dbb7a6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 18:48:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/366a194c7928421bb223543d2e4329cb : SUCCESS in 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93a6428198b674a83c2fd3a8559b884239142026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 19:50:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/11469b7256ea4f45a03858014201c60a : SUCCESS in 24m 59s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/34b7a829f91a4f0ba16cc6eab01ec875 : SUCCESS in 32m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a161dbe7be474e1a839efad8031a2cec : SUCCESS in 1h 35m 55s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0b58e05ec4d248668b8820537aaba126 : SUCCESS in 1h 26m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3907dbe8d88d46a3acf3e07aa8074fc1 : SUCCESS in 1h 57m 08s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/6c890ff96a614d9490d37f2a22403be5 : SUCCESS in 44m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b97d0389124c402a86f29b538261af30 : SUCCESS in 21m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ba6ea404e51443bcb57ffacc46cde8be : SUCCESS in 38m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bad62bab749f453a94eee6674a995ff7 : SUCCESS in 1h 16m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e0379ac55f94611aef06644eeb6938e : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b86ddfe01174bfe981dfe47e47ae673 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4cb55401f054cf4ae88605695ddd84f : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3361b470275b4a068c6815f88a5dcae8 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a553a22c63645fea964b76ac0f3fe4e : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e0a51c53bc44cea912184742a12a4c : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/280f20ff659742deae99109f25f694c7 : FAILURE in 35m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bec2f98fbeaa4a0d9840ee489defc283 : FAILURE in 36m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4de5b915b8f4031a12a3abcbcb5cc01 : SUCCESS in 1h 14m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/28f715c9a9e443dda691cf3e59124550 : SUCCESS in 1h 12m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9d2f7c1757564170af9a9dbd3ffc10af : SUCCESS in 1h 12m 57s","accounts_in_message":[],"_revision_number":1},{"id":"0e34fe010829ba4bb00fc8836be5c41c400fd509","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 19:58:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f65b6d1f0c94cb4c01a64b467de69a5ad7da680","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 19:58:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4621399b087e66b5e985f37bec0ee3462df752ac","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-08 19:58:52.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c22da534f3f8d1f793e144ae08d6fbb8c3794067","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-08 20:37:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/abe7d40d5c6847a2922725a528c45a3c : SUCCESS in 30m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bbec31540954dd80d6aa5edea0fb53ca99278251","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:46:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b85500ecb0e42e59e518177f246e63a : FAILURE in 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9a9c6042161dfc94e22f8ae55df86721c3b8c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 22:09:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/18bdc639d59d45fda67adb0340214135 : SUCCESS in 22m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/93363851b16f450e99d7b5b0824c71e5 : SUCCESS in 32m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e8952bcba9fd4e05a9256d7dfaa58f0f : SUCCESS in 1h 12m 46s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/024eca5cd2c9415383c021108662cf25 : SUCCESS in 1h 20m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a91123654d2645f181b38cdeeaa3f7d7 : SUCCESS in 1h 20m 36s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b5b6b4c80de24e809781c311186f1af3 : SUCCESS in 50m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7b927e3626e44f0b9582b02509d8d468 : SUCCESS in 27m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fc0257fc255e451cb31aa54d307b71ca : SUCCESS in 18m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d5f91589800144048edf26671fcb2284 : SUCCESS in 1h 02m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58514c6329eb457a86dea8419d3e47f9 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a7bf28228d424f8a4c7bdf891c2301 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/161eb37b577a464d85d56683ccc9c2c0 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c35357cba3849b8ab486f6a981db537 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1da7b5abc764739a6ec32f58fc42110 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/456e97e25efd4edfba774ffac295fa0a : SUCCESS in 8m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ba43acda95454a548f34a4ba39f12870 : SUCCESS in 27m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b5a1d6edb1d451f9f8b53491059cf63 : SUCCESS in 27m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/acc5637bbf754af3be064c0ab3f89927 : SUCCESS in 1h 04m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f68bdc207444b0386412dab460680d8 : SUCCESS in 1h 03m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7d096a2fc05b40849845d91b2a824cf5 : SUCCESS in 1h 09m 49s","accounts_in_message":[],"_revision_number":2},{"id":"314679fb3cbbfc96c8def100e3667ee8a9fbf6d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 22:10:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dbe4f4def108724d12e92ccc2934f2ccc30ab410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:34:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2f388392c4204ad8969ece92887685d7 : SUCCESS in 34m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/834c0ff067494d3693c684f22ea146eb : SUCCESS in 1h 12m 02s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fdcd06ef0cc84995837984881f6db909 : SUCCESS in 1h 21m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b10fbd7b1fdb464c88abf29bbe39be27 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e4fa36a9559490e954c0ec851239704 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b007c93b6a0043d9acf38f0a921c742d : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4a1f55556f24993a3e1462a654b1652 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef0a8431dd1d4078922d79ad3af283ee : SUCCESS in 7m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/66d5a61a88fc4033955bca53ac426b9a : SUCCESS in 46m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/03df2e864c2d406287e5832dcc46d22b : SUCCESS in 1h 07m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cdf0d48a1be84941902530c6f8c0298e : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8870697429264698bd060a77573b86efd19cafba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:34:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bb51cdb38f29bd3b536e204b19e6cdf6467bcc9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 23:37:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224db2f0689b4fbda9e5d59cc9446364 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"89f46e9450564af37a9ea7f86dbcad248be453f2","revisions":{"d26d1454b6d86b0ac4ca66e1286b6c3558a90b15":{"kind":"REWORK","_number":1,"created":"2020-04-08 16:25:43.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/11/718511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/718511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/718511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/718511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/718511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/718511/1"}}},"commit":{"parents":[{"commit":"f1f1d854594da23321ad94ae6358ecbdc119d18e","subject":"Upgrade RabbitMQ to 3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1f1d854594da23321ad94ae6358ecbdc119d18e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 16:24:30.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 16:24:30.000000000","tz":120},"subject":"qdrouterd: drop Debian/Ubuntu part","message":"qdrouterd: drop Debian/Ubuntu part\n\nPackages are no longer present in Debian nor Ubuntu so let us drop part\nwhich care about those distributions.\n\nChange-Id: I74069ac30684f875c26abb670812e9ff30e158f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d26d1454b6d86b0ac4ca66e1286b6c3558a90b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d26d1454b6d86b0ac4ca66e1286b6c3558a90b15"}]},"branch":"refs/heads/master"},"89f46e9450564af37a9ea7f86dbcad248be453f2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-08 19:58:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/11/718511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/718511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/718511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/718511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/718511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/718511/2"}}},"commit":{"parents":[{"commit":"2d1017be3fd658011fc5a86bd38b7265da306ea3","subject":"Sensu: Remove ceph and process-checks plugins from client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d1017be3fd658011fc5a86bd38b7265da306ea3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-08 16:24:30.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-08 19:58:49.000000000","tz":0},"subject":"qdrouterd: drop Debian/Ubuntu part","message":"qdrouterd: drop Debian/Ubuntu part\n\nPackages are no longer present in Debian nor Ubuntu so let us drop part\nwhich care about those distributions.\n\nChange-Id: I74069ac30684f875c26abb670812e9ff30e158f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89f46e9450564af37a9ea7f86dbcad248be453f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89f46e9450564af37a9ea7f86dbcad248be453f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
