)]}'
{"id":"openstack%2Fkolla~718612","triplet_id":"openstack%2Fkolla~master~Ia9feac726a60250215b1bc78bf90dc68ac6f956a","project":"openstack/kolla","branch":"master","topic":"718612","hashtags":[],"change_id":"Ia9feac726a60250215b1bc78bf90dc68ac6f956a","subject":"kibana/debian: use upstream package on all architectures","status":"MERGED","created":"2020-04-09 08:13:10.000000000","updated":"2020-04-19 13:27:32.000000000","submitted":"2020-04-19 13:25:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718612-1587302752014-7130f63b","meta_rev_id":"750c0616b0c1a2c4ad29e253559609f9d5a47543","_number":718612,"virtual_id_number":718612,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-19 13:25:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-17 09:13:52.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-19 11:08:40.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-04-17 13:37:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-04-19 11:08:40.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 09:13:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-17 13:37:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-19 11:08:40.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-19 13:25:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e89a97f43a8f14d83361f99a9b425112d20d8941","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-09 08:13:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cc09164c6cc22b1fc466b7c832272633ed3bdfa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-09 09:13:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/540a29f61dcd4edc98226e0ae67e0511 : SUCCESS in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48be8863c79b5fcde655f730ecba3b3652c5e195","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:16:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eebe3057e53440f28a47e996424439a0 : FAILURE in 51m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6da670f8e8f8c6afbeb6d58c5149deb59925c05d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 11:09:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0afc23ed49794333836c20d9e309510b : SUCCESS in 24m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6f3388266dc243eab1e0049eb1a4f6b6 : SUCCESS in 31m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/acdbfd19b1c54f99b99940582f237394 : SUCCESS in 1h 28m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c50a3f6cd1154104919ffbd52cd74910 : FAILURE in 46m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6de3f9c6f01d43e8b2a1695fe3e2791a : SUCCESS in 1h 25m 21s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e621f68777dc4dc7bef85057bcc348ff : SUCCESS in 48m 44s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/83d51b5fe9124dfa9b5932d4b4478a12 : SUCCESS in 32m 59s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f802ffd871364de289061205c32728d9 : SUCCESS in 2h 02m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/88599ebf56564f7ea7b8d391c8a7fe35 : FAILURE in 1h 18m 21s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a49cd5e47f2b49d5a0900a65e00e7725 : SUCCESS in 2h 14m 33s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1c92137b277c4170978f59551256f654 : SUCCESS in 1h 54m 59s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/ef646db3d91c4a19a77bbe4da27808b2 : SUCCESS in 1h 18m 16s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e9a4a1199c984029986fbe1ac7e2dbe2 : SUCCESS in 22m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f2f43dc4d3f242b28d112cecc39fe624 : SUCCESS in 44m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8380bad75b80445988dd98ac604f1d94 : SUCCESS in 1h 17m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08a0e95485b949a2b9316ba2d4348125 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d16fb8110e44e9588d250d06b0141c9 : FAILURE in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b4496c8a39b4c54ae4e64180888fcd3 : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81f033ec66ac4aed9a19501de1dfb98c : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8382227d4f14134bc3453cc6d068a27 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c312e88aac844fb9778c660f543f523 : SUCCESS in 10m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c55f8256b4243558e651810bc2259a5 : SUCCESS in 26m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/181790fa6ded4db7accec6f169af8702 : SUCCESS in 34m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64e586c7535f458c877daceb017b650b : SUCCESS in 1h 05m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c6e7037064404bc695e16a5dfc1993f5 : SUCCESS in 54m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6144d7e0cd5c40d49ed7305d0d75837c : SUCCESS in 1h 10m 32s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e608b655a6234e649b66afc5a244abf8 : SUCCESS in 1h 39m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/43421d30fcdc4215b69ba52b1bde6f09 : TIMED_OUT in 2h 43m 12s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/daf2ba60512448d9ab14087bed5caccd : SUCCESS in 1h 35m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5052c2c41ab248fa8bdb2a102dcdd0da : SUCCESS in 1h 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5428b91b61f2a064f866d80a77c5504743d486df","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-09 12:40:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"983096d97bf31023ae74ed63fad110a16f31a67f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 13:29:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ae1eefb0ec441b49319fe51bd1e8aee : FAILURE in 45m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"189e75067a2a86182eb85496f69bedb41873d13e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-09 13:54:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/68a7255acc2f4b669abb2b50b333f194 : SUCCESS in 36m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"81894f464988fddb1a38974e07a3b129381d7aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 16:17:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c2ba0a84bed3416dafea0fc947dcd002 : SUCCESS in 26m 15s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0a35924968c941e09604d6b305cb2b58 : SUCCESS in 34m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/22f454a0881a4120a0ed32ee43009132 : SUCCESS in 1h 22m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3ea44b3122f842feb29491ad8975151a : FAILURE in 1h 03m 20s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fbea893654244fbba49021e521d70c6e : SUCCESS in 1h 37m 30s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/51222b17078f424da88276eba3c19d1b : SUCCESS in 58m 31s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/5fb5fb291ecf48acaf44c2c73f27307c : SUCCESS in 33m 11s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/034177957d0b482cb776a0bb892b2a47 : SUCCESS in 1h 47m 47s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e936f4fcf104486c9531a8c89144f5c7 : SUCCESS in 2h 13m 11s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/8d11bb6263e847b0aa35ca9f4188ea2e : TIMED_OUT in 2h 11m 12s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6ce23d0118194071ae36ec95726cfdab : SUCCESS in 1h 49m 30s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/b03d5cdc77b24d229fd0f458a88cce1d : SUCCESS in 1h 16m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c88b626da11046fa835b31a01108f784 : SUCCESS in 24m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8f79020d7a2f4408a77b433d4ff2ab01 : SUCCESS in 18m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1b07cea527104f799297954d85dc6492 : SUCCESS in 1h 05m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65a9ba56909047509ea62c3c662fed14 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0649ffcff9ca48c299d7c9c80b5582d0 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/198e1217b2ec43079bffa2a4318bcc11 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f5344c066cb4fadababf4de14d89b67 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a13443105e65408d990626f7bbfcfa5e : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2681d7b530943029652595aa4ad872f : SUCCESS in 6m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b22db50a20de47479ff4d0fd9ac35bd7 : SUCCESS in 23m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/486db48beabe4491afd23b3e6b4aa715 : SUCCESS in 24m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/902b1372504c40bbad94b8d1615c1506 : SUCCESS in 1h 07m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/321dc96aa3254087855687133ca3afdb : SUCCESS in 1h 02m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a8699eda5cdd4078914934e8dbe02357 : SUCCESS in 1h 29m 10s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/48ead43f5d204287a88028002788dcb3 : SUCCESS in 1h 36m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8572c0b363bf469aba3c232ef9f739ae : SUCCESS in 1h 53m 08s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/afec656af4fb41bbbeb95bc6d8c79e4a : TIMED_OUT in 2h 09m 14s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/be03411d1277475c9bf97810ec583b0d : SUCCESS in 1h 04m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c1d9f41e1ac5267f4e140386733dd1d55c42b5f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-09 16:38:07.000000000","message":"Patch Set 2:\n\nand no compat issues?","accounts_in_message":[],"_revision_number":2},{"id":"bc03b00334eb1b70f77cdfb23bb0d38edf3adf36","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-09 17:12:28.000000000","message":"Patch Set 2:\n\n\u003e and no compat issues?\n\nNormally I fetch \u0027x86-64\u0027 package, unpack, remove node directory and repackage. This patch does the same without creating packages in meantime.","accounts_in_message":[],"_revision_number":2},{"id":"e183d78d9e9f139b0fe510b9d85e78e918b705bf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-10 09:26:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a079a95125a36ab8184e7db610b2b60fb4d273f9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-10 10:04:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/58fddad035144e25ade0e2b3c33b54ea : SUCCESS in 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ca98276ec449a4f7cd2da21964da60a2c5975321","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 10:25:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/92a697c796b94b629a854bf1af440cf2 : FAILURE in 46m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a071a825cbbf4e6ab237cf43e81f0c18 : FAILURE in 44m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"793796826cac2393f5f44980dcd3538f680d400f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 11:54:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4ed634ac216249748ec84aba0e068dbf : SUCCESS in 1h 10m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/854565116ada43b88932337f1f239ce6 : SUCCESS in 1h 21m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c14b30a3166d42729ab152aa661c5ee7 : SUCCESS in 1h 55m 04s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ef2a1d1163e9457286dada654c9d74a6 : FAILURE in 1h 23m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b725ce9620394db3bdf914609d7737b3 : SUCCESS in 1h 20m 23s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/959affe16da444718b46301b1b88a3ea : SUCCESS in 1h 17m 13s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/bdaf5e86c5204bcd9d18fefc9dcc8ad3 : SUCCESS in 31m 03s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d82e9586f1764a1390f1b68277f2fabc : SUCCESS in 1h 54m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/4f9ac0b69c0544239885f6a82963b8c4 : SUCCESS in 2h 25m 33s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/e64a4c3b684b4ac59ed80de38e051e62 : TIMED_OUT in 2h 07m 45s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c09860dc69584500a790e2d06237a6ab : SUCCESS in 1h 41m 27s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/7051a38f88754a6ca8e0fd4191abe55a : SUCCESS in 1h 14m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0fc10908a47a40cb9ceff4da78cd98e7 : SUCCESS in 21m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e1e8ab201c1f4d23ae2ea1c032bf96d2 : SUCCESS in 19m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/295b06101e44431aaf2cb10d6a9dc188 : SUCCESS in 59m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/867aa435b62a41dc8b84d1a03c498305 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbee6c259c754265b6192e7c5e7fd6d4 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d34ee6a7d8ac43d284ba963911aad45c : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a03dd6a8430c458fa80626b4d44dcf10 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fbf71175ef1464085301ad051cb917f : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6175d93eae94303857267755fe5477b : SUCCESS in 7m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8511d33d41b44cb2959dab7306f50859 : SUCCESS in 19m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b6817d03e384f25a846ea9e457e7dfc : SUCCESS in 22m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eaa30d1f357844f29c72fd18a7f892aa : SUCCESS in 51m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1dcbc5cce2b94b6eb8b9e055b83a783f : SUCCESS in 57m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f2d2b23d4647421089ea44096ae4111a : SUCCESS in 1h 05m 37s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/da361d6f5c164ccfa731f0f9c1e076c9 : SUCCESS in 1h 33m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a76ae7941bb747519914d01ab5e3e282 : SUCCESS in 1h 49m 01s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2b0f96d5ec814cdabc84a842584602c1 : TIMED_OUT in 2h 07m 20s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/8f87c22e772a4571ac0c10f6fb298388 : SUCCESS in 1h 00m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e86670eb25c6bcce73d516ab8082885ebe23cd14","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-12 15:13:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a5555c5c5d2229b73b317521d521907bb09e2201","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-12 17:25:15.000000000","message":"Patch Set 3: Workflow-1\n\nneed to check why it fails on aarch64","accounts_in_message":[],"_revision_number":3},{"id":"20bca9300bbb0d985169a6567f8d2e1dd3123e31","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-12 18:15:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"acb2e0504ab9dbd8ac64f45c640db6b39d358043","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-12 18:49:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/1d49041b1dbc46088925a1e5fbbeaf89 : SUCCESS in 32m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f327fe4bac91f4fab7c1f6830eadb002454bd16a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 19:04:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/182c9cdba4ed4219aba5a5470b523faf : SUCCESS in 48m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/64b5e802acf64a4094179631303ec9b6 : FAILURE in 43m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"085faa8792a52cfc067473cc3610ccb08799422f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 21:44:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c18e636706144d30909d4f5fc8d77c13 : TIMED_OUT in 3h 04m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f093499ddddf44688d3f43da218c7373 : SUCCESS in 25m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f94f794eda6442d39b28453fd06c8ee3 : TIMED_OUT in 2h 02m 39s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/69888efc284d4b5db93e516df6e58af8 : TIMED_OUT in 2h 03m 05s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7462d211f8014f00ad0496f15657ebf8 : TIMED_OUT in 2h 32m 57s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d71c0589d5d4408cb39007849e56ba70 : TIMED_OUT in 2h 07m 25s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/916666d2208e4c13a10c3b8658aba1fa : SUCCESS in 1h 14m 23s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/7c1cbc1748984dd097e7b2d92db65c0a : TIMED_OUT in 2h 33m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/370917055a664d0f87d4560a22de70d0 : TIMED_OUT in 2h 33m 50s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f6c4aa0bfaaa48ae98c14a666ea4fc55 : TIMED_OUT in 2h 04m 41s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c657fb1d330f4472a1b2afba57e7151d : TIMED_OUT in 2h 03m 36s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/5fa3eca9bb9f4753afd6d6c7914e0881 : TIMED_OUT in 2h 03m 03s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7e8ca10d509445a8a8ff724dc5be086e : SUCCESS in 44m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3edf2f83bcdc4d589de217f185bce939 : TIMED_OUT in 3h 03m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8be1b529133e40199fe5bf39067e6ef0 : SUCCESS in 1h 16m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ab270d994eb4fb8af3ff22a45f4c74d : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a28a673f3e428498e38a6fd8c38fd9 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b2d8d3d316d4afd900cca573791089e : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82528a0086254a71813df2507c1f1f86 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7b24877ba2046be93e05b273956d798 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e52b7bae8c8c4f21a3105b4dcc6f9999 : SUCCESS in 6m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b87d96d95aa44ad8b2ae9b95c6fafe6 : SUCCESS in 2h 56m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/86d5bd1ff6164d0da90804d38e9f03e4 : SUCCESS in 2h 54m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e744381d7f9e4063babc9fc7f290be27 : TIMED_OUT in 2h 03m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/23d64fe71847422091df6e47bb567935 : SUCCESS in 51m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc63663e83e0403591bb7954ed39e13a : SUCCESS in 1h 07m 48s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/adbd9690e3b941fe9b47d934798b7275 : FAILURE in 1h 37m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/77fd825e98db4a77a993bbd20d9b7b2e : TIMED_OUT in 2h 32m 58s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d507ef1e11fa478ea559e66d44b76602 : SUCCESS in 1h 48m 13s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/098de113dabe48ce85b063cbd190bd83 : SUCCESS in 1h 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"41ab9d498c75d3aeeb6ce5cd98298110cc7b30a8","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-13 14:56:58.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0f9ef3d44bcbe5ee4c646eba4fae7bd5fbf280e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:46:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fa44f3a8d09a48279bb935c417ff2489 : SUCCESS in 48m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/633e3bf2e35e4f30813add7c1c20d5b0 : FAILURE in 47m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2ee65f164ad1904ab39618cd9b97f91123b292c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 17:09:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5fe8496cbc9b41748bf3dce5c2128415 : SUCCESS in 29m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/eeb886807d44463182cb8a89cbcf7078 : SUCCESS in 34m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b694d06e146047ba8a651e20ae51f32a : SUCCESS in 1h 20m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/428ab2fa05724021ae79b16492b4a7be : FAILURE in 48m 48s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/208c45c1038c4fc6b42867f88796b868 : SUCCESS in 1h 28m 05s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d595641b40ae4c82b08583af9ee168ed : FAILURE in 46m 13s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/08532956fcea442982fb4bdbcd069609 : SUCCESS in 30m 13s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/cc500764af4e46479a3dfb69f1b66362 : SUCCESS in 1h 46m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/35f59e4b76a34c948d27c0ad6ace407b : SUCCESS in 2h 11m 00s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/6a2b57c00d594d69bf21d2d09ad02c00 : SUCCESS in 2h 05m 20s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4822faa91abe49f1adb1296ddb2ef4ed : SUCCESS in 1h 41m 41s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/7e88d6e678964ac29fe7fb98c226c264 : SUCCESS in 1h 16m 58s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e3af90c86cfa4a78b258c16711d151b4 : SUCCESS in 25m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5546ab7d036d431398ee4f8e3df257d0 : SUCCESS in 22m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/489ff151b39441a3bf5ff5fd87179477 : SUCCESS in 54m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ad5766308b74706aceddee4fab836de : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f96ffce96f35461c874e1093ff0efd82 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a32d5a22bdf945f9aea4d5b0b0654da4 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13c4df7631544ccba8af5d9e12c01402 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/030b7d7f42634bdb9acab7a835669545 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de4391c2eded41b5b9712b402decfdcf : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/44462cd62fff461498e0ee57a0d14490 : SUCCESS in 28m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/52154ac4c4d74f8d89f3b759083b23e7 : SUCCESS in 24m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/12194bd05cfd44e4b4d00555cccb6a4b : SUCCESS in 1h 04m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87e4cf47097e4d17ab9bf416695b049a : SUCCESS in 58m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3294812336be4467bd04f3e3ce2db267 : SUCCESS in 1h 05m 23s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/73a783cef9754a65ab9c358aa63c870f : SUCCESS in 1h 50m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a360386d979d4dc8be763cc2d3effba1 : SUCCESS in 2h 03m 06s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/834273eb9f094712873007a2d8f83897 : SUCCESS in 1h 31m 47s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b33dd4e1c4c842b480bd1b58aa4b25ca : SUCCESS in 59m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"72c9f68b98f4b2d7817b79fe7a0efa5955ebb1b6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-13 17:27:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ff761edc526b4dba8014d69b9c2b9b02 : SUCCESS in 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a198da9f545c967d7408d49fff42e83a49f3d3fa","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-13 19:53:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"50ea089325a6261c9df27813c63560294f4be50e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-14 12:21:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4b90aaa2cd9786735e741378e6b97950b73d424e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-14 13:36:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/481eef49d9354bbf8aac974889ded237 : SUCCESS in 26m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fe535251e4199ee0190cca3500e01d2858b95ecc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 13:46:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/073f919b99a14a6fbe5a305bb89efb99 : SUCCESS in 48m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b64d558a2c4040349d2e7d6eaa6d548f : FAILURE in 48m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8a983d48aab7f219707a384391503365e5384029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 15:07:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/831fc54d781441d38419a5d440603185 : SUCCESS in 22m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c952011b83ba4f08918dfb561c285b18 : SUCCESS in 30m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6e740b949f4e472a87ab1eee963e1225 : SUCCESS in 1h 16m 18s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3937050f97734ad6b54b2fb04dfd1ac6 : SUCCESS in 1h 09m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c520519a05d846aab1a1e462eefafa05 : SUCCESS in 1h 18m 27s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/9a2ee1526d964c31836c3d9fa9db48cb : SUCCESS in 52m 23s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/a119837445394678847bd9e2403b74fb : SUCCESS in 33m 31s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/be3d047677b44a509a6b3a6516b427dd : SUCCESS in 1h 52m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d5f1c3eb98c94f799a1911c923838ab7 : TIMED_OUT in 2h 38m 58s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/c9dcc229fa6b4265a8b6a05ba920dcaa : TIMED_OUT in 2h 11m 20s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/18b07684095b4375a97709b328857857 : SUCCESS in 1h 56m 43s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/cf68eb83408146a9a51962aad37607c5 : SUCCESS in 1h 20m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0979afa33b714da58cb932ed7d865920 : SUCCESS in 21m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4af203956e504ce688f97eab54190da5 : SUCCESS in 19m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ef148bcfbac946cd85e22ccd4e8ad3c1 : SUCCESS in 1h 00m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/543df1f15904457b815c5a12f71e37be : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33da2b9c27c748ecb9298df9fe6986a6 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4558ea5af9524e37a002bfefcd5a306c : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/717881e5b1524ffb9af3be32d585113b : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f50e6f9c845409b8a8ee802de20c2ea : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0586073d83743f9af8d266ae20dc185 : SUCCESS in 10m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c3a833cc82c40c0ad193eec4449627d : SUCCESS in 23m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2d1ae3484f84b9fa55cbf67539a4f94 : SUCCESS in 23m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/52dc0789166443f7b95d1e97d020d42d : SUCCESS in 1h 02m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aea5815a46e749f6a88f26efa4c0a023 : SUCCESS in 59m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/991c63311adb445597e4b2bf925b24f2 : SUCCESS in 1h 08m 39s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5e4c411efdd741b896038af7b30475bd : SUCCESS in 1h 29m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ffd0bac998d14fdf871497146e8da247 : SUCCESS in 1h 43m 56s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7c5ed719e7194744bcadbeafaa465df5 : SUCCESS in 1h 41m 16s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/12d06707693a438cacde9e5068327f11 : SUCCESS in 1h 02m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6ae8e0385df63ee5147d1c136a8baf213dd36564","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-16 08:34:21.000000000","message":"Patch Set 5: Code-Review-1\n\nmatrix update is missing","accounts_in_message":[],"_revision_number":5},{"id":"2d74abbc17159b29f42d34533fbdd69e9060f3b4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-17 08:47:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"681326dedbd27b4f7dc7a0a53270e46180ea17af","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-17 09:13:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/f16982284a8540728db58d9479e52154 : SUCCESS in 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0e3cf1d0d7b5e4664b843236459019afb87ff067","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 09:45:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/22672c7dcc304ff29ca699bfc9a75e55 : SUCCESS in 47m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f581bc9f4264450c896837de7b21e14b : SUCCESS in 51m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e5687df4312303e39ac974c2a42ced9c27113b9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 11:17:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/31865ae64dc947dd87b75e5515b573f8 : SUCCESS in 56m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b67c189048874c9c89452b97a4a14e9d : SUCCESS in 58m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fbc96f92d60d44d1bee9c2b78107d13c : SUCCESS in 1h 20m 55s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bf656513195049c59244de0fdc37ddc3 : SUCCESS in 1h 16m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/07632a9b0c1449538b2ba28651e25cc3 : SUCCESS in 1h 28m 28s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c4085fed4b784227bccdd04d4e0d647e : SUCCESS in 54m 22s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/84ed17895bc9480788ac0678252f371a : SUCCESS in 33m 40s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5ae6c952e8804877aaf0615a7dfad5a1 : SUCCESS in 1h 52m 39s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/077e1d60114a43979b1f1ce4e3e00efb : SUCCESS in 2h 04m 16s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/68b5cc0e7cc04ba5b490483acd59980e : SUCCESS in 2h 02m 56s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/e4e849f89ace436bae077453a237fe2b : TIMED_OUT in 2h 14m 06s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/225c17388e82487ab9fadb14b369cef6 : SUCCESS in 1h 19m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9c6e18543a3c49b1be3045ffdb03b72f : SUCCESS in 20m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2c8f1686670640e985ae2553c0447c04 : SUCCESS in 39m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7912b10920ad4c179a528b1019fdec7f : SUCCESS in 1h 11m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f7b16007c4248e581fdbae846c53d5c : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32ba63ab1074f45b8d29a7734e9e35e : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d8dc08656624e86b8c4f37de2e17e97 : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/455ea4bc3619446fbdbba98689f7a86a : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ed75e604015418eba366c46922e5f3e : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdd3e55d4e2441488d2d1039d307a9b2 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d4078edcd7ea4d92ad697ab0a703cbf6 : SUCCESS in 26m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/27b19b5b26df4d5389339cf3276f89a1 : SUCCESS in 27m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/13e99671873546cc8ada3640a65e4d74 : SUCCESS in 59m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1251675c1abb432fa2191bd1800b9f57 : SUCCESS in 58m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/114893e99e8d4ad5b9f3be9d980c5158 : SUCCESS in 1h 09m 27s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ec3639b8f4bb4b49bf4d333772d7a952 : SUCCESS in 1h 32m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a42c052dd0774c508c89b1d83a8d5152 : SUCCESS in 2h 00m 24s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/eaabf64f4d834fc0a8fe56f8657cdb87 : SUCCESS in 1h 27m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/220778a3518743149f05093c91334f7f : SUCCESS in 1h 16m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0ad743c614ec0ff7932e5f8675656baa937a1049","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-17 13:37:40.000000000","message":"Patch Set 6: Code-Review+2\n\nhacky but no better ideas around","accounts_in_message":[],"_revision_number":6},{"id":"5538a2305e6c6328e56caab3f5f1b78b46a5aa70","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-19 11:08:40.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f8dc4c630f87b7b381d88ca83fbfde9117898f9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 11:08:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"bb383c293fca658cce01056db9619193260b877c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 13:25:51.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3aeac88f72d94d6da0e1615daa1c8d61 : SUCCESS in 38m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/68028e5d98f04305a4d8ce413a904534 : SUCCESS in 1h 18m 00s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2350f9d6608d4e889f3f47d9e6646991 : SUCCESS in 1h 15m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d69d81322ef849939215b5abe738cf54 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efe7b06efcb8485cb05e0ee47baab6f0 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35be8a1ec8a54eeaa3851c2d186b7396 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d1fee715f674ed6b7e17d483e8eb847 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fafc5c9d3eb64ee7a952c7ec867cad42 : SUCCESS in 9m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf755ac921544f3e86a232b055cf0fa1 : SUCCESS in 38m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4c1dd0904b3425eb7d139878590b30c : SUCCESS in 56m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/49e4f27ca7d149a585dc5a0aefff39c2 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":6},{"id":"df730351d3b7f7a56e0adf4fc4f0a58f3c645ead","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 13:25:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"3a68b04eae7b0817db274a699952392c2c9be952","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 13:27:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d690f6191e43489fbb92c3c8c889dc : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"345f0555bcae10056b0ef6a79d32cc124f07d5a2","revisions":{"c3cd269a94fcd0639da1c3b6a3e5a8b9c45da819":{"kind":"REWORK","_number":1,"created":"2020-04-09 08:13:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/718612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/718612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/718612/1"}}},"commit":{"parents":[{"commit":"a176c6e080be2e71829e8ca5e8b330fdda70038a","subject":"Merge \"Add docs and release note for CentOS 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a176c6e080be2e71829e8ca5e8b330fdda70038a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-09 08:10:33.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-09 08:10:33.000000000","tz":0},"subject":"kibana: use upstream package on all architectures","message":"kibana: use upstream package on all architectures\n\nUpstream kibana package ships x86_64 binaries of NodeJS. I usually\nrepacked package without them to have something for other architectures.\n\nInstead let us install upstream. Then on non-x86 archs we add nodejs\nfrom distribution and remove x86-64 binaries.\n\nChange-Id: Ia9feac726a60250215b1bc78bf90dc68ac6f956a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3cd269a94fcd0639da1c3b6a3e5a8b9c45da819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3cd269a94fcd0639da1c3b6a3e5a8b9c45da819"}]},"branch":"refs/heads/master"},"4ab41f46fa48cea060b958756b72add08311e24a":{"kind":"REWORK","_number":2,"created":"2020-04-09 12:40:53.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/718612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/718612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/718612/2"}}},"commit":{"parents":[{"commit":"1b33dd61fc93d8228d22ae73c858af2de8cc593c","subject":"Merge \"swift: use python3 for scripts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b33dd61fc93d8228d22ae73c858af2de8cc593c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-09 08:10:33.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-09 12:40:39.000000000","tz":0},"subject":"kibana: use upstream package on all architectures","message":"kibana: use upstream package on all architectures\n\nUpstream kibana package ships x86_64 binaries of NodeJS. I usually\nrepacked package without them to have something for other architectures.\n\nInstead let us install upstream. Then on non-x86 archs we add nodejs\nfrom distribution and remove x86-64 binaries.\n\nChange-Id: Ia9feac726a60250215b1bc78bf90dc68ac6f956a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ab41f46fa48cea060b958756b72add08311e24a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ab41f46fa48cea060b958756b72add08311e24a"}]},"branch":"refs/heads/master"},"31e75b24b37fbb1ef942e201afd32771993ec3ea":{"kind":"REWORK","_number":3,"created":"2020-04-10 09:26:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/718612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/718612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/718612/3"}}},"commit":{"parents":[{"commit":"cea4bdf0d0cc72b4058dd4c6cb73f88f7ee9171b","subject":"Merge \"Add collectd-pmu package for Intel PMU (C8 only)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cea4bdf0d0cc72b4058dd4c6cb73f88f7ee9171b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-09 08:10:33.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-10 09:26:02.000000000","tz":120},"subject":"kibana: use upstream package on all architectures","message":"kibana: use upstream package on all architectures\n\nUpstream kibana package ships x86_64 binaries of NodeJS. I usually\nrepacked package without them to have something for other architectures.\n\nInstead let us install upstream. Then on non-x86 archs we add nodejs\nfrom distribution and remove x86-64 binaries.\n\nChange-Id: Ia9feac726a60250215b1bc78bf90dc68ac6f956a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31e75b24b37fbb1ef942e201afd32771993ec3ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31e75b24b37fbb1ef942e201afd32771993ec3ea"}]},"branch":"refs/heads/master"},"43e9768e0e0509efa12cc96c00dc6a4fca893bb5":{"kind":"REWORK","_number":4,"created":"2020-04-12 18:15:44.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/718612/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/718612/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/718612/4"}}},"commit":{"parents":[{"commit":"8518744133ec9cf5401021c09211fd985c386fe8","subject":"Merge \"Allow operators to use \"ceilometer-upgrade\" parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8518744133ec9cf5401021c09211fd985c386fe8"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-09 08:10:33.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-12 18:15:30.000000000","tz":0},"subject":"kibana: use upstream package on all architectures","message":"kibana: use upstream package on all architectures\n\nUpstream kibana package ships x86_64 binaries of NodeJS. I usually\nrepacked package without them to have something for other architectures.\n\nInstead let us install upstream. Then on non-x86 archs we add nodejs\nfrom distribution and remove x86-64 binaries.\n\nChange-Id: Ia9feac726a60250215b1bc78bf90dc68ac6f956a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43e9768e0e0509efa12cc96c00dc6a4fca893bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43e9768e0e0509efa12cc96c00dc6a4fca893bb5"}]},"branch":"refs/heads/master"},"4e2d48eb6ccc1d0537f25d76fc6354626478e0ec":{"kind":"REWORK","_number":5,"created":"2020-04-14 12:21:52.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/718612/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/718612/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/718612/5"}}},"commit":{"parents":[{"commit":"504c76d0db9d65fcf289762268f53ee5296329f3","subject":"Merge \"Upgrade Elasticsearch and Kibana to 6.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/504c76d0db9d65fcf289762268f53ee5296329f3"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-09 08:10:33.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-14 12:20:24.000000000","tz":0},"subject":"kibana/debian: use upstream package on all architectures","message":"kibana/debian: use upstream package on all architectures\n\nUpstream kibana package ships x86_64 binaries of NodeJS. I usually\nrepacked package without them to have something for other architectures.\n\nInstead let us install upstream. Then on non-x86 archs we add nodejs\nfrom distribution and remove x86-64 binaries.\n\nChange-Id: Ia9feac726a60250215b1bc78bf90dc68ac6f956a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e2d48eb6ccc1d0537f25d76fc6354626478e0ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e2d48eb6ccc1d0537f25d76fc6354626478e0ec"}]},"branch":"refs/heads/master"},"345f0555bcae10056b0ef6a79d32cc124f07d5a2":{"kind":"REWORK","_number":6,"created":"2020-04-17 08:47:00.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/718612/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/718612/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/718612/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/718612/6"}}},"commit":{"parents":[{"commit":"f0defeafd295459c0532b8296c86543c31a95728","subject":"Merge \"CentOS 8: Switch last mentions of yum to dnf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0defeafd295459c0532b8296c86543c31a95728"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-09 08:10:33.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-17 08:46:44.000000000","tz":120},"subject":"kibana/debian: use upstream package on all architectures","message":"kibana/debian: use upstream package on all architectures\n\nUpstream kibana package ships x86_64 binaries of NodeJS. I usually\nrepacked package without them to have something for other architectures.\n\nInstead let us install upstream. Then on non-x86 archs we add nodejs\nfrom distribution and remove x86-64 binaries.\n\nChange-Id: Ia9feac726a60250215b1bc78bf90dc68ac6f956a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/345f0555bcae10056b0ef6a79d32cc124f07d5a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/345f0555bcae10056b0ef6a79d32cc124f07d5a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
