)]}'
{"id":"openstack%2Fkolla~718659","triplet_id":"openstack%2Fkolla~master~Id106cb71624e0b0d8393904afd2a62469e6066f1","project":"openstack/kolla","branch":"master","topic":"re-enable_rdt","hashtags":[],"change_id":"Id106cb71624e0b0d8393904afd2a62469e6066f1","subject":"re-enable collectd-rdt plugin","status":"MERGED","created":"2020-04-09 10:46:45.000000000","updated":"2020-04-09 20:38:01.000000000","submitted":"2020-04-09 20:38:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718659-1586464681691-5dbef688","meta_rev_id":"7f168add5db9131ef241ff02174a74125416fbbd","_number":718659,"virtual_id_number":718659,"owner":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 20:38:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-09 11:15:46.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2020-04-09 11:54:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2020-04-09 10:52:06.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-09 18:41:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-09 18:41:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 10:52:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-09 11:15:46.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-09 11:54:53.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-04-09 18:41:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-09 20:38:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b1cc5e3449f2d26a05185ea982a0b51a4e83525","author":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"date":"2020-04-09 10:46:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b87507091b42a278393120f2dd4adfc240e49d3e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-09 10:52:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce76bfe0c494c5620f3bb0fa6c7dfe5fafedccbb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-09 11:15:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/1bbf10b66737466895a61aec1a9af599 : SUCCESS in 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"910a327c50d7f2c9e83533c8f1e28ffebf4dfaf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 11:44:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/336aea270cab49e1aa8d9da91e6a03f2 : FAILURE in 48m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f9aec7ac32cb1ccfff5ed5c2106d33d0fd4437f","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-04-09 11:54:53.000000000","message":"Patch Set 1: Code-Review+1\n\nthank you, looks good to me.","accounts_in_message":[],"_revision_number":1},{"id":"88eb568f60f22f96cf76993e85d8eeb4a6460b33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 12:27:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e927e45b2121420fa09dcd93283e8a0b : SUCCESS in 36m 07s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d5ee9590306b49c69a00805d08c35c86 : SUCCESS in 44m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9e1f5a5a171c4156892660d99b8a18c3 : SUCCESS in 1h 23m 45s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d4f1e5daad8a456e9334bc46dad1fda3 : FAILURE in 49m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/07f10c996282457abb1656df756d6991 : SUCCESS in 1h 25m 59s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/4084fba3d4a243d395feef5ea069eb58 : SUCCESS in 38m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eba15fba73134600a09ec0898435eade : SUCCESS in 21m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/308d0f1fa19f4cf49c7bfca21f2239ed : SUCCESS in 32m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/934157a13bd446aa973fb3b96694acc0 : SUCCESS in 1h 07m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fc4210eaf384d1b8821a77b264d1d8f : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be7ffa4eec649ac9a31830b53f82088 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c083f568fb574bf5aebe4347e8e46270 : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c637d1b2350474da8402467b860a51b : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4c62b9300b54b938398db1497a61abd : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41ebf7d093a4702b4e0bae717f1b00e : SUCCESS in 9m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/90f514a7c08b4e1a811f31d8426c65fa : SUCCESS in 57m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2376e3a7fb464a3684ef5ca0b45aa651 : SUCCESS in 26m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f686858eaca45d9aba7b84a7a029932 : SUCCESS in 58m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57c0078c2ffb47ffb925b12fd5ea43bb : SUCCESS in 59m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9a8571d5fe4c4be5b8adb96ec4f7dfef : SUCCESS in 1h 10m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8ff9831b6bc258eb3b96571d5ad019b3aac4ab25","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-09 18:41:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7953407ff825ee5315e92d99113dc37d10811098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 18:41:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2114ac9a513b3003246e2b7b91413ea1af86ed41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:38:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"27765b80535edd2c700e0929db7a4452b9c10cd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 20:38:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/35be46c00f0f41158ccf5385102463ea : SUCCESS in 47m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a7c75999818e4b7b97943c716c93d34f : SUCCESS in 1h 15m 21s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9d7c024c4101453d9891670e5ea499ad : SUCCESS in 1h 53m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dee838eff474679bdaf9b5013466df6 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e2b8a5ab7c4c59b32ded20c4405111 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c56a77a17b94702aa19c3a1f16a5b53 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dab24b1b9f4743abaac3fa5e3f8399b3 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ae3b134f844340bffd8573b75ee3a7 : SUCCESS in 6m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f77758270d7404597545ca521085851 : SUCCESS in 27m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a5ea3061281d4cd786845104b0ff25de : SUCCESS in 1h 01m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf1a0a19dee846d7b7803206e81693b7 : SUCCESS in 1h 09m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"157e0754c84cbedd30513e4a35d1f420508cfebe","revisions":{"157e0754c84cbedd30513e4a35d1f420508cfebe":{"kind":"REWORK","_number":1,"created":"2020-04-09 10:46:45.000000000","uploader":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"ref":"refs/changes/59/718659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/718659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/718659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/718659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/718659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/718659/1"}}},"commit":{"parents":[{"commit":"a176c6e080be2e71829e8ca5e8b330fdda70038a","subject":"Merge \"Add docs and release note for CentOS 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a176c6e080be2e71829e8ca5e8b330fdda70038a"}]}],"author":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-04-09 10:25:30.000000000","tz":120},"committer":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-04-09 10:25:30.000000000","tz":120},"subject":"re-enable collectd-rdt plugin","message":"re-enable collectd-rdt plugin\n\nChange-Id: Id106cb71624e0b0d8393904afd2a62469e6066f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/157e0754c84cbedd30513e4a35d1f420508cfebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/157e0754c84cbedd30513e4a35d1f420508cfebe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
