)]}'
{"id":"openstack%2Fkolla~719364","triplet_id":"openstack%2Fkolla~master~I071eebc4f60094d881a68e201e8c3bb6e6742ff5","project":"openstack/kolla","branch":"master","topic":"wip/debian-use-ussuri-binaries","hashtags":[],"change_id":"I071eebc4f60094d881a68e201e8c3bb6e6742ff5","subject":"Debian: use Ussuri binary packages","status":"MERGED","created":"2020-04-12 18:18:41.000000000","updated":"2020-04-30 18:54:47.000000000","submitted":"2020-04-30 08:33:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719364-1588235639634-39316be2","meta_rev_id":"40e31f8fd4e82ddb65054120c9bf677e89b81a46","_number":719364,"virtual_id_number":719364,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 08:33:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"date":"2020-04-29 22:12:43.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-30 06:59:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-04-30 03:58:48.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-30 06:59:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"date":"2020-04-29 21:11:31.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-04-29 21:10:04.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-04-29 21:11:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-30 18:54:34.000000000","post_submit":true,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 21:11:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-29 21:11:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-29 22:12:43.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-30 03:58:48.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-30 06:59:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-30 08:33:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-30 18:54:34.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"}],"messages":[{"id":"2748200acd4d34095a6caa27b35c62028da074e2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-12 18:18:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34da7a7a9daa55f8036d48f36edf23e9e1b9d354","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-12 18:46:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ae4b6107e7034f6382c3393b90558696 : SUCCESS in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8821ad593d2d34171f0f8751b2bbf83f5b9a121d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 19:16:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/033b7d19d0bb489eb187824ffacf65da : SUCCESS in 49m 38s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e71f6758f75440c08b2a5be1094e04b2 : FAILURE in 53m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22cf0d54e94e6c4127ebc5588b41c523d22bd421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-12 20:30:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1ba949bbcc7a4df2891ad0c2e7ff68d0 : SUCCESS in 24m 26s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/66747cbc07064e5289e0e9a8e244828e : SUCCESS in 47m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7429304390a6405c835c370dcab7a632 : SUCCESS in 1h 10m 55s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f1abe4005f8b4e21a19deba122737aab : FAILURE in 41m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f3074ad56d7646d084f46867ba98ece6 : SUCCESS in 1h 16m 51s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/187b483dd66b465bb6a1e9309b32c7c5 : SUCCESS in 40m 56s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/b8a10c2be46046d08dbf5b2e7acfd852 : SUCCESS in 29m 57s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8fd34300a6ef42ba88ef060d00b8f3a1 : SUCCESS in 1h 48m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/eb296512bf8340c0bf58581d5a09fb43 : SUCCESS in 2h 10m 45s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/e17ec26cc4ef41768f5056cef177186c : SUCCESS in 1h 42m 45s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c12c790186dc4e3ebef1e79ff09c8067 : SUCCESS in 1h 38m 31s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/48372c5ddc054fa5931a2d3829536abe : SUCCESS in 1h 19m 05s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/54fba08110e741d3a3d681d4a11ac8c0 : FAILURE in 18m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/497a717e027c4be9a1bc35e98cea8f84 : FAILURE in 17m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/967a781b63c24eb5a22a99be12052335 : SUCCESS in 54m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3a1af77188f41a1a16729108f4aca47 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2806c1468b44a818c07578290f1eb0d : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fd58f9a669c412790fb092c41a295ef : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/763e78bccc39441782ffeaacdad8b77b : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06592aa498d84e808653fbea13a77e14 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f597e9cb3bc747d180196a3eb8b395d1 : SUCCESS in 5m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8287c26c8ad24976986395d4e15577fa : SUCCESS in 22m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b4f6996a8674fb1ae4754b96e5424df : SUCCESS in 26m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0e184b9609b482d947d1e2672bee8e7 : SUCCESS in 57m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6348bced53c4aafb27210f501d6c26a : SUCCESS in 56m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/04c0ae2d83cf4655b8269e327e6509b9 : SUCCESS in 1h 03m 39s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d313fc3ab781482dad274daeb426d46c : SUCCESS in 1h 40m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/435cf676a52e4cecac591932ab0e9ea8 : SUCCESS in 2h 00m 06s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6c39aee74a4941ce89f88cdc9a58110d : SUCCESS in 1h 38m 21s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0dd1b4011ec94e889e9845ccdd73bd77 : SUCCESS in 57m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49644ed95aa2209a3ffa7520a2216e4ab8f9102f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-13 09:15:34.000000000","message":"Patch Set 1: Code-Review-1\n\nHmm, looks like this butchers debian.","accounts_in_message":[],"_revision_number":1},{"id":"fa037e3e225af6ac3d20cc1b76a7f4170ea67d99","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-14 06:52:44.000000000","message":"Patch Set 1: Workflow-1\n\nWIP\n\n\n22:47 \u003c@zigo\u003e hrw: We don\u0027t have Ussuri packages just yet, currently they haven\u0027t released even rc1 for Ussuri.\n22:47 \u003c@zigo\u003e So we do have clients, oslo, and such, but that\u0027s about it.\n22:48 \u003c@zigo\u003e What you exeperienced with Horizon is quite normal, we have Django 2.2 prepared in our repository for Horizon, but no release for the dashboard plugins...","accounts_in_message":[],"_revision_number":1},{"id":"319aefc1295ba8104034f3aeb28c19278e0502e2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-27 13:21:57.000000000","message":"Patch Set 1:\n\nfew packages left\n\nhorizon needs vitrage-dashboard which needs some deps handled\novn-* are broken","accounts_in_message":[],"_revision_number":1},{"id":"7d1756f20cda0b5c9c43cff716be6e813345b6c7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-27 21:18:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4323ba694269da09cd4f7bd04d339fd3dc73bf58","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-27 21:19:05.000000000","message":"Patch Set 2: Workflow-1\n\nStill not complete","accounts_in_message":[],"_revision_number":2},{"id":"4225f3ba0687e8635cc9b0a7ac697189a977c1bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 22:18:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/5592a61323c049bea58749cc07c39ae2 : SUCCESS in 50m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7230d25998c243098317ca1f694adab8 : RETRY_LIMIT in 2m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec2881a6563b4adca3adc307a52ed14e : FAILURE in 4m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a4ccd493a8440dc879cac8e4a64862c : RETRY_LIMIT in 2m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bfc276cacb729c235621b6ff1b1ab3771a8bd8f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-27 23:08:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e1bc7560719e47caa609c92132a6ae05 : SUCCESS in 32m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d94d8f2650975ead46ec4519f5b43b574d93beb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 02:44:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/157505ff2545479499ec8de51be236db : SUCCESS in 26m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/be60e28ddcfd46a38d565172fe49c054 : SUCCESS in 36m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2b056ffe5f2e44b79d622016c6459878 : SUCCESS in 1h 20m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4efc655ba77442e4b4fc46107846cb3a : FAILURE in 6m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5d6091fcde9a4fee9f43d2ddb446d082 : SUCCESS in 1h 29m 51s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c9d56d7559b7444bb55b9a5b34e89507 : SUCCESS in 54m 32s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/3815f75ba4b64e669ea758b4c445c8a0 : SUCCESS in 33m 50s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/cf4815a795d34eb9a35f71e2558e841b : SUCCESS in 1h 46m 32s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/50126f2e4d3d46eea3346c3f761ce77d : TIMED_OUT in 2h 44m 14s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/3d1bc25e8a6641498564e4d7f13b8fee : SUCCESS in 1h 51m 27s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c9cdaac727094b3dad91d74a5b58dd8d : SUCCESS in 1h 48m 05s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/650adf740950414097532fc0e3034c6f : SUCCESS in 1h 18m 33s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/d3586003e1e246809b54fe8c4fd51c9d : SUCCESS in 2h 01m 14s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/75dc8f3fc2354348809ad4826e9b64db : FAILURE in 25m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a2103e1ad9bc4493872a7ab3a424630c : FAILURE in 20m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/76535833beff48bf81be92997de0a55e : SUCCESS in 1h 03m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0bb4fd1519a4aee94793ae423f6b2a6 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b32af54cb154449971839f5a4fecede : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/810be9edad97418686ef4815fada8580 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0d5af8621cd4a57903057f3e50615be : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40724be6278c4430add8a29ac059e048 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a312ee023d4677bfbf54b47ea139c8 : SUCCESS in 6m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1264c00504e04610a34b2f00fdeb518f : SUCCESS in 24m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/db8d0229ee6c4dfca8e89e53c0014c65 : SUCCESS in 26m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ccde61cb69004f11934f6fec07aeb62b : SUCCESS in 1h 02m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/672cbf9a88114e2983b6d61229a8de09 : SUCCESS in 1h 04m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3d537bd4a7a944a69c9a51995c6805ea : SUCCESS in 1h 11m 26s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/814be1cf70eb42c8bbb09f80201ff4fe : SUCCESS in 1h 41m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/92b124f1d4884250b0a3b625ad7c38d7 : SUCCESS in 1h 56m 07s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/764fc0b434e64ba19dacbfe8e5956f66 : SUCCESS in 1h 32m 50s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7a3bc032a00345758eb45671ca88d04c : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/c28e487b53794eedbf315c851ed46df3 : SUCCESS in 1h 39m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50c37b6db34b9287b0c9e1e626f09d7c02e49418","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-29 21:09:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"94bfbe5b1e88b752c10eedd0bb620591d92b3258","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-29 21:10:04.000000000","message":"Patch Set 3: Review-Priority+1\n\nAll images should build now. Please review.","accounts_in_message":[],"_revision_number":3},{"id":"fff0338ffca30ce395c383d1d7d3a2cb493b3ee8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-29 22:12:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e61953e16776433680ab8b54e384333b : SUCCESS in 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dc244f622fc455dd7022b8739c89fa6f4b98a84b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 23:07:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e69b7d9d4ecf4c948ba288879d601eb6 : FAILURE in 8m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f76d8c34cec4be6971e6d68d9c8dff5 : RETRY_LIMIT in 2m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb0181e3089845ce9548db634f206ee4 : SUCCESS in 1h 42m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b5749532f7946ebaa2410c4533b1070 : RETRY_LIMIT in 2m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"abb46cb9e12d8ab10772a6ba705430d218021be5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 00:03:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0897cce3517b4e738f031ec552a8065c : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9c86c8f446f7406f8740e3c3b1d37357 : SUCCESS in 28m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cf6139c9a9c34590977f4a39c2c568bd : SUCCESS in 1h 22m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5935c16f134b4e0eb5e8709b15b14178 : SUCCESS in 1h 03m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c176c98c95e34d26a422c972364e6141 : SUCCESS in 1h 55m 03s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/afc31683ee224cfdb57c36f2030afd3f : SUCCESS in 43m 18s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/d82b5e6766e24c6facb0ced99a20d72e : SUCCESS in 30m 42s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/807d93c587ba485dac1645a2a7e072fa : TIMED_OUT in 2h 50m 03s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/07d790767fbf4416ac660c86c4c792de : SUCCESS in 2h 38m 02s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/c98cb05366db4798b99e4d876cbb3c8c : SUCCESS in 1h 52m 21s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/85620a5b7bc14fdb8f28dfde30efffe8 : SUCCESS in 1h 40m 43s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/50a7ace7789349c7aa8728d8b54debf3 : SUCCESS in 1h 21m 04s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/9708db9ffae24c4aa8f3ff7cd3bf090d : SUCCESS in 1h 47m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c5abf162ca2e41789ec7fb4d773a3d20 : SUCCESS in 28m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3547808c5c5046909d83a902b4fc81f5 : SUCCESS in 20m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9938a44f5ff2422fbefafdda80571adb : SUCCESS in 1h 02m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a65c22c9fd9346e7b8a160e2b96966ce : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00e437bd6a2c4b4d9f9738cd131ccc21 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa183a4fa07c4d16903e51ba4dec51b4 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93292598eb394ab8b35f5424f3a4960d : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5719742f944c47a6917e4c6269c30f37 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bce47d695c546d1bd52db12b6a2c3bb : SUCCESS in 7m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/327fdafb408b454d9890e71829c69ea3 : SUCCESS in 38m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/94ac89ad499d480d9ce4a31a88eac553 : SUCCESS in 1h 01m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2c65741162754ae081bdda8091cd65f1 : SUCCESS in 1h 22m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b879c1a59396455ca3c7019da51af89b : SUCCESS in 1h 01m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d43c24e7ce894adaa574395996fe8307 : SUCCESS in 1h 20m 59s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c460a6dd2bbc45ffb2993a92caa9eaad : SUCCESS in 1h 27m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/140fffc3582b4d97a519f94739aa3dd2 : SUCCESS in 1h 44m 20s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0bbc5536bdad493199e3b843a10e6f16 : SUCCESS in 1h 25m 16s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b0bacd8e768e40a986876b5bed182bd6 : SUCCESS in 1h 07m 09s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/5283b14e3dcb4aa494c6ec1e02667a0f : SUCCESS in 1h 26m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dc84e0a6a6b34ee195277d1fd40be75267f75daf","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-30 03:58:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4215e466f45ba78525b96ea3ceea3c07535f8cc4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-30 06:59:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\ntrusting hrw is doing the right thing","accounts_in_message":[],"_revision_number":3},{"id":"93839033ba41a44c4e579d2b2e371107d9d88d69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:00:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ad5775f33b7e77f00b543f7d5624987d02fc254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:33:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e5686c668d8f4c1f9ee13e90a28ec0d8 : SUCCESS in 24m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f459aa37d62341a6a21ce5b18ceac208 : SUCCESS in 1h 32m 20s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/441d7318698d461ca605747ad2810345 : SUCCESS in 1h 25m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cad41ad8f424f41a98c98e4bddd7ff2 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/261747c3cefc495a8466ef8f0e17bbf5 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f71cbe4db4747e3a9380b64bccf7dce : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef8e3273982941eaa83adb3636e042e3 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c445ab35f1444e93b3f7ef7ab9b0ef4e : SUCCESS in 8m 01s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/53ad67894c39474585e11f02cb8ad0b6 : SUCCESS in 27m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/433faa7269c041deac4d8d91834208ab : SUCCESS in 1h 09m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6a2fc629b87b4a0aa11627a349561953 : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":3},{"id":"346e85f1d436a75d4dd64c814f7d4bac99729dc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:33:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9515bf2d933acd73433a8e7a77ca722a3e57568f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-30 08:35:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7761fb7b316438e6d39ce02c3be78bb2bb5d56f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:35:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79bb1ebd60fd4919a73ab84bfd208a4e : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3},{"id":"9edefd5a02534e018034a0deeb351a197a7e82dd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-30 18:54:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f608f5e32393b3a06f77a4f383def25d8c0fdad4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-30 18:54:47.000000000","message":"Patch Set 3:\n\npinging Zigo","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"62c8dc95bd79a1b72b8633712847d0ba44c89126","revisions":{"a0cdfc6dc8eb543a9a52d1bea814583726bc8595":{"kind":"REWORK","_number":1,"created":"2020-04-12 18:18:41.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/719364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/719364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/719364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/719364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/719364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/719364/1"}}},"commit":{"parents":[{"commit":"8518744133ec9cf5401021c09211fd985c386fe8","subject":"Merge \"Allow operators to use \"ceilometer-upgrade\" parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8518744133ec9cf5401021c09211fd985c386fe8"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-12 18:17:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-12 18:17:15.000000000","tz":120},"subject":"Debian: use Ussuri binary packages","message":"Debian: use Ussuri binary packages\n\nThere is a repo with binary packages for Ussuri. Let us use them instead\nof Train ones.\n\namd64 only\n\nChange-Id: I071eebc4f60094d881a68e201e8c3bb6e6742ff5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0cdfc6dc8eb543a9a52d1bea814583726bc8595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0cdfc6dc8eb543a9a52d1bea814583726bc8595"}]},"branch":"refs/heads/master"},"b0b9781a143d1932114fa771317763b3545bc83f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-27 21:18:47.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/719364/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/719364/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/719364/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/719364/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/719364/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/719364/2"}}},"commit":{"parents":[{"commit":"54a644409115a1db22dfb08d8f087f550ddef55e","subject":"Merge \"aarch64/centos8: no fluentd, rabbitmq, kolla-toolbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54a644409115a1db22dfb08d8f087f550ddef55e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-12 18:17:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-27 21:18:19.000000000","tz":120},"subject":"Debian: use Ussuri binary packages","message":"Debian: use Ussuri binary packages\n\nThere is a repo with binary packages for Ussuri. Let us use them instead\nof Train ones.\n\namd64 only\n\nChange-Id: I071eebc4f60094d881a68e201e8c3bb6e6742ff5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0b9781a143d1932114fa771317763b3545bc83f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0b9781a143d1932114fa771317763b3545bc83f"}]},"branch":"refs/heads/master"},"62c8dc95bd79a1b72b8633712847d0ba44c89126":{"kind":"REWORK","_number":3,"created":"2020-04-29 21:09:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/64/719364/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/719364/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/719364/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/719364/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/719364/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/719364/3"}}},"commit":{"parents":[{"commit":"d7a377260fbfb045a4fa11c97a8701f9c7eb471c","subject":"Merge \"Revert \"Add stunnel container for Glance service\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7a377260fbfb045a4fa11c97a8701f9c7eb471c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-12 18:17:15.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-29 21:09:06.000000000","tz":120},"subject":"Debian: use Ussuri binary packages","message":"Debian: use Ussuri binary packages\n\nThere is a repo with binary packages for Ussuri. Let us use them instead\nof Train ones.\n\namd64 only\n\nChange-Id: I071eebc4f60094d881a68e201e8c3bb6e6742ff5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62c8dc95bd79a1b72b8633712847d0ba44c89126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62c8dc95bd79a1b72b8633712847d0ba44c89126"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
