)]}'
{"id":"openstack%2Fkolla~719431","triplet_id":"openstack%2Fkolla~master~I4541c3c070ca70a2ee62f2a0e4ac5d70c57ca6e9","project":"openstack/kolla","branch":"master","topic":"bug/1872432","hashtags":[],"change_id":"I4541c3c070ca70a2ee62f2a0e4ac5d70c57ca6e9","subject":"Fix that the cyborg database init failed.","status":"MERGED","created":"2020-04-13 10:03:18.000000000","updated":"2020-04-16 07:05:02.000000000","submitted":"2020-04-14 13:25:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719431-1586870758137-a88869dc","meta_rev_id":"2a164b49537fe44b3d7e03ebe4d30df0556f9746","_number":719431,"virtual_id_number":719431,"owner":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 13:25:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-13 10:33:59.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-14 11:38:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-13 12:57:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-14 11:38:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 10:33:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-13 12:57:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-14 11:38:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-14 13:25:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f5f0261e7a957698ef7472bbeddd98f557376e4","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2020-04-13 10:03:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fb907a32f335d2b89da0567c443ba4a7901372c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-13 10:33:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/278fdc5997aa42e198982437446e8eb7 : SUCCESS in 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e2fcf23e8e705723d7e01bcf8b909f2418d2897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 12:21:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/646d9a8faaef42dfa3fadb812d3a56e9 : SUCCESS in 25m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d801bc0f42234bc69c3060035bf323a0 : SUCCESS in 28m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5c17681e56fa46cc8b9cacd017f16431 : SUCCESS in 1h 12m 40s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9a6ecb1a0fe34a268314da805d5bfb1f : FAILURE in 48m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1bc585b568694747a138414d04860c9f : SUCCESS in 1h 29m 38s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/dd370b62cdf54b119694a31409238eed : SUCCESS in 57m 31s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/239bc2bc88de4a6ab0e91c1ec4fcf2c9 : SUCCESS in 1h 44m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/500bb51a309a4ec5a0856a559caf74f0 : SUCCESS in 25m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/435c016e7e1348f1bac5ee9235c0f0c0 : SUCCESS in 59m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ec01289a0ce42a2a496bf8cd0894f00 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07436a70b3d3403c8fa0bced409930e2 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f196c09bb2334977b554eb52897e46a6 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7880463126494f4f9388654cf775da04 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cba54ddb9e81402faf71e8797526e9e5 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4501a625fa2d468ba52a0f5471abdf3d : SUCCESS in 7m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2b05e362ad6349309ce2c8b0a2c1ffce : SUCCESS in 1h 59m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4bb06a314a843b4bb517bf3577fcd0d : SUCCESS in 1h 57m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b9c577b0b815459e8d8c583de2562f15 : SUCCESS in 1h 01m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d4c1f3c3e734332a65d5e2a78cda7a6 : SUCCESS in 2h 01m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/75f652d827424d1b9f558eb65da22112 : SUCCESS in 2h 10m 43s","accounts_in_message":[],"_revision_number":1},{"id":"49823bdd2adfcc8c1787c2b0ec7b15d26d53b178","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 12:29:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e5c99ce9fad1402bb9468b5db9e04026 : SUCCESS in 46m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d27e8b1c798c479eb130174a7a93e2c7 : FAILURE in 48m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b47b74ce5d62dcb3570fe2e41c363076f25bbf6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-13 12:57:12.000000000","message":"Patch Set 1: Code-Review+2\n\nAccepting no release note because it\u0027s in an unknown state overally.\n\nBy the sources it looks like both methods should just work: https://github.com/openstack/cyborg/tree/master/cyborg/db/sqlalchemy/alembic","accounts_in_message":[],"_revision_number":1},{"id":"6f66c289cdf9bc9469e3c17672afa3ab4dba2ed6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-14 11:38:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d46193a0822ef8cb54f003d956218a5354384613","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:44:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"801acec6a05743f3552f8e8a58846c32a43c3090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 13:25:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b249cdac58ba4554bf60442c1766b594 : SUCCESS in 30m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a241ce15f7ed4ed7a5739b6e23dfb6ec : SUCCESS in 1h 21m 26s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5478d9fab5d44634b1751545f931c63f : SUCCESS in 1h 28m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11b6fa49975b4a419162a8673822d68b : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e67dea9d01345c2ba62b84b6e93e1b3 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02621888149b481eb9973b9f6aa32124 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae24116a573f4fb0ab15085a8e31c62e : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c45e7e776595429cbfbf593f5077fb88 : SUCCESS in 6m 55s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/984fc1e00e52449898df9c2f88ad9fdc : SUCCESS in 30m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b775bf91af64a65bdff727d318ceac7 : SUCCESS in 1h 00m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/33baa41916fc4463bf9ac2980a766b68 : SUCCESS in 1h 15m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c122cb6cf9200e97f59052da96049e40ead8ba85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 13:25:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5e8cc2b3a4b02fad9263d396edc55068a0a3d91f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 13:30:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1540e4b2a3fe4c1a93bdfe9a3b9f70bb : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7aeaf2d874719e9b7584052439056ad793a7b774","author":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"date":"2020-04-16 07:05:02.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 95d82da7e1d85c840646eb6954f812c946c0de17","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43de719c3ac17085423b381e5938d1bf83e3ff65","revisions":{"43de719c3ac17085423b381e5938d1bf83e3ff65":{"kind":"REWORK","_number":1,"created":"2020-04-13 10:03:18.000000000","uploader":{"_account_id":27614,"name":"ya.wang","email":"me@littleya.com","username":"ya.wang"},"ref":"refs/changes/31/719431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/719431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/719431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/719431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/719431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/719431/1"}}},"commit":{"parents":[{"commit":"8518744133ec9cf5401021c09211fd985c386fe8","subject":"Merge \"Allow operators to use \"ceilometer-upgrade\" parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8518744133ec9cf5401021c09211fd985c386fe8"}]}],"author":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2020-04-13 10:01:24.000000000","tz":480},"committer":{"name":"ya.wang","email":"wang.ya@99cloud.net","date":"2020-04-13 10:03:11.000000000","tz":480},"subject":"Fix that the cyborg database init failed.","message":"Fix that the cyborg database init failed.\n\nThe database init command should be \u0027cyborg-dbsync upgrade\u0027.\n\nChange-Id: I4541c3c070ca70a2ee62f2a0e4ac5d70c57ca6e9\nCloses-bug: #1872432\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43de719c3ac17085423b381e5938d1bf83e3ff65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43de719c3ac17085423b381e5938d1bf83e3ff65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
