)]}'
{"id":"openstack%2Fkolla~719943","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ifae18c407210c12745d29fc4c95dca69aeafe6a8","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1866017-stable/stein","hashtags":[],"change_id":"Ifae18c407210c12745d29fc4c95dca69aeafe6a8","subject":"Ensure proper JSON in keystone bootstrap","status":"MERGED","created":"2020-04-14 12:44:12.000000000","updated":"2020-04-16 09:52:26.000000000","submitted":"2020-04-16 09:49:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719943-1587030582575-eaa21e4c","meta_rev_id":"fb7cca94b36169237ddf6f3d6a3a90ba519b6361","_number":719943,"virtual_id_number":719943,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 09:49:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-14 15:08:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-16 08:35:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-16 08:35:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 08:35:17.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-16 09:49:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d09a0a59d19ae7871cb05d4d79d2bba1df2ea04c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-14 12:44:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"069c76b366304e96496f29b0f46f8bf19563b040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 15:02:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/030f8bc263144c62b17a2732ce0de7ba : SUCCESS in 52m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9ef39daaf47b4bcfb5002e09283c9fed : SUCCESS in 55m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4e8a533a651446fe976a168647be1586 : SUCCESS in 1h 29m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/08b230d40db5455d97c22bd8778d1d54 : SUCCESS in 1h 12m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f39095e118ff401a8214fb9e092cceb4 : SUCCESS in 1h 29m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/30d70fc4ec6040df950a8bcffd62141f : SUCCESS in 1h 02m 05s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1fcb62431c92470883ed98f3ea500e28 : FAILURE in 49m 25s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c904005d5ee54db8aa0a92bcd11aa972 : SUCCESS in 53m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0b0fef3d1414b289bea714a153278e3 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d423949c18714667b82c03cca1e21d68 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d023ed2c09514368aa703f8c09778071 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4b8caa966344a91a6a79b66179335d8 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fbb419a1e80450e9cd8756a4cfb17d9 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24074ad1757641bd8febfca0b4d76b74 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82b3c491579f4e4baa89eea99835683b : SUCCESS in 6m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eedc7001c2e549a7a054ef21e3cae1a1 : SUCCESS in 25m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/40cc718fd2ad4bee95e49333ef2b193b : SUCCESS in 31m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d65061f2326d47f584e2ec69bb2e07b3 : SUCCESS in 57m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b7f56ca274ea4e5b90d215cc0b1e0e6a : FAILURE in 32m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f9685d5b982e42908715cf8d70be2f83 : SUCCESS in 1h 05m 41s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"5abff1463b051e647081363f257d1c9de27c7fa8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-14 15:08:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0454ce762f12b21d628f039867fc924204e8964b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-16 08:35:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44d6241fcf369ac6c045d1f2750f65ba6551a2a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 08:35:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c846dd1f036b6cffeccdc50362035d4f97d16ec5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 09:49:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"63add5183ee34c69216244a47ae90ca64ad0462a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 09:49:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/132eae3fc5494b719b751ac12ec10d4c : SUCCESS in 58m 37s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0e63d1834bf74f20b6637f4bef69fc66 : SUCCESS in 1h 13m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af71eea97c6b49e79b147eac921d6bc1 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ea4b40a1074336ad4c4335ef7eac5c : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42158c00dedc4604944fd1149b851131 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c953644c9cfa4c5a93bfce126c750389 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67a2fc38259d4e7e9fa75b5b284f8837 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7151b192f9f7404d855ecc4510e14e51 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f51ecc00cb554aef8956036f34444468 : SUCCESS in 6m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3005e558c12b4270b4906e5718948c09 : SUCCESS in 44m 47s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"39a3ed253f88d6bf299aa67e0d72094075120d99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 09:52:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68fff8a0661d4e88b6f491c65babdd76 : SUCCESS in 2m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfde3049b09848a4aeaa4930d5e479fb : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab8bcf5a4f4b1a6a8ae698eda80f2a93d00aa84e","revisions":{"ab8bcf5a4f4b1a6a8ae698eda80f2a93d00aa84e":{"kind":"REWORK","_number":1,"created":"2020-04-14 12:44:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/719943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/719943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/719943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/719943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/719943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/719943/1"}}},"commit":{"parents":[{"commit":"80caffa8c81f3d0d79fcdd7682acbcffae3efdce","subject":"Sensu: Remove ceph and process-checks plugins from client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80caffa8c81f3d0d79fcdd7682acbcffae3efdce"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-04 09:22:19.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-14 12:44:12.000000000","tz":0},"subject":"Ensure proper JSON in keystone bootstrap","message":"Ensure proper JSON in keystone bootstrap\n\nThis fixes the issue when keystone-manage output included\nbackslashes and/or double quotation marks which broke JSON string.\n\nChange-Id: Ifae18c407210c12745d29fc4c95dca69aeafe6a8\nCloses-bug: #1866017\n(cherry picked from commit 33d184a731458bb45d1206bdae4fb172629a76e7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab8bcf5a4f4b1a6a8ae698eda80f2a93d00aa84e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab8bcf5a4f4b1a6a8ae698eda80f2a93d00aa84e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
