)]}'
{"id":"openstack%2Fkolla~720159","triplet_id":"openstack%2Fkolla~master~I519c2a59208376ae05e0a6f5fa4de303670f2674","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I519c2a59208376ae05e0a6f5fa4de303670f2674","subject":"gnocchi: install missing python3-wheel package","status":"MERGED","created":"2020-04-15 09:02:54.000000000","updated":"2020-04-25 21:06:29.000000000","submitted":"2020-04-15 15:50:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720159-1586965825532-ef3b6f3c","meta_rev_id":"fdb1e291d10dddda52ccf82e5344a51a5257795a","_number":720159,"virtual_id_number":720159,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-15 15:50:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-15 12:18:29.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-04-15 13:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-04-15 11:46:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-04-15 13:43:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-15 11:46:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-15 12:18:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-15 13:43:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-15 15:50:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8517ebe6b4dd6a06d546e210c43edd0c0e7062db","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-15 09:02:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1860ccd44b454fa98a0e65881cc90b98685a5e28","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-15 09:33:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/184f01e55f0447309b806f4c2a42bdf1 : SUCCESS in 29m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1684653e87f080b309ac1da1457bf1fd770b227","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:45:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9600018af1a94f6db48f270b1a4fb49e : SUCCESS in 53m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a711cafb4e0440b5810fd4a89d2ff1d3 : SUCCESS in 59m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0206f5b46c955ded0a1448b51fe1c8b835172fec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 10:53:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5e4d0b978b11416d9457181edfbeb251 : SUCCESS in 22m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/17cbe11097ce46aaa06f5690e41b2c58 : SUCCESS in 47m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/992b34d7ea094ae48d16086cd118e8f8 : SUCCESS in 1h 46m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7dc2e9d198ee43339982d230c70e2cd9 : SUCCESS in 1h 08m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/50c0c98ce44243ef890b07cc5d81475b : SUCCESS in 1h 15m 18s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/14ea6e7a9c57436895cfe1696f604187 : SUCCESS in 44m 07s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/54b94d473eb14996b69344f4cb93ea55 : SUCCESS in 20m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0daae3c33bc04c21b6a023602aa48c4c : SUCCESS in 18m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/060aad38638543e882e688c61bc1fd7d : SUCCESS in 1h 02m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be63f9d3be8f4ca0b2bb7ee51704384a : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49994c184aab478d93bc0ca81a36f7ae : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af1bf7d2059e49f782066b3509a1d1ad : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7edabc6a19e64d1aa3bd8f5dd6f062b1 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc6fc37b6a75481782a277a4059fa6f7 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b950373cc275438689a464bac08b55ee : SUCCESS in 7m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4bd8aa93efcc40e3a12979f3ff99d259 : SUCCESS in 32m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e47d59d8afc94da887c862c427fef5fa : SUCCESS in 25m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1358843684ba4b1ebf327b97ec4c3d21 : SUCCESS in 57m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c42e802fe054490f94474a401d9f594d : SUCCESS in 57m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a3b7a32a7001438ab8bcf6bacf35a157 : SUCCESS in 1h 04m 51s","accounts_in_message":[],"_revision_number":1},{"id":"802e3338317d3890a545e549f08e800fc3b7514f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 11:42:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2d7516d321f0e6f521035d78dd4d82f41cbbb6d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-15 11:46:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"75b205743d18d90d333d720b006806b6e44c170e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-15 11:46:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5cc49cdf9588d01cc0eec02f6a02e27f2d8d9c0c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-15 12:18:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/2e4c5e4ed05444a9bb5b371cf742beb4 : SUCCESS in 30m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8858f6a014a4b8afc10c2675692956ccd10ea988","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 13:08:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c902b2f02af5428281490c07393ce417 : SUCCESS in 48m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a136f731f41d2c437a59e9af37d5e05f2e058775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 13:38:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6ee2bdc7a81e40c294c41576631d5874 : SUCCESS in 34m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3712ff8948bf4444916cec84253011b6 : SUCCESS in 29m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cacd187240e34d449f02455dbebedc7a : SUCCESS in 1h 25m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/355516a5309346c186065c9aaa2761bb : SUCCESS in 1h 16m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8b0d9532df294b86b24abcd450bc8444 : SUCCESS in 1h 35m 18s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/4154ab1ec23f48e7a2cbcafe9d7e8104 : SUCCESS in 47m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/84b2a466116a4cae8e640883cb87ee41 : SUCCESS in 36m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/089ce7fccbb04cc78ec90ab4cded4c93 : SUCCESS in 18m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b8f2e54974cf48df8e69423aa49922f1 : FAILURE in 13m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ff750d6dfd3412a93953783e805f0c1 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/836e1a302a61463282d87187b6e1638b : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dfd5e51f5f343e99b0bb41d0658dc9d : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7854b87c0fbb425cbcfa4899e51456b5 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce28370f24254615b894c2a4c37e1123 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb051ba5bee34094a68445415ca74fba : SUCCESS in 5m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a603b47e768e4f14958b59b5b9635b92 : SUCCESS in 24m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/44504752df4a406fba9a7bae38cf9ebe : SUCCESS in 26m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e169ebdeb5444caa9bc6be11220745bf : SUCCESS in 1h 20m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/39de351e1768499da1933aa01e63887d : SUCCESS in 1h 01m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec7e4af3bd6b4065a1e7020516cd9e4a : SUCCESS in 1h 08m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b22caa36ed86922768761a686c492c130a35e32e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-04-15 13:43:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4e8432cbe66abfd15c752c7fb0304d0dde224310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 13:43:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3cc53fb1f0e2816a082bf8849bb2720ed8cda3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 15:50:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/97c1a2627ab44e408c74c2a6ed83fde8 : SUCCESS in 36m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ad2016af2414457787c9f9540c472569 : SUCCESS in 1h 14m 28s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c3a5b489635641d6bda053ac44ffe2e9 : SUCCESS in 2h 03m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9211139d89c74bef96a2b9e8566e6d03 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c5342f875264ac39d9e385f91eb98e9 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b4b3b058d0f4f7c8cd441495c4ed6aa : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c456a48cc78c4f7080cddfd4e0d80168 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4930059de166410c9aa4f9a06dbe1dc7 : SUCCESS in 6m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed55176ea0b44fdc9b15fc3a65538ce1 : SUCCESS in 27m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/88058eeda414456289699c73121d6eed : SUCCESS in 1h 00m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f0a43b79573244e395b11096b08b1606 : SUCCESS in 1h 10m 42s","accounts_in_message":[],"_revision_number":2},{"id":"efc9a40327bc96e9bcf3c6640f294b27dbd9c615","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 15:50:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"18aeb45d8f9b00e9595cacfd58606abf37a3da8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 15:52:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f901b461fa443b9ad8bb580af803c73 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":2},{"id":"13a6a9c2177189727a8676dfc8fe5699c4894713","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-25 09:42:15.000000000","message":"Patch Set 2:\n\nIs this required for stable/train?","accounts_in_message":[],"_revision_number":2},{"id":"fdb1e291d10dddda52ccf82e5344a51a5257795a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-25 21:06:29.000000000","message":"Patch Set 2:\n\n\u003e Is this required for stable/train?\n\nNo","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e70187c13a4cb2255c2be79920d2a1db0e2acb30","revisions":{"96546e98c079028cb0547f86215a49a1c2339ef4":{"kind":"REWORK","_number":1,"created":"2020-04-15 09:02:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/720159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/720159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/720159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/720159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/720159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/720159/1"}}},"commit":{"parents":[{"commit":"4cbb9a5bfdd906c6ebc4c2fc5b0ac4431602bee4","subject":"Merge \"Fix that the cyborg database init failed.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4cbb9a5bfdd906c6ebc4c2fc5b0ac4431602bee4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-15 09:01:39.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-15 09:01:39.000000000","tz":0},"subject":"gnocchi: install missing python3-wheel package","message":"gnocchi: install missing python3-wheel package\n\nStep 5/8 : RUN pip3 --no-cache-dir install --upgrade cradox     \u0026\u0026 truncate -s 0 /etc/apache2/ports.conf\n ---\u003e Running in 3a850b4d6319\nCollecting cradox\n  Downloading https://files.pythonhosted.org/packages/30/a8/f4790a3f5af7c7616dda7bd5cddbe31d3cf7c35319b700ff49e14ffa8153/cradox-2.1.2.tar.gz (44kB)\n    Complete output from command python setup.py egg_info:\n    WARNING: The wheel package is not available.\n    WARNING: The wheel package is not available.\n    ERROR: Failed to build one or more wheels\n\nChange-Id: I519c2a59208376ae05e0a6f5fa4de303670f2674\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96546e98c079028cb0547f86215a49a1c2339ef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96546e98c079028cb0547f86215a49a1c2339ef4"}]},"branch":"refs/heads/master"},"e70187c13a4cb2255c2be79920d2a1db0e2acb30":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-15 11:46:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/59/720159/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/720159/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/720159/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/720159/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/720159/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/720159/2"}}},"commit":{"parents":[{"commit":"4cbb9a5bfdd906c6ebc4c2fc5b0ac4431602bee4","subject":"Merge \"Fix that the cyborg database init failed.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4cbb9a5bfdd906c6ebc4c2fc5b0ac4431602bee4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-15 09:01:39.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-15 11:46:07.000000000","tz":0},"subject":"gnocchi: install missing python3-wheel package","message":"gnocchi: install missing python3-wheel package\n\nWhen there is no cradox wheel available on pypi (for non-x86 archs) then\nbuild fails:\n\nStep 5/8 : RUN pip3 --no-cache-dir install --upgrade cradox     \u0026\u0026 truncate -s 0 /etc/apache2/ports.conf\n ---\u003e Running in 3a850b4d6319\nCollecting cradox\n  Downloading https://files.pythonhosted.org/packages/30/a8/f4790a3f5af7c7616dda7bd5cddbe31d3cf7c35319b700ff49e14ffa8153/cradox-2.1.2.tar.gz (44kB)\n    Complete output from command python setup.py egg_info:\n    WARNING: The wheel package is not available.\n    WARNING: The wheel package is not available.\n    ERROR: Failed to build one or more wheels\n\nChange-Id: I519c2a59208376ae05e0a6f5fa4de303670f2674\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e70187c13a4cb2255c2be79920d2a1db0e2acb30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e70187c13a4cb2255c2be79920d2a1db0e2acb30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
