)]}'
{"id":"openstack%2Fkolla~720850","triplet_id":"openstack%2Fkolla~master~I0743137577964f949eef949bad8163a990098bb0","project":"openstack/kolla","branch":"master","topic":"james_kirsch/bp/add-ssl-internal-network","hashtags":[],"change_id":"I0743137577964f949eef949bad8163a990098bb0","subject":"Add stunnel container for Glance service","status":"MERGED","created":"2020-04-17 23:24:32.000000000","updated":"2020-04-26 19:06:37.000000000","submitted":"2020-04-23 11:48:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720850-1587642504034-602cea05","meta_rev_id":"68e61540f32c8159f025d34cc193d44ac9a86b31","_number":720850,"virtual_id_number":720850,"owner":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 11:48:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-22 16:12:27.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"value":2,"date":"2020-04-23 10:00:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-22 16:10:56.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"value":1,"date":"2020-04-23 10:00:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 16:10:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-22 16:12:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 10:00:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-23 11:48:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e587c77528482c79862dba72bba134368f59e9c","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-04-17 23:24:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"037ac7c491d0bbf24f82f79bae4aec5e42467fad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-17 23:55:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/52092c2f49ee4406a3fe6aa78ba3f2ac : SUCCESS in 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e4ec3ff05b14828ae7617d8e2206ecd6e23d3d93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 01:01:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd853cffa6b74064a858239390271e26 : SUCCESS in 44m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c06b1cd82de9476b90f1d78f3aee5667 : SUCCESS in 1h 36m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6dd2dd534cc46f3be1770623757086b : SUCCESS in 47m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b86c8e04540d3cae55f284d46cf566021424abf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 03:33:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/81a1464a60f8461b92b4d515df010ed3 : SUCCESS in 26m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/70f76c16e0354b6596fe638feca2c156 : SUCCESS in 32m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5ba426d9e4ef4143a011430b531be2eb : SUCCESS in 1h 15m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/34dfcee8b4d1429c96ac043856fdc869 : SUCCESS in 1h 09m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/348b000084c04889adae4b23a7771b91 : SUCCESS in 1h 29m 37s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/a56a5becf3be47d69654c71182b79563 : SUCCESS in 40m 49s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f64056b0f5714774a070f6c911a1e761 : SUCCESS in 1h 56m 21s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/136ca841c2434c5c856a1331a90c4241 : SUCCESS in 2h 00m 20s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/496e7643355b4cc6b713986d5faf2951 : SUCCESS in 49m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d3fd0a9d49eb40b188470823cbededf3 : SUCCESS in 47m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dd1c3881d30948938f3e7aa7f71574d2 : SUCCESS in 59m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a872ca87ac6b4a14b9e3555dd5c168e9 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc2aab5827141f8b464c5042a8656a1 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b58596a8c8c442ed9113f0961bcd9565 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8972d5b79dbc485e9811a7193cae6edf : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6797a4e75f954318a0b753f40877b232 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0021e09c214845b0a876e7c177fdf349 : SUCCESS in 5m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b428d7a2ebe34f188f451688f542bc01 : SUCCESS in 54m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f3574644f84f49a696d8ed3256abcf2e : SUCCESS in 54m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8c83f70390b405b93e06486f0fd7ff3 : SUCCESS in 1h 06m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d723c4de69f439d9767d8e97b3cd743 : SUCCESS in 1h 17m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8b0cd6053aa043bdb4cb88b683c0c3fa : SUCCESS in 1h 30m 34s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b28856719f614aa6820e9987e646314f : SUCCESS in 1h 30m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f97ff8ee4c0049c386e01c17b4f64e0e : SUCCESS in 1h 55m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"592a545ab2be4058f7dd178e352f8e93ff65ba65","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-04-18 16:59:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8aa55b48facc2a684fab2bba9b72d1a5b1cbce3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-18 18:47:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e12b5e30bc004205beb29a87adc60dfe : SUCCESS in 29m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ee5d9ff3956925e4f197c7c2663c5fd3b96d44cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 18:52:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c7e1fecdf0644c3a64a9b3db2152ffc : SUCCESS in 44m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5d88ea7adf14d55ba3633201312f13b : SUCCESS in 1h 39m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7294e8069b3344ac831e432cb5b5c220 : SUCCESS in 49m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8219aac77316bdf712cbda6af4bbee8f0ed2cd07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 13:42:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ad86c4d6d91b4f91b4b2143786d21ba7 : SUCCESS in 23m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3b8da15e1fc14e9692d1e9a3a7ed8632 : SUCCESS in 42m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d936b92969fa4bb0a90fb518dee1ddba : SUCCESS in 1h 26m 44s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/17c57737afb4463e83a56027c5b01d89 : SUCCESS in 1h 16m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c1b359236c474334894dcd6f570cea1c : SUCCESS in 1h 21m 20s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/6f844a9087054525901e8217e4af1ae1 : SUCCESS in 1h 01m 24s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/860ba94901bf47cea8eaa4bd9ff2a18d : SUCCESS in 2h 04m 01s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/60f40f35e85a46c49bad4cfbd947c28f : FAILURE in 2h 11m 57s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/417fc4e5d13541909b556aafd8f7af4a : SUCCESS in 1h 54m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0b9fc6f9296f4be1bf9ba79ef1e1e048 : SUCCESS in 24m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6359e2955fc94815b50987be5d4078ce : SUCCESS in 18m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/79f98a255acb4091a83c6d601b301014 : SUCCESS in 57m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d087568926574a129470312d82d45ce0 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fba487c2eb69467ea5b64df9a580af82 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc03402bbfa84ed8bb4b844cebd8e1fb : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74fbb570ce204b8b9ecf274d85cb2521 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b81b6b3f92c43d580a0c5b4dbdd671f : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e2f5b8a9b624f02a8fd8fbd1156d38f : SUCCESS in 9m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d5d155dab21142ac9395a3b6b5bf9c8f : SUCCESS in 34m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4084e6cbcd447bca166f8a59e6f9391 : SUCCESS in 39m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6a96eca7c95b43d0af7fdaae4c16486c : SUCCESS in 59m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f8133ac63a9f44eabf22771e86145e7b : SUCCESS in 59m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b815b2ee3eb2432886d58617c8f5ad68 : SUCCESS in 1h 07m 49s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/af58509beac44b2f95dbe672ee32cfa5 : SUCCESS in 1h 47m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ca90a1554ab64d15bbcc532746808235 : SUCCESS in 1h 41m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2a26f50a7c19c8d555c2676bb49f98892b3a3e9a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-20 06:47:04.000000000","message":"Patch Set 2: Code-Review-1\n\nstunnel can be a general, separate image, inheriting just from base.\n\nWhat about heat? Is not it a bug per se?","accounts_in_message":[],"_revision_number":2},{"id":"de5c21b21b418ce12e44e82e6f698fa7494d3152","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-04-20 15:15:31.000000000","message":"Patch Set 2:\n\nwill update re: stunnel.\n\nFor the kolla_httpd_setup to execute, user must be root (not heat).","accounts_in_message":[],"_revision_number":2},{"id":"b60e14e7ef12b169f5547df8ee521211efd2a17f","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-04-20 16:41:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0dbab88e1c1209b2fe81805013299d61cadb6f57","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-20 17:39:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/78c8470242a04155829ebed039c8bedc : SUCCESS in 29m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1863506e2c716e8c4dfcb8a3ad9ce3aa1180a626","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-20 18:15:36.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nCould you add a release note for the new image, and add it to the support matrix in the docs? doc/source/matrix_*","accounts_in_message":[],"_revision_number":3},{"id":"c52362566da87679d7e3f66017ca7cdfbac13a54","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-20 18:28:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4ed356e529e9280946c9651322fcc94e4ef7151","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 18:33:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/26b13a033a574da0947e4bbbad8848f9 : FAILURE in 57m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/88edaeea3f534d6da9e2c52e5c3aadbb : SUCCESS in 52m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6bc3c636ace445858ad1b91492f5396d : SUCCESS in 1h 41m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/006bd9e46fb94bebb7daf7c6b3d258db : SUCCESS in 58m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"057941cebdafd4b07422c2007d2ea50497b0c604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 19:02:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/29e08d6ee5fd4eb79094c07e802f8087 : SUCCESS in 26m 14s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e39636da6a7d42fa9d6d945d3762ee87 : SUCCESS in 1h 11m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6fb435ec904347aabaf20a2dca301027 : SUCCESS in 1h 19m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3210adb08dfa4ceb8d999083cbb3d292 : SUCCESS in 1h 50m 24s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6f0850d192834e878e6b13c1597d4e00 : SUCCESS in 1h 25m 59s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/16622e4c8ef9466c96a599a765031cb1 : SUCCESS in 40m 17s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/2178fec1961d487589b2b27ecbcabcb4 : SUCCESS in 1h 48m 09s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/96d0be43664947ee87b1eb891293d904 : SUCCESS in 17m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5a7446e323c34152bd2ba2c897a5c54f : SUCCESS in 33m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2add6e1120e346d1b41d0f206aabe733 : SUCCESS in 1h 11m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce7ea3e418ae42a2a7ec3de72c7ea10d : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497037ec8d8449068c6077dceb146600 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa8f39e005ed4c978753166c6c490e37 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8d9f3c1dae142039b0c557254a7a961 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88a4d77906f64705984044a06477ce0f : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8fb6b9d2a384521983bf7bb4c9ef9f0 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1479a9c2482a4e58bf7770e6c56f85bf : SUCCESS in 1h 09m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/afb9013bf61e4ace9d857ec47bfb8208 : SUCCESS in 22m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7ed52dbc91541fa95df9a621a17921f : SUCCESS in 1h 00m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/11da72584f694f17b26c5b3583a6ede3 : SUCCESS in 56m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d51c4f7f55234119b3a8e1ae8ffeced7 : SUCCESS in 1h 34m 30s","accounts_in_message":[],"_revision_number":3},{"id":"17640a45c7c91faa619fbeb816838cc2e3b69fa1","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-04-20 19:37:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a859d85427b41355c5c622a42409b41533a7050f","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-04-20 20:52:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e47c8b93249e2b62150c29cba6d396e821d35ebd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-20 22:08:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/68504398b7814735a97e529bad833df8 : SUCCESS in 1h 11m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"88c2172e000ac80d555e0b6f0de4edd776bafbff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 22:47:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/25d37e10006c404ab32621233b8d2c52 : FAILURE in 47m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7c8039ec80b4beb965e340550f636a0 : SUCCESS in 45m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/75d18eab3ac04529927626a610a21188 : SUCCESS in 1h 37m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4227fe34b6d648148865d1126d53c845 : SUCCESS in 50m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4eba68bf3f2bd1ea673d625bdca2a8d525436ba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 23:03:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/00494e6bc9f64d619a58a5bc69b6a12b : SUCCESS in 29m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d148147ca85d483cbafb4b04fdf4ae41 : SUCCESS in 30m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c3a504ea19df45bdb2e475b8c6d3fce9 : SUCCESS in 1h 21m 07s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b00d0f3568434715af61d938327e2a2b : SUCCESS in 1h 39m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ec3dd7fc7d8b460e96d50cbc179535e6 : SUCCESS in 2h 06m 22s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/533cc1db134d4f0ca3e5a9d42a66243f : SUCCESS in 1h 18m 34s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/cb2af1ec19624346b272060efca81e71 : SUCCESS in 2h 04m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/922bff048f334353932026f64b0aaf2a : SUCCESS in 21m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6be25511922c4867b64e74cb86359970 : SUCCESS in 1h 11m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/54fb3a0d215144919626957e81b31211 : SUCCESS in 1h 37m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b7b502f1e924c53a4bf3b91145d32fe : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69714043616640baa036c95f201db3cd : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/581e964362da4414b7b43a73e86d2b95 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c95115533d764ff88fb6c3379bc60bc9 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4833a3caf7946568f8862300adb7d44 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29791b4e8eb54dc8961a5e5d611ae110 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df5b6c47f80d485fb46e0f975ba8d211 : SUCCESS in 5m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8bc98d4609834886b77795f15cb41871 : SUCCESS in 24m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cabf1c0a7c048ffa4389d5f11d03900 : SUCCESS in 28m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/33686aef72d54b838c118fd0d9abe905 : SUCCESS in 59m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6a0f2ffffe9347df9d0c8ed952d2c6e0 : SUCCESS in 58m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8bbc133fc93242b4916510d6cb117875 : SUCCESS in 1h 33m 23s","accounts_in_message":[],"_revision_number":4},{"id":"78f93b9f249327b7cec363a952fe2265b1098fc6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-21 10:23:31.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d0b88df3a596f890d4c6722be948928be52c6a45","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-21 10:24:01.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"1899dcf0dbac28edfd6ad431724fbf2988f75b70","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-04-21 16:20:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f6cc3fa2f9ec60731898b15bdbb256c911ec2853","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-21 18:04:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"40b124cc645b27a129a32f6b1a165646e12fb0ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 18:05:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/4e0da3b68217461c840af534ebaa0dcd : FAILURE in 50m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/66dcb4af8e434c8fafa779f0790b3d35 : SUCCESS in 51m 09s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8c2db0358c4c4d2188c131399a6cc795 : SUCCESS in 1h 39m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/411f14e9c1f2411a8d3fa360b1eaf6d0 : FAILURE in 47m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9aff982d295a848fb29a2f10621337e931725a69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 18:20:07.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0691a3cf5ede4f8f836c1c9d776b23fc : SUCCESS in 50m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5a70f86f9bb44ecf832d573d6082d7df : FAILURE in 50m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a9f2e98b97cd44b2a0ef1eece4025e81 : FAILURE in 33m 12s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e1160c9931054c67afec32794f30ae48 : SUCCESS in 1h 13m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/86123f5aae1e4ba5a0c93c2207688518 : SUCCESS in 1h 21m 06s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/5ded39a146164e47991c7334287514bf : SUCCESS in 44m 51s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8a00c75a8a9f4aafb59ad70319402e61 : SUCCESS in 1h 44m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/29460bd687db463b9d81caef4aee69ae : SUCCESS in 56m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5d08600443d64b6daf5ebfa70737f5af : SUCCESS in 50m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a8d13d5607394ecb87b5c176a8fbe809 : FAILURE in 28m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd71024b51bb48328691b3697c5f6efb : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/328829f0a72d473f95e8b58a39ef377d : FAILURE in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da236834a3834d4fa5656372a5bd5317 : SUCCESS in 8m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/512f041af3234803a11e979a2c5ace0a : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c19acb49e2d74667b6915cfa738667c5 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b451be3aa804e49a8eeca76b7294511 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90d3e122418b4b518e9c472728ce5faa : SUCCESS in 6m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/567217a7bb544d02afb76427a71a7294 : SUCCESS in 26m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/701cddcd7f584fec8b701faa4c68335c : SUCCESS in 24m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/90e3a012ffa2411ea7f1e124cbd16597 : SUCCESS in 1h 01m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d80d440e2cd34835be2638d5921177ae : SUCCESS in 57m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4dac6f095582448ebdab58db271bc970 : SUCCESS in 1h 16m 41s","accounts_in_message":[],"_revision_number":5},{"id":"19a888e25ca081d907c18dfeb5d4031eb68b8ad2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-21 19:17:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/934372ee5a7f4d9087890e5695559372 : SUCCESS in 27m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e400737942ec6b344fe09ca990d64e166a59c1ea","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-04-21 22:25:28.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"335dc4871e7c8174ce77746282af71c1e57a65ab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-21 23:03:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/95e641cd75cd400ca9895df0e0da2d31 : SUCCESS in 32m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9d094b3ba7d4cc592234a918fe9ec391b4eee0f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 00:10:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b82cc5944a8046dd83005507f9cba0e7 : FAILURE in 51m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5305727386484226bc7beae626bb80c5 : SUCCESS in 50m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f73be26e98fa4c98b8b63c7485dc85e5 : SUCCESS in 1h 40m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9501a3a3845f4e18addf89e1995d2162 : SUCCESS in 47m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a5a911b7d48a55ba43b81bfc0d9c01e82ab3eaf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 00:50:40.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3a7f5faa8ec8499981f6e3816c8330e0 : SUCCESS in 23m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/669d7b91f0584349a2e832a02c72948d : SUCCESS in 54m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e00c4718c7144ebd8d2a7e824817d5d6 : SUCCESS in 1h 08m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cc50cef2a0684b948091bcc77ef0ea86 : SUCCESS in 1h 11m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/18c0199e53494420b3653ac819a6df40 : SUCCESS in 1h 35m 43s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/cd026b2cac914f97ba207991e27632f8 : SUCCESS in 41m 49s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/dbd4472aba7a40b9bf4fe3112ce64c64 : SUCCESS in 1h 54m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/57d740a98df7478cb8e4bda245c809cf : SUCCESS in 21m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b29c0a1b8601479d90dc974d54bf68b7 : SUCCESS in 20m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a9dd63bdf32a4a08ace46dc7935d5eb5 : SUCCESS in 1h 04m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c20c9f86b2d44f14b09f2a298d02c754 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6e86b44e8ec4d20b99a6b1de79cfe82 : FAILURE in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d9f5657b7694787a43eb48014c42c58 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6b6764999bb4a598ad078cf7dfe5bce : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/853647373c0a4abf8c453698184d2113 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b5711163d24ba58866b0a4888145cd : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/170466e9ee41434b94988b8ecbc63f24 : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3671fb62410d4379bc3f37d25575dfb5 : SUCCESS in 24m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb21b709f3c248539724d72b8e7c471d : SUCCESS in 46m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64328157bd6641598c6ed96a59abd8b8 : SUCCESS in 1h 11m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4563558c081f41248c5d7f60806eb15c : SUCCESS in 1h 05m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/606c49aced0743cf935c7a03ca8b5374 : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":5},{"id":"183957515335d7f6f89d8679b7d2441f81b00d5e","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-04-22 00:54:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"109c5b27fda09bbbcd1695f0f0297774bbdc29de","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-22 01:32:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/4900270853e1417cb21c974a65e0d42b : SUCCESS in 36m 09s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d5797bb1fcb192ac6e24e6b7316063d1450dc120","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 02:41:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c45ebbf4b874425a8ca7b761f6074c9 : FAILURE in 52m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bbcd7cdebedf4a1a90d9cc9ad1561e56 : SUCCESS in 51m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4aea746a0c4e46839feba32999445192 : SUCCESS in 1h 41m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b3ed0a56525943d9a3bd658aaed8d7f8 : SUCCESS in 48m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"10c63faf1a4e84a99c098f69facda1f81a1960bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 02:48:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e858dd70ce6b40bcb34aead20aa927f7 : SUCCESS in 22m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b7c46d755efa433d8c78278d33d867ef : SUCCESS in 28m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/86ecb835abde4b7d89e4a18a5ca81077 : SUCCESS in 1h 14m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/07ffc19874ff4d2dbbb3bf9f6e681818 : SUCCESS in 1h 10m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c7f30a1b21b043c1ae283c203803ccd2 : SUCCESS in 1h 19m 14s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/70d7cce74cdb422daa5181e05d0715b9 : SUCCESS in 43m 40s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d9a393f06d4540109ddb81d36094df18 : SUCCESS in 1h 48m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fb8c6456cb6f4b1ca743a50323720bba : SUCCESS in 18m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0972ad963a2e486a93863f3303acc67c : SUCCESS in 19m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/16291015b65c45268b296544ef94e1b4 : SUCCESS in 58m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fe86194192a4d7aac4dc8f02f98d527 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6de333e54d646e58ce53079299c98d3 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e9128ed5b434737bef157df222e7b8a : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d77ae9519f2b4cd286df081a45767d10 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98d2b42d944b442a94cec7d63c9e4167 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abcbdb5704d94c6b925b5ed332aff941 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92de5428d44241f78577ca5e6f2bac52 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/482eb154603441de8e19b4636c86a917 : SUCCESS in 20m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2f045315524442aa9db40a724b9289f : SUCCESS in 25m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cd25fee910f4118b30d744fdbbbf7f3 : SUCCESS in 59m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94e933d4e7ee4d6a97b8bd90fd3b499f : SUCCESS in 55m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92577c5c759742aba21b74dc62b9f434 : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":6},{"id":"375b73b351544efe99dc760cb6962e2606b32a15","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-22 07:21:36.000000000","message":"Patch Set 6:\n\nI think we should have the heat change as a separate one. It\u0027s convoluted now.","accounts_in_message":[],"_revision_number":6},{"id":"78cce9b2ee1c9db322601b4eb7ca748783acb900","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-22 07:23:20.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1ef6f2d6a226e92b5c89f68c65e17092a8f0b5aa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-22 07:23:28.000000000","message":"Patch Set 6: Code-Review-1","accounts_in_message":[],"_revision_number":6},{"id":"131484d5f89848a1252094e85765abc3de964af9","author":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"date":"2020-04-22 15:13:10.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"57ea4ae8952b19488e825c45f67a167912e74bb6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-22 16:10:56.000000000","message":"Patch Set 7: Code-Review+2\n\n\"Initially\" :-)","accounts_in_message":[],"_revision_number":7},{"id":"03ff16ad433ecb0be98d83451417156fa066a984","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-22 16:12:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/64742028079d4599b0bac040bba2f84b : SUCCESS in 33m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"de7bf5dbb585ed9ab00b58bc38bbfafee61905f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 17:02:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a86b160c43434859a58ea63c5b3aaec3 : SUCCESS in 1h 04m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3fb753fef9334549998f4d59c70ba76b : SUCCESS in 1h 05m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/218742b243b649be8cc6a0a2313f1e17 : SUCCESS in 1h 22m 45s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a1335229155a4a60b1951644b321e7f2 : SUCCESS in 1h 45m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/276488dd185247279f202736b39256d8 : SUCCESS in 1h 23m 16s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/26aeef9560e44d60b32bfb5b559e2e3f : SUCCESS in 42m 58s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/11f2a7a2a92a4ca28bc8626d60768e04 : SUCCESS in 36m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f3d47c32caae43b4afd0e760325ec7fb : SUCCESS in 44m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b18ebef2e16a4ab895cb206979fb9a46 : SUCCESS in 1h 09m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c829cd1b50e4f69aee20ec067653acd : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43a8c663e4d344c4b66033edfb9e4cb1 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85a780ed29e440e9bd22d2cd28d19db6 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/471cc4b0704246c9979459ae1709d5f5 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ccdedb73fb84d35ae758069ba0db50e : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ebe931acb1840e39a2f8bec091f8a8e : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5af92427535d454b88bb078863786941 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/454fa6af51094e82a2ce5369a32af774 : SUCCESS in 22m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/583f55be8c844ad682cc87b30f0e5900 : SUCCESS in 33m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0cd89294698b45598397bd4850fd7512 : SUCCESS in 1h 14m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/42fbc88473d74a08b078eff2a9a29fed : SUCCESS in 1h 01m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9038138b63b44999188e88557377037 : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":7},{"id":"e5873e711164f487abdc42a248116398b279b696","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 17:33:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2afe8c760a7418a8744430b6560cc01 : FAILURE in 1h 11m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a72a48e7c8694256bf82b271d1fe36c4 : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/55c25ac6ec314d2ba1cfc6de095942d8 : SUCCESS in 2h 13m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c6585bd1d244aa3b3553ff21eb0e46c : SUCCESS in 1h 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fe8a0de6d68bb0110f97e0ef7a672fc4a704f301","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 10:00:17.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b0804f07e51b7070eee903fd532b217c8ea191e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 10:00:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a83f87a813d4cd4a4625b8fd8484fe5ac93be798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:48:23.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6ceac064e1c24fa6b7522a32197d0703 : SUCCESS in 31m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7a86b230556b4bda83299d9f0421bb18 : SUCCESS in 1h 29m 17s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8e15c747c3184dc5af9fee5d68818ef5 : SUCCESS in 1h 26m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/079f7c33366b4110a283f6ee33fbc369 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8b2c24b9d9417f8cce626f7016677c : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab010db5d05b4d1c8bf6893ff05a53a7 : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e2a682074eb43719e2ce2254fa8d524 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a53ff87dee40d780c0b22c6dbba77a : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b99153936b1846eaace0178add05a0ad : SUCCESS in 7m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2c2a97beb78489eaf407b9cd5fd8bd2 : SUCCESS in 1h 11m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/75825027836545f5b1240babb073b96a : SUCCESS in 1h 36m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9e4d601b28ef4828be20966ac4686292 : SUCCESS in 1h 45m 33s","accounts_in_message":[],"_revision_number":7},{"id":"6e9203f821019df1d937904494afd15c889358da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:48:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"405324db951300821aa401b87519196c095cc324","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:50:07.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d3068ee24c4e75a5ef4c525b139779 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a26177bbb16342ada03d081fd4234ca5 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":7},{"id":"11b5616b9c3a4517520c4502f28d5c772b52d4e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-26 19:06:37.000000000","message":"Created a revert of this change as If1140b293423ad3173cb3a4e1f19abbfecdaf2f6","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"935adb59131caf66455a7d59c2af7cc42186e282","revisions":{"fb1c09cf057b2b5eea5c4292438712fa29aef59f":{"kind":"REWORK","_number":1,"created":"2020-04-17 23:24:32.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/50/720850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/720850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/720850/1"}}},"commit":{"parents":[{"commit":"b6d152e92dd7a6b59f3b9f248735dacadd26797e","subject":"Merge \"CI: add kolla-ansible-debian-source-aarch64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6d152e92dd7a6b59f3b9f248735dacadd26797e"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-17 23:24:17.000000000","tz":-420},"committer":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-17 23:24:17.000000000","tz":-420},"subject":"Add stunnel container for Glance service","message":"Add stunnel container for Glance service\n\nThis patch creates a stunnel container for Glance service to perform\nTLS termination.\n\nChange-Id: I0743137577964f949eef949bad8163a990098bb0\nPartially-Implements: blueprint add-ssl-internal-network\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb1c09cf057b2b5eea5c4292438712fa29aef59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb1c09cf057b2b5eea5c4292438712fa29aef59f"}]},"branch":"refs/heads/master"},"3d0fb8b28652aa4f41f88a05df9ba2ae0355cea2":{"kind":"REWORK","_number":2,"created":"2020-04-18 16:59:03.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/50/720850/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/720850/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/720850/2"}}},"commit":{"parents":[{"commit":"b6d152e92dd7a6b59f3b9f248735dacadd26797e","subject":"Merge \"CI: add kolla-ansible-debian-source-aarch64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6d152e92dd7a6b59f3b9f248735dacadd26797e"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-17 23:24:17.000000000","tz":-420},"committer":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-18 16:58:49.000000000","tz":-420},"subject":"Add stunnel container for Glance service","message":"Add stunnel container for Glance service\n\nThis patch creates a stunnel container for Glance service to perform\nTLS termination. Also removes \"USER heat\" from Heat API containers so\napache setup executes as \u0027root\u0027 user.\n\nChange-Id: I0743137577964f949eef949bad8163a990098bb0\nPartially-Implements: blueprint add-ssl-internal-network\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d0fb8b28652aa4f41f88a05df9ba2ae0355cea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d0fb8b28652aa4f41f88a05df9ba2ae0355cea2"}]},"branch":"refs/heads/master"},"3c0053350c8a24509ad41de66ef0f425fe884078":{"kind":"REWORK","_number":3,"created":"2020-04-20 16:41:06.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/50/720850/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/720850/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/720850/3"}}},"commit":{"parents":[{"commit":"b6d152e92dd7a6b59f3b9f248735dacadd26797e","subject":"Merge \"CI: add kolla-ansible-debian-source-aarch64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6d152e92dd7a6b59f3b9f248735dacadd26797e"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-17 23:24:17.000000000","tz":-420},"committer":{"name":"generalfuzz","email":"generalfuzz@gmail.com","date":"2020-04-20 16:41:05.000000000","tz":-420},"subject":"Add stunnel container for Glance service","message":"Add stunnel container for Glance service\n\nThis patch creates a stunnel container for Glance service to perform\nTLS termination. Also removes \"USER heat\" from Heat API containers so\napache setup executes as \u0027root\u0027 user.\n\nChange-Id: I0743137577964f949eef949bad8163a990098bb0\nPartially-Implements: blueprint add-ssl-internal-network\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0053350c8a24509ad41de66ef0f425fe884078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c0053350c8a24509ad41de66ef0f425fe884078"}]},"branch":"refs/heads/master"},"b242f13c3db664341fc2f97eb1fd4542698035eb":{"kind":"REWORK","_number":4,"created":"2020-04-20 20:52:29.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/50/720850/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/720850/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/720850/4"}}},"commit":{"parents":[{"commit":"b6d152e92dd7a6b59f3b9f248735dacadd26797e","subject":"Merge \"CI: add kolla-ansible-debian-source-aarch64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6d152e92dd7a6b59f3b9f248735dacadd26797e"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-17 23:24:17.000000000","tz":-420},"committer":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-20 20:52:19.000000000","tz":-420},"subject":"Add stunnel container for Glance service","message":"Add stunnel container for Glance service\n\nThis patch creates a stunnel container for Glance service to perform\nTLS termination. Also removes \"USER heat\" from Heat API containers so\napache setup executes as \u0027root\u0027 user.\n\nChange-Id: I0743137577964f949eef949bad8163a990098bb0\nPartially-Implements: blueprint add-ssl-internal-network\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b242f13c3db664341fc2f97eb1fd4542698035eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b242f13c3db664341fc2f97eb1fd4542698035eb"}]},"branch":"refs/heads/master"},"5680dd00ee0bc9180ff8dadeb3d050c56bd6259d":{"kind":"REWORK","_number":5,"created":"2020-04-21 16:20:37.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/50/720850/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/720850/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/720850/5"}}},"commit":{"parents":[{"commit":"b6d152e92dd7a6b59f3b9f248735dacadd26797e","subject":"Merge \"CI: add kolla-ansible-debian-source-aarch64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6d152e92dd7a6b59f3b9f248735dacadd26797e"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-17 23:24:17.000000000","tz":-420},"committer":{"name":"generalfuzz","email":"generalfuzz@gmail.com","date":"2020-04-21 16:20:31.000000000","tz":-420},"subject":"Add stunnel container for Glance service","message":"Add stunnel container for Glance service\n\nThis patch creates a stunnel container for Glance service to perform\nTLS termination. Also removes \"USER heat\" from Heat API containers so\napache setup executes as \u0027root\u0027 user.\n\nChange-Id: I0743137577964f949eef949bad8163a990098bb0\nPartially-Implements: blueprint add-ssl-internal-network\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5680dd00ee0bc9180ff8dadeb3d050c56bd6259d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5680dd00ee0bc9180ff8dadeb3d050c56bd6259d"}]},"branch":"refs/heads/master"},"4d1a676573223e08408fd2eb6c286f1bdc93fb3d":{"kind":"REWORK","_number":6,"created":"2020-04-22 00:54:31.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/50/720850/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/720850/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/720850/6"}}},"commit":{"parents":[{"commit":"b6d152e92dd7a6b59f3b9f248735dacadd26797e","subject":"Merge \"CI: add kolla-ansible-debian-source-aarch64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6d152e92dd7a6b59f3b9f248735dacadd26797e"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-17 23:24:17.000000000","tz":-420},"committer":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-22 00:54:21.000000000","tz":-420},"subject":"Add stunnel container for Glance service","message":"Add stunnel container for Glance service\n\nThis patch creates a stunnel container for Glance service to perform\nTLS termination. Also removes \"USER heat\" from Heat API containers so\napache setup executes as \u0027root\u0027 user.\n\nChange-Id: I0743137577964f949eef949bad8163a990098bb0\nPartially-Implements: blueprint add-ssl-internal-network\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d1a676573223e08408fd2eb6c286f1bdc93fb3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d1a676573223e08408fd2eb6c286f1bdc93fb3d"}]},"branch":"refs/heads/master"},"935adb59131caf66455a7d59c2af7cc42186e282":{"kind":"REWORK","_number":7,"created":"2020-04-22 15:13:10.000000000","uploader":{"_account_id":30810,"name":"James Kirsch","email":"generalfuzz@gmail.com","username":"generalfuzz"},"ref":"refs/changes/50/720850/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/720850/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/720850/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/720850/7"}}},"commit":{"parents":[{"commit":"b6d152e92dd7a6b59f3b9f248735dacadd26797e","subject":"Merge \"CI: add kolla-ansible-debian-source-aarch64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6d152e92dd7a6b59f3b9f248735dacadd26797e"}]}],"author":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-17 23:24:17.000000000","tz":-420},"committer":{"name":"James Kirsch","email":"generalfuzz@gmail.com","date":"2020-04-22 15:12:37.000000000","tz":-420},"subject":"Add stunnel container for Glance service","message":"Add stunnel container for Glance service\n\nThis patch creates a stunnel container for Glance service to perform\nTLS termination.\n\nChange-Id: I0743137577964f949eef949bad8163a990098bb0\nPartially-Implements: blueprint add-ssl-internal-network\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/935adb59131caf66455a7d59c2af7cc42186e282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/935adb59131caf66455a7d59c2af7cc42186e282"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
