)]}'
{"id":"openstack%2Fkolla~721251","triplet_id":"openstack%2Fkolla~master~I5610f56cd0c3419ee5b1bff02655ff864614de7b","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I5610f56cd0c3419ee5b1bff02655ff864614de7b","subject":"use Ansible 2.9+ binary packages","status":"ABANDONED","created":"2020-04-20 12:17:11.000000000","updated":"2020-05-17 13:34:05.000000000","total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"22291a883dfb3a62c66aadb9e5ec57d68bec217b","_number":721251,"virtual_id_number":721251,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2020-04-27 10:26:48.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-22 12:42:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-22 12:36:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":-1,"date":"2020-04-27 13:41:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-2,"date":"2020-04-26 04:08:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2020-04-27 13:41:13.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2020-04-28 13:48:06.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 08:51:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-22 12:36:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-22 12:42:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-26 04:08:55.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-27 10:26:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"10ce124fccdf3e4879115a962729fc722446e7c0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-20 12:17:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08c1e9afcdc4f952d7a171260c3025874f4a38fe","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-20 12:26:27.000000000","message":"Patch Set 1:\n\ncentos8 can be moved by installing centos-release-ansible-29 package.","accounts_in_message":[],"_revision_number":1},{"id":"5027cae447ae68d8bfa6a32bce9adc88259aefb0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-20 13:06:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/21c1eb2b8e714138a28325f6006f6e8b : SUCCESS in 28m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb52399bb608f4e51d811217dd97f4bfa71235f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 22:05:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b4d0eb50eaf64757aefb49faf7c36b49 : SUCCESS in 44m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cfebbf625a7d41eb960ea4f08828c125 : FAILURE in 12m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e46f4d71ab244a69efe74715ab1effe : SUCCESS in 48m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f48fc24354e8f1971a51e85043600a687fa45604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 23:08:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/95ed081092db4f06a9ee528a936d7aa9 : SUCCESS in 25m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ae55bc3c43b446c7872349844a4769b2 : SUCCESS in 30m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/01228b93213e416fbdd3170a3fd09b51 : SUCCESS in 2h 06m 19s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4dc8aa58819948eeb0057530001a85af : SUCCESS in 1h 16m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e10071736e3e401cb630ec80ab9005a9 : SUCCESS in 1h 14m 08s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d73d37e8c21b405f8796a073e17af40c : SUCCESS in 45m 50s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/ba4d6c0c77d640fa994467adc378ccaf : SUCCESS in 29m 51s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/71c4215308164ab69c93ef97422b173f : SUCCESS in 1h 52m 17s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/24fe57d2f1cb43baa9a110979a42e445 : SUCCESS in 2h 33m 43s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/cf78e4631f0a43f685dcaba1c887f512 : SUCCESS in 2h 01m 55s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/3b5c35f5dd8c4010861299c13211ebde : TIMED_OUT in 2h 07m 45s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/8bd7772f207e480ea0887428157c9b30 : SUCCESS in 1h 27m 19s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ebfa87566d914775989072f00a4288b3 : SUCCESS in 18m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a07c15bb86ab4deabdc47673c1bf7ee3 : SUCCESS in 1h 20m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ffbb9df69e1c4c3e81b23208006dc13d : SUCCESS in 55m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f23514863bf472396a64959eb04d5e9 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640ba1c47df44e3f845f109f0a551c5b : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b504a168db034618b0eb2d5527b6c4e2 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cf2bb75478f441fbafa6881500880ee : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7e7a3c4a1e84f1e96c17223dc64d1b7 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fa3903dd8c4496a6c260e1dc7b72bd : SUCCESS in 6m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bf4ec07f14cc43cb9cea56cbc1ddc306 : SUCCESS in 31m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6689c3eb1cb40a0a58fa1b95e8cf310 : SUCCESS in 24m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9210fc11d4e94196a666d32168bd96fa : SUCCESS in 1h 02m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fa71223889474aa1860012dc603d9b3b : SUCCESS in 1h 02m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d75e95fe076d45d4bf29064f392538e8 : SUCCESS in 58m 36s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5e34e573df694b02aae134c05eb66810 : SUCCESS in 1h 25m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/bb003a9e37974abb8dc05418fc632418 : SUCCESS in 1h 55m 01s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8b493a506bad41e8a1ccdbfd96e3eb9b : SUCCESS in 2h 08m 37s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/59893b4cc2c246c6b8b937a8bdd8de09 : SUCCESS in 1h 02m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74b7e59855eec28ab7586488577abf6f39ff1da8","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-21 04:48:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"57b19616361b1244fe7c568ce93464ad428ec959","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 07:08:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce710420c33cee934e882b3ce2776d8507c6e437","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 07:08:23.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"db64c4898d63f7de8c7b9e74557904ed3188e030","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-22 07:37:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b85a4c9055664cc88291aee134a828fb : SUCCESS in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cdfdb018b0caeb3c2e91f9a30b15686d6f90d94d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-22 08:51:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4610e3c1871a3a215c6ce5c102d3412881ddb66b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 09:09:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/df11fc453a7348ed8a28bbcd6603a6af : FAILURE in 13m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/76ae270f8a4f493f9a4a9d78e5473218 : SUCCESS in 1h 00m 56s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2031086fa2754d9ea23a00ff0188be5e : SUCCESS in 1h 52m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/84588ee8f2554d93804b636fb85af936 : SUCCESS in 57m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"299cb3303795ad18f644cd5f5c481077919d1136","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 09:55:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7d6264ec4dd8d8d42d3ed67f6e4d269b985c4749","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 09:55:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"525c582649db5a22de9ffb683aa65f2f0eef2bc9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 10:20:28.000000000","message":"Patch Set 3:\n\nc8 has wrong baseurl in repo file. Problem got fixed, new centos-release-ansible-29 in progress","accounts_in_message":[],"_revision_number":3},{"id":"74e04a39c64a6ab9bf7ddda9d068578f9608f3a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 11:54:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8090d1c7e0a4ac48402b45cfb8c316c : FAILURE in 1h 01m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/64c02b7327ea45608436a51a44e5d99d : SUCCESS in 59m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1100db5717954fb3a1556560e4e0eecd : SUCCESS in 1h 53m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fcbdc92ab06441f1a6ed584526d9290c : SUCCESS in 58m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"10aebc96f383fa2f85dcb753eb453ea6bdab6919","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-22 12:36:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/73caf00ef83f4a498479faf41cea70b6 : SUCCESS in 31m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e3cd2bf416d504a72abdc97a1bd72eefa34fee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 12:42:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5e0ebe91670e482daafed66e771ed0c3 : SUCCESS in 31m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/39820b04e3ec48d7a7820e51d9012fb7 : SUCCESS in 39m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b6861156d40f4081a12ce2a47916a7ee : SUCCESS in 1h 16m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0a86460898f14a77bc0050c9e13a3059 : SUCCESS in 1h 32m 10s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6e98681a9dc24b7bb7f69bd8f927c47c : SUCCESS in 2h 03m 05s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/885cb514c72e43be8fa971cbd54dc6db : SUCCESS in 50m 01s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/3728923f14fa4578badd33ebf7e33d4f : SUCCESS in 30m 47s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/3c292362d5864c1bae6d1fee7e4932b3 : SUCCESS in 2h 16m 22s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/577773edd951421497ba7ffcb007f75a : TIMED_OUT in 2h 44m 28s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/3ba73d85c7b84bcfb5782520b6063a79 : SUCCESS in 2h 04m 14s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/b4445749d568453ebf4a74c5b84026ca : SUCCESS in 2h 00m 41s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/dc581a3a2a5448aea706f9663d4ddbd0 : SUCCESS in 1h 32m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5df4a6cad54d408daa6735c543e678b7 : SUCCESS in 53m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/77b9ca1fb03e4da9a5ee06a4a21e9ccb : SUCCESS in 21m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b23c10f1ea9047a197a5747b0bfd6029 : SUCCESS in 1h 15m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94418ce4f103444bb3a8d9bf7202095a : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f79a5fd184d8488dbb88180c5ced22ca : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6799b7857d74e02a7c9f087886fc82a : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9032f49fac174ee794229c97b5cfe07f : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4301c38720924098846129b06b67ce12 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52489df104574e788544a2e3b43cb8e8 : SUCCESS in 7m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f881a72666cb4a539645ebcd9d0a1df6 : SUCCESS in 42m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8724872456814902bf916ab4daa9e366 : SUCCESS in 40m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/34c26e7bc55449c4b33b3b493f0b6bdd : SUCCESS in 1h 17m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1cb802a084804c788c307a26255b390f : SUCCESS in 1h 14m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ffcbc6c451e046bb8ad940871a9d4fae : SUCCESS in 1h 11m 35s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/fbfc9cdb17f34af18e757e25751b8c58 : SUCCESS in 1h 33m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/820acc39b2cd47e696e2962a68c57382 : SUCCESS in 2h 20m 48s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/272b03bbbf334333a13b626eabfb60c5 : SUCCESS in 1h 38m 12s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/bee1f84ccc674a9eb3aac24559fca6fa : SUCCESS in 1h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"12f4c84a512c01645d2efb50b351e6d92dd6b65f","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-23 10:21:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8cfd02e9eaa566b5b86e78544ce94ce6d2bc0f84","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 10:29:18.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"10dc3f9a8d18dfc083e89b497956628ac81b05c7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 10:29:32.000000000","message":"Patch Set 2:\n\nMaybe roll into the next patch?","accounts_in_message":[],"_revision_number":2},{"id":"b94697837b2ee0d9d5a1aa0861300c6bfcd8d8a9","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-26 04:08:55.000000000","message":"Patch Set 3: Code-Review-2\n\nI found out something interesting. Ansible from this repo has dependency with Python 2.7.\n\nhttp://paste.openstack.org/show/792721/\n\nDo you want to continue? [Y/n]\n\n:)","accounts_in_message":[],"_revision_number":3},{"id":"92ffe8238c375f9235bfc16574958580e0a620da","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 10:26:48.000000000","message":"Patch Set 3:\n\n\u003e I found out something interesting. Ansible from this repo has\n \u003e dependency with Python 2.7.\n \u003e \n \u003e http://paste.openstack.org/show/792721/\n \u003e \n \u003e Do you want to continue? [Y/n]\n \u003e \n \u003e :)\n\nGood catch.","accounts_in_message":[],"_revision_number":3},{"id":"76d36a3398966d433ce0a637344088f7d6af4b10","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-27 13:41:13.000000000","message":"Patch Set 3: Code-Review-1 Workflow-1\n\nubuntu...","accounts_in_message":[],"_revision_number":3},{"id":"ecb95f97ade70e14b511087e47db71fc9f53135f","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-27 22:00:15.000000000","message":"Patch Set 3:\n\nThe good news is Focal has Ansible 2.9 from official repos and it comes with Python 3.","accounts_in_message":[],"_revision_number":3},{"id":"b1fdd0149fb62e7124c99eb347bce21bbee40946","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-28 13:48:06.000000000","message":"Patch Set 3: Review-Priority-1\n\nwill go back to it in Vicky","accounts_in_message":[],"_revision_number":3},{"id":"22291a883dfb3a62c66aadb9e5ec57d68bec217b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-17 13:34:05.000000000","message":"Abandoned\n\nk-a patch abandoned so this one is not needed too","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7c956bf7351bf3a605f6e8a6efd4b384a021a0a0","revisions":{"8f8336388713e80b86fe512f1082604964761e2d":{"kind":"REWORK","_number":1,"created":"2020-04-20 12:17:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/721251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/721251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/721251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/721251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/721251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/721251/1"}}},"commit":{"parents":[{"commit":"26ba50c16f4560bcc3a405650d8f608c31b9df8d","subject":"Merge \"kibana/debian: use upstream package on all architectures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26ba50c16f4560bcc3a405650d8f608c31b9df8d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 12:12:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 12:14:39.000000000","tz":120},"subject":"Debian/Ubuntu: use Ansible 2.9+ binary packages","message":"Debian/Ubuntu: use Ansible 2.9+ binary packages\n\nDebian has 2.9 in buster-backports, Ubuntu has PPA with it.\n\nChange-Id: I5610f56cd0c3419ee5b1bff02655ff864614de7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f8336388713e80b86fe512f1082604964761e2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f8336388713e80b86fe512f1082604964761e2d"}]},"branch":"refs/heads/master"},"4ff44fb26fa86932c2227a0764b9ffea2a9d1317":{"kind":"REWORK","_number":2,"created":"2020-04-22 07:08:06.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/721251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/721251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/721251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/721251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/721251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/721251/2"}}},"commit":{"parents":[{"commit":"d318e5c1a193914ee90eab1c9a762d61d1b07acb","subject":"Merge \"Zun: add CNI loopback binary to the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d318e5c1a193914ee90eab1c9a762d61d1b07acb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 12:12:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-22 07:07:13.000000000","tz":120},"subject":"use Ansible 2.9+ binary packages","message":"use Ansible 2.9+ binary packages\n\nDebian has 2.9 in buster-backports, Ubuntu has PPA with it. CentOS 7/8\nhas repo with it.\n\nChange-Id: I5610f56cd0c3419ee5b1bff02655ff864614de7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ff44fb26fa86932c2227a0764b9ffea2a9d1317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ff44fb26fa86932c2227a0764b9ffea2a9d1317"}]},"branch":"refs/heads/master"},"7c956bf7351bf3a605f6e8a6efd4b384a021a0a0":{"kind":"REWORK","_number":3,"created":"2020-04-22 09:55:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/51/721251/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/721251/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/721251/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/721251/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/721251/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/721251/3"}}},"commit":{"parents":[{"commit":"d318e5c1a193914ee90eab1c9a762d61d1b07acb","subject":"Merge \"Zun: add CNI loopback binary to the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d318e5c1a193914ee90eab1c9a762d61d1b07acb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 12:12:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-22 09:55:08.000000000","tz":0},"subject":"use Ansible 2.9+ binary packages","message":"use Ansible 2.9+ binary packages\n\nDebian has 2.9 in buster-backports, Ubuntu has PPA with it. CentOS 7/8\nhas repo with it.\n\nChange-Id: I5610f56cd0c3419ee5b1bff02655ff864614de7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c956bf7351bf3a605f6e8a6efd4b384a021a0a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c956bf7351bf3a605f6e8a6efd4b384a021a0a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
