)]}'
{"id":"openstack%2Fkolla~721252","triplet_id":"openstack%2Fkolla~master~I564fa390004a749f5ce2ba524dda271cbf1145b7","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I564fa390004a749f5ce2ba524dda271cbf1145b7","subject":"kolla-toolbox: use only binary packages","status":"ABANDONED","created":"2020-04-20 12:17:11.000000000","updated":"2020-05-17 13:33:38.000000000","total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d299ea563a38f18e66a270dd63826ff2ead7794","_number":721252,"virtual_id_number":721252,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2020-04-23 10:32:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-22 08:24:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-04-22 07:53:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":-1,"date":"2020-04-22 08:55:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"rejected":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2020-04-28 13:49:57.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-21 05:13:07.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-22 07:53:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-22 08:24:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-22 08:55:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-23 10:32:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2aa68151705d35d039af3f1c5867493b310f8bc7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-20 12:17:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f044481babe77c551b6af369b1df593f4e567365","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-20 13:06:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/42b594b12a2548ce97711cda32ef6443 : SUCCESS in 28m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc4dacc8be9f28d3a7309ff2daf864f7c9e3686f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 22:13:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/23438e2f14464101b0ca08fd633e8f39 : SUCCESS in 50m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6f6b66104cc45e2972cbece61e493aa : SUCCESS in 1h 43m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c80843ba0577408e9a7d22d2f7fcd1de : SUCCESS in 53m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7e9f68bc5ba0350cab76b8069ae99061f73b926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 23:42:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4a47a958af5c445eab24ffa65593e91b : SUCCESS in 42m 04s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/194ab675cf9d461882775825ed8527b8 : SUCCESS in 29m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f2dea10d0df5444e8bc41b195eabf883 : SUCCESS in 1h 22m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d0a7c0fded1f42acb08ca31f0c8aef13 : SUCCESS in 1h 25m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3eb6130a1f07418b8b5240da1a9fe20b : SUCCESS in 1h 23m 41s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d5eab7ce99ea409cae94626e64680905 : SUCCESS in 43m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fca7d46d2f1540ee8ef29161e7db93fa : SUCCESS in 20m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c4b37976e1864b47915d91e7bb08ce93 : SUCCESS in 21m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/102458bb12164c38b07edd2234417533 : SUCCESS in 1h 01m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14a15f8a2e0a4d189bcd76a64a37f298 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbeed7447aaf42268075dafe28b9b4ce : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dac3f9af2ff94b07bb563bc9cc338ec9 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5ca1c1489194050a1c065ad2831eb88 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ff21de7699244f388d5af6d409e808b : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96e027de2584c8db4e7cbc046467eb9 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/31eea93762b44934801ec10a5867679b : SUCCESS in 28m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f48a883da684a37a7650ea8b7438f85 : SUCCESS in 43m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76ae63dec6e44302b708c5969ebcd833 : FAILURE in 24m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d5edc64dd424ed1a1fb3b13c3298cfc : FAILURE in 34m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/acd7b90aa5ba441b98b08d00397d1641 : FAILURE in 1h 31m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3119194266b03a16cb65767e1f82dcbccb8e93b1","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-21 05:13:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a5a8d68155f1e7048d28faee00abbc2fc52755a1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-21 07:27:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e510a6a7c7647a2565d8629e7408c6cc58d49959","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-21 08:02:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/169e222bfccd43768ac9627e6c784f0c : SUCCESS in 31m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d232ccdf836f9c782579a422f61ba634e40a8e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 08:56:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ebfa5a16cd2c444b8dc8a7d7e6372f34 : SUCCESS in 26m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b75cc00fa4d04d298ccdb44e6b1c1f4b : SUCCESS in 57m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d8ef8312de6e4b0b8f9ef55ab68d17c8 : SUCCESS in 1h 21m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0e4f62b652f746e2b40565ed21c2afb6 : SUCCESS in 1h 08m 06s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b49ea2d68556442ea8bfee36499cbc3c : SUCCESS in 1h 23m 50s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3a8d4bc939054dc186b3cac2ead4cf94 : SUCCESS in 45m 43s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2c8c5868aaf6442f9f58586b0ea8b9a8 : SUCCESS in 26m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/11f99f98e401432aa7ede68b592e5bb5 : SUCCESS in 21m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1dbd7d724918457a85526627e878202b : SUCCESS in 1h 02m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3682c534d3546c59e1c43635fc58878 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb0e60a9d4846b788dc5fd928059521 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97c00b6da45c44a8a5aa659383096071 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c651294152d4358bbb19e00ad0246fd : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e991d3239fe34853942febdecaf97118 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f0e0e1ee41d4c66b20410562c0f2008 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d38dc8a21989471c822cd3e7afa860be : SUCCESS in 20m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/54efaaf6149349a6a0613a99652c9535 : SUCCESS in 29m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/41f1e28258b64b1181bceafbd14879ea : FAILURE in 24m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d747bc8a5f66466ebfcc7dbaf2505243 : FAILURE in 26m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cee9c32e27324733994409e063bcff90 : FAILURE in 56m 29s","accounts_in_message":[],"_revision_number":2},{"id":"e2978127b289d8f5e9aab1ff4a89a422d616d1a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 09:07:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c96f032f1015450ea3c68c8380c6f9a7 : FAILURE in 51m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9469232b987e4f068a9cdc65ce162d46 : SUCCESS in 52m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d89b462abd4473bb7c1d722e60c9ddd : SUCCESS in 1h 34m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/14e7f1d66aeb4acb9a5214e837715781 : SUCCESS in 51m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65ecb174247716c1dec387b8ab409ba46d56a32d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-21 13:01:56.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nWill need a reno about removed jinja customisation points.","accounts_in_message":[],"_revision_number":2},{"id":"fb523f6b58ba43f9bb10e075ed9ad07e58d5ed25","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 07:08:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a34fcf5923b0366d472d1e6f7633dfd6819ce9c7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 07:12:19.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"68d59006c429364443451d5c66b2119a65f40368","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 07:17:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"85cfbb658c75e10b1cb49e0f26ed10517823205e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-22 07:53:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/df1c1a9015034fc8900180312bbb2115 : SUCCESS in 34m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"88c3992cffb0a62a42d7c1a3e24e0532a945ec93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 08:24:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4ada320adf854ef7b832a511d020c2ec : FAILURE in 23m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/791d763ea1a14c7e9cfede6519a916b2 : FAILURE in 27m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8646e59aa9ca492ca9f359376a36b05f : FAILURE in 24m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4d6a60b90a724cfabf7aa90767db3dc2 : FAILURE in 20m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a01b781fcc4c40eba2cd8183dc226b27 : FAILURE in 24m 44s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f96aeb5de9b64c81a80ac866a7595bdb : SUCCESS in 49m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/31effee938aa4716ac18b57ba6c7b18d : SUCCESS in 21m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cd36cb59b83546268d772b5fa8860774 : SUCCESS in 19m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7049590d833545c9b05709e9a86022e6 : SUCCESS in 1h 01m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b32814ed59a147fba9e81146efc219a2 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b52f5f3959c4543b851a2c29cc364ad : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb5318fd25614eb0b033290ba2849cd6 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b73896b16e649388354c5297db8cbff : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa98e5852eab4d24856bde7515ec0270 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0884e86e27e4461ea0b08737cc5dd3f9 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7956a2957a864c189e52133e84006376 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eba2fd07a7b14a7393f386063e2481a3 : SUCCESS in 23m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/553963c17a1a4b4ab4bce1d59973944c : SUCCESS in 23m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d228c84cebf74e05afdc817c07e00385 : FAILURE in 27m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d7feedac3f65464695e936a606ab56fc : FAILURE in 25m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0b8f7417fbb84a1e9e477c7bbf196ba6 : FAILURE in 57m 27s","accounts_in_message":[],"_revision_number":4},{"id":"2ff1dc91d1dd0a7af0ede71447e5110c6f7be3c6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-22 08:55:40.000000000","message":"Patch Set 4: Code-Review-1\n\nI thought we were leaving the source flavour in place. Now we solely rely on distros to provide OpenStack tooling, even for our source builds, which is not what I imagine as a stable solution. The point was not to do source for binary, but still do source type...","accounts_in_message":[],"_revision_number":4},{"id":"a6e4c32515f9b56bf5fe672bde05684d6f156493","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 09:09:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6f1c29db27e4fcc8b3405c9f249ace4 : FAILURE in 13m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7a1410cf2cc42f5a15a1a446bae755b : SUCCESS in 58m 50s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8bd60415a23b464f85d2d863bc257f12 : SUCCESS in 1h 45m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c458c88531334d97805fd4f905603b1f : SUCCESS in 55m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1cdd941a654b23533c5dbd281a3f1e0c2a6de177","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 10:20:31.000000000","message":"Patch Set 4:\n\nc8 has wrong baseurl in repo file. Problem got fixed, new centos-release-ansible-29 in progress","accounts_in_message":[],"_revision_number":4},{"id":"006b1eb5277119bea3b349956226b419eaa36879","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 10:32:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"49b162b25e821faa81f17a72e820187288e9deef","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-28 13:49:57.000000000","message":"Patch Set 4: Review-Priority-1\n\nwill go back to it in Vicky","accounts_in_message":[],"_revision_number":4},{"id":"1d299ea563a38f18e66a270dd63826ff2ead7794","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-17 13:33:38.000000000","message":"Abandoned\n\nnot every distro will be up-to-date with python3-openstackclient","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"062202ea6c66d86389ce28b33243176a9fa6744a","revisions":{"18bf69399a056943574d789712c2e352469b68ee":{"kind":"REWORK","_number":1,"created":"2020-04-20 12:17:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/52/721252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/721252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/721252/1"}}},"commit":{"parents":[{"commit":"8f8336388713e80b86fe512f1082604964761e2d","subject":"Debian/Ubuntu: use Ansible 2.9+ binary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f8336388713e80b86fe512f1082604964761e2d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 12:14:50.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 12:14:50.000000000","tz":120},"subject":"kolla-toolbox: use only binary packages","message":"kolla-toolbox: use only binary packages\n\nWe use virtualenv for all Python stuff. But all those packages can be\ntaken from distribution so we do not need to install compilers and\nheaders.\n\nResulting image is over 300 megabytes smaller.\n\nChange-Id: I564fa390004a749f5ce2ba524dda271cbf1145b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18bf69399a056943574d789712c2e352469b68ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18bf69399a056943574d789712c2e352469b68ee"}]},"branch":"refs/heads/master"},"37306baf9c86b4d9660ec10b7de1d1ef6332e483":{"kind":"REWORK","_number":2,"created":"2020-04-21 07:27:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/52/721252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/721252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/721252/2"}}},"commit":{"parents":[{"commit":"8f8336388713e80b86fe512f1082604964761e2d","subject":"Debian/Ubuntu: use Ansible 2.9+ binary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f8336388713e80b86fe512f1082604964761e2d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 12:14:50.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-21 07:27:35.000000000","tz":0},"subject":"kolla-toolbox: use only binary packages","message":"kolla-toolbox: use only binary packages\n\nWe use virtualenv for all Python stuff. But all those packages can be\ntaken from distribution so we do not need to install compilers and\nheaders.\n\nResulting image is over 300 megabytes smaller.\n\nChange-Id: I564fa390004a749f5ce2ba524dda271cbf1145b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37306baf9c86b4d9660ec10b7de1d1ef6332e483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37306baf9c86b4d9660ec10b7de1d1ef6332e483"}]},"branch":"refs/heads/master"},"0441e8da6fac467323b969089a27732e2abc2b45":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-22 07:08:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/52/721252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/721252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/721252/3"}}},"commit":{"parents":[{"commit":"4ff44fb26fa86932c2227a0764b9ffea2a9d1317","subject":"use Ansible 2.9+ binary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ff44fb26fa86932c2227a0764b9ffea2a9d1317"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 12:14:50.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-22 07:08:31.000000000","tz":0},"subject":"kolla-toolbox: use only binary packages","message":"kolla-toolbox: use only binary packages\n\nWe use virtualenv for all Python stuff. But all those packages can be\ntaken from distribution so we do not need to install compilers and\nheaders.\n\nResulting image is over 300 megabytes smaller.\n\nChange-Id: I564fa390004a749f5ce2ba524dda271cbf1145b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0441e8da6fac467323b969089a27732e2abc2b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0441e8da6fac467323b969089a27732e2abc2b45"}]},"branch":"refs/heads/master"},"062202ea6c66d86389ce28b33243176a9fa6744a":{"kind":"REWORK","_number":4,"created":"2020-04-22 07:17:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/52/721252/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/721252/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/721252/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/721252/4"}}},"commit":{"parents":[{"commit":"4ff44fb26fa86932c2227a0764b9ffea2a9d1317","subject":"use Ansible 2.9+ binary packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ff44fb26fa86932c2227a0764b9ffea2a9d1317"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 12:14:50.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-22 07:16:54.000000000","tz":120},"subject":"kolla-toolbox: use only binary packages","message":"kolla-toolbox: use only binary packages\n\nWe use virtualenv for all Python stuff. But all those packages can be\ntaken from distribution so we do not need to install compilers and\nheaders.\n\nResulting image is over 300 megabytes smaller.\n\nChange-Id: I564fa390004a749f5ce2ba524dda271cbf1145b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/062202ea6c66d86389ce28b33243176a9fa6744a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/062202ea6c66d86389ce28b33243176a9fa6744a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
