)]}'
{"id":"openstack%2Fkolla~721329","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163","project":"openstack/kolla","branch":"stable/train","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163","subject":"switch to official CentOS 8 repositories","status":"MERGED","created":"2020-04-20 19:02:54.000000000","updated":"2020-05-18 10:40:57.000000000","submitted":"2020-05-18 10:38:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721329-1589798328421-ed83af0c","meta_rev_id":"25282dfa61191eb878535509881364354a8773e5","_number":721329,"virtual_id_number":721329,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 10:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-05-18 08:05:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2020-05-18 08:56:33.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-05-18 06:51:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2020-05-18 08:56:33.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2020-05-18 10:40:57.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2020-05-18 10:40:57.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 19:03:50.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-04-24 05:49:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-05-08 17:03:03.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-05-12 10:59:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-18 06:51:10.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-05-18 08:05:05.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2020-05-18 08:56:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-18 10:38:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e67ba8a9c8552babb89e121101fc06eb917c1c89","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-20 19:02:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6af4b21c951fbd05da29a38de922186721130700","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-20 20:10:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/535f4b94f1c64970abba47a22ecb8e16 : FAILURE in 26m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d0a8840cebec6059f981741e235636e67916d7e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 21:39:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/169410b22c6c40ef93263fe1197996f2 : SUCCESS in 1h 51m 10s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3796910744054a1c82f53cf71a05360c : SUCCESS in 1h 00m 01s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/de7236da863845278db34063c43a4de0 : FAILURE in 25m 26s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d6884b4987894a6e99e43ea23eb03e4d : FAILURE in 36m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/295d6bf3d1ba4d9e847caa40dc5c9786 : SUCCESS in 1h 39m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/013e222d0a59465ead1e8e0b3c4d966b : TIMED_OUT in 2h 02m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7278d03e6b224a8f9f23fbc8797ff725 : SUCCESS in 1h 29m 54s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0686e7bed0404d3b9692703a57b8a814 : FAILURE in 48m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/63c61bec10634564b881abb13493ee8a : SUCCESS in 1h 52m 40s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4547665118804742a797dacba1a542f5 : SUCCESS in 54m 12s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/0ca8795c76f44a65bb0611ad27e20ab8 : SUCCESS in 58m 28s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/ff83a36466634c9ba3f7753958c18ecc : SUCCESS in 34m 34s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/9d2e21e30b244ec1a04b0e2234fc15be : SUCCESS in 2h 24m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/ff2309f93c794e7384d91d620b91442a : SUCCESS in 2h 30m 04s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/0f3b23e7630c48c4a246d6fc59306360 : TIMED_OUT in 2h 02m 34s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/335ed18d9ab74976bc10fa5722f3e9ce : SUCCESS in 1h 47m 44s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4f257a24ba774043b8a1ec6c13e447dd : FAILURE in 1h 30m 52s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/60b2e434aba24ba89aac687ad3bd8300 : SUCCESS in 1h 26m 17s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/28f64083e9834d33b6b925b1ca04bd2a : SUCCESS in 1h 15m 53s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/33a44cf606884b99ab30424d57d77fff : SUCCESS in 24m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2c355f37e7004c36aa9d39d28e2c9ec4 : SUCCESS in 18m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/86aef03bd0d94270adc93b92a44c9064 : SUCCESS in 1h 13m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7540bdcf9dac4408870f52c0caaf21ff : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16e6587ecda94627bf4a8941dfcd9a90 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb87b61f6bda4c3a914e34fe9bb68760 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22a03eac0c7242f8a6ac6751c2c9792a : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf29a366119f42d6857134f87ed0e69a : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55796bef70a54579be32101b9774f6bf : SUCCESS in 10m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49b73a30a92548ed949feeb9795f9e57 : SUCCESS in 47m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/549abdfe5d0146698f26376db060af22 : SUCCESS in 51m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba8eb710b9aa45ce87f5ca081ae75da0 : SUCCESS in 1h 09m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/113ddec730164e229b59858ef1d7dbca : SUCCESS in 1h 10m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d2f6b5c409d41f1afafd497c3a88898 : SUCCESS in 1h 21m 11s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/123ee1f2df034016acd2b0e5560fec2f : FAILURE in 1h 18m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a21daafaa42f4f388348fdc4fb0aa953 : SUCCESS in 2h 00m 03s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/30e9fd8d0b1649a7934380f17f4625be : SUCCESS in 1h 27m 29s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d77de7e399a141d8b7776154a9787830 : SUCCESS in 55m 05s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"733ec85fde4e48371eaa4a672689022c022aa829","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-21 07:26:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e844138def3c1ae5aac96e522a5e80d75a7e507b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-21 08:01:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/56409364be564e9b824608ffdfb2b905 : FAILURE in 31m 21s","accounts_in_message":[],"_revision_number":1},{"id":"14b18ef14eb5279d402b2483b90071e15aa68a67","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-21 09:12:09.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0b13a7174c9bcf03c8dafd551fde69ea92cef419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 09:46:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/24b71004eb3c44b4836180228602fbde : SUCCESS in 41m 23s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/32ab2b183c9a4f1184fc929d05ef8ea0 : SUCCESS in 1h 17m 41s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/79ae4c5f87794893953e5b044178e84d : SUCCESS in 1h 06m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c787c0cac3944534b338569ee3c6d550 : SUCCESS in 1h 13m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/57c96f8974fb43fc932fe3f182e71c46 : SUCCESS in 1h 48m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cacc3322bfef42b4a0183e6379dfb8d9 : SUCCESS in 1h 04m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1a1c88777b394744ba40d9068dee387b : SUCCESS in 1h 48m 12s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/339767b1383b4103beeb7141c4197b2c : FAILURE in 1h 41m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b12f7b4e01744f0e8cd8dd8a2e82a64e : SUCCESS in 1h 45m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/df3d7283e80e4d0b886c6b9f5d6de589 : SUCCESS in 1h 38m 58s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/efd2af8c74da41cf8972d681c79ab11a : SUCCESS in 54m 16s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/13d4e53a3e7542b8aa932e2a107f01c5 : SUCCESS in 29m 05s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/4227bbb348bb444cbb5be704493f4472 : SUCCESS in 1h 52m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c039f3546ada4071be8b50b1407b1e1a : FAILURE in 2h 15m 17s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/8e93f8eb703c4b0c9e270a3ef0ca5caa : TIMED_OUT in 2h 02m 47s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/fc6a4288278d40df8d11bb7dbce7002a : TIMED_OUT in 2h 03m 23s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/dcec002cb2ba48f6874f2058ad3ef45b : FAILURE in 2h 01m 38s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/9c55115ed0f94f21b0e6bca6d8a9c8a7 : SUCCESS in 1h 35m 31s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/283827b742cb402f9cb20d00c4d0088d : SUCCESS in 1h 28m 00s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/46322254ce2b4b06b3ee4a7a040058f6 : SUCCESS in 40m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/daf8c05033ea492098e39a29dc75930a : SUCCESS in 48m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7c01d0c49a674d22ac8f344009219cf9 : SUCCESS in 1h 04m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/deb1f7a17e544af699bc0704cc40906e : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0aa1a03f9e94f3a95517e252cf7a2d1 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d59b49db3264412494082e85102ca85c : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1910dcd5ed04d9b8f2e5ecba5b26b92 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aae1b12295e64e399fbe6d674131a1db : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b231cf02eb1e42878a0f13d94ed4d0d6 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/29e27d39b69745b2ae48d2ae70196400 : SUCCESS in 37m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/da5e642f95ce4c619b40f22439e2508e : SUCCESS in 25m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2310d608c0a64e0c890d9564658cec5b : TIMED_OUT in 2h 01m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3fbc6871f66c49088e6cb0c1bb9bad66 : SUCCESS in 56m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/84cd73b73e974e92a7437482df2e45ed : SUCCESS in 1h 11m 01s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/80de3aeb95df42da8353e3af45018179 : SUCCESS in 1h 26m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/992151f785e84fff8a7c93f1d8e2f754 : SUCCESS in 2h 17m 49s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/4b841748b94a41fba7934281eac1e66a : SUCCESS in 1h 20m 31s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d71c7393c3864493a6df1f59dc7eaa25 : SUCCESS in 56m 48s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"34adf7ecc9f1efc10d37bc328ffe39faefb91249","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-21 10:24:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8c7012f8cd42fbb4829d4cd062519ce39257003","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-21 10:55:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/a40b14f985d9409fb69398f516a9f2b6 : FAILURE in 27m 51s","accounts_in_message":[],"_revision_number":2},{"id":"cb93c1c00af3dd34fa88a621a351d43c02b165e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 13:32:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5db12c37708e4d62939219af49ad324b : SUCCESS in 46m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/93eedbf496d14f0b973a8db3e21997da : SUCCESS in 43m 10s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1068e2960cdc4dac8c3b0ebc91137072 : FAILURE in 34m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b0209dbca5b64c4698141504e4f8d037 : FAILURE in 28m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/452556865e004cc582af89837885613b : SUCCESS in 1h 08m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a68c4bd563ad440ab980e0659e06a460 : SUCCESS in 1h 30m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1550e46f117e44ddb22e66e57f0b4aee : SUCCESS in 1h 21m 36s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/684e7925ea4040f7826cd1618c89ebe2 : FAILURE in 24m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3e5aa8901c864b578fc1e4a04a6cc25d : SUCCESS in 1h 31m 08s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6268ae89c2da43d68d143dfb8e5c2881 : SUCCESS in 51m 07s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/c5953fd0a18049f993680ac521de0839 : SUCCESS in 42m 44s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/a83c7ec1497a4fcab4bcd5fcfec87441 : FAILURE in 18m 06s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/b0289c8d19c24439a99f6e6ac9dd1070 : SUCCESS in 1h 53m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/18bc76a336aa4d50beb9671b455a3b8a : TIMED_OUT in 2h 33m 08s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/ca7ed2eaa644479182e219e10158bf4c : SUCCESS in 1h 50m 58s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/0eb5ffc37b324e308f7200bc67f14c44 : SUCCESS in 2h 01m 34s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/fa0309fca1eb44c6a625b80e036ee96d : FAILURE in 1h 37m 06s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/031bab08e5aa4bdd9dd252063851f8dd : SUCCESS in 1h 33m 15s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/ef9499d9d0c24490b9fac5fa704a8670 : SUCCESS in 1h 21m 20s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/633df467ece04bd0858ab2dcd854f850 : SUCCESS in 37m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/428016e6a0c64dbca5d59811c4f8d3af : SUCCESS in 24m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cf2d4c4133c3493d882763dd6c50795e : SUCCESS in 58m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60bdc755956143c19fc8e7eb0011eef3 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506d88d730634ff6b93da9c330450cb7 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a8adeb5d0c1481cb8554e22511c55a7 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dec4e5c4c4e4e97a8803bbc923dc4ec : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caf84b4b58fb4b65b8cdcdc2fd99c935 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151ca8394ff94657b54a0e78b6e6128c : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7451cacce3c44e2780013e985b79cddf : SUCCESS in 34m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ba3beaf10fd477c8f458f3d068bacf2 : SUCCESS in 1h 08m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/27bd39a5d8d34e4eb86f808c50d81d3b : SUCCESS in 1h 03m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c7da9bb23ab34121a80b78ba8b677ee6 : SUCCESS in 59m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/29ad3234f0cf4b19994b4d193f6dbdfd : FAILURE in 1h 16m 20s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/2c683664906745eebaa743ac48f844e4 : SUCCESS in 1h 20m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/2d46b98e81c44719a595b03ccfb2602a : TIMED_OUT in 2h 33m 24s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/058f556119dc4eb9a2f94e3f19a2fcfb : SUCCESS in 1h 18m 43s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/4f14327120804d92996371dfe76665dd : SUCCESS in 58m 08s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"568aa4ed3a210afa79bcf354b8725ac500d75e9a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-21 17:47:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5fa29fc6bbf5d4c1b290619d17362c29df93f852","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 06:55:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac64c2392ba2033aad3e7358747c0de47bad3984","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 12:10:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ca63aedef7a3ef33cc88326ecb2cc2df08e49521","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-22 13:07:42.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/c92fda8832f4462bbe921121ca5735b6 : FAILURE in 27m 48s","accounts_in_message":[],"_revision_number":3},{"id":"7765d9b417f109aa669d63ef29370c2a4f81cfa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 14:26:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/df863a84ab044fddba27bdbca2c79954 : SUCCESS in 52m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/17c8d985c3e84c4682c54d9f944c139f : SUCCESS in 1h 01m 08s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5986ad4843d842f4b07fc6618378e932 : SUCCESS in 1h 19m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ad5c3be29036463e87ba97198c26c27b : FAILURE in 1h 24m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c1c5ae9b0f204fbf95bc28ce4a50f06d : SUCCESS in 1h 36m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d87034d13b1c416a9a6df675c74a5961 : SUCCESS in 1h 24m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/56f2155097f84c108830fd063981fcb4 : SUCCESS in 1h 56m 01s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ffe3eb081d7d45e48e2760f0284ab568 : FAILURE in 49m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c7c1dc65147c455f8a58da73b79db19d : SUCCESS in 1h 47m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/69bc7ad7f3f145689f9b5216d3b797b9 : SUCCESS in 57m 58s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/a8077dcf245d48b9bd2188e9d1560140 : SUCCESS in 59m 55s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/18500d521fd34d23b6a2ea740e17ffd4 : FAILURE in 32m 32s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/ccb6e43afc8f41e8b3c0e292d094e06f : FAILURE in 1h 43m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/ca751a8a2b2d4baf8c26be08084a5713 : FAILURE in 2h 12m 13s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/4eb799ee3a6846fc9cbb9e43561cc31e : SUCCESS in 1h 59m 38s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/341797331d9e4c4abd919e83363c00cc : SUCCESS in 1h 42m 45s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4aa5845470464bbaa76fc551076d9625 : FAILURE in 1h 50m 21s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/2833bda838d34b0db57f82b40eff09ef : SUCCESS in 1h 25m 11s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/300721457cd54b039bfb2e800eae4e9e : SUCCESS in 1h 20m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5730232b9d4f4e0b977b56b0085344fa : SUCCESS in 29m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f7992f9cafe342f39285ac6abf5e5158 : SUCCESS in 19m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/25a37fe9178b436bbac4660cc6a77bfc : SUCCESS in 59m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3ae15d25c504a03a3e99056111511ef : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e5c64c97a27425ba685b12d19518d8f : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9cc636314fa34ccfb6c5a4f98873d858 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72a97319b0cc45628e42ff44120c626d : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b8d944cdccb4824a09e44564f408de5 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c96ee7ea4040cbaf82540dec3e357c : SUCCESS in 8m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bd44ae5bfd714f1eb58e814529255173 : SUCCESS in 22m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/52c08824c03c447ba12b6f60c3fefa41 : SUCCESS in 1h 14m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7265cbeecc94443296bdf9d2b89a416c : SUCCESS in 1h 28m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eed6c785bb3b4cf8a4ee9cec70bc4ecd : SUCCESS in 59m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d9a9521deae04fb482674a93310361f9 : SUCCESS in 1h 23m 51s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/acd785914ccb4f599b10cf6b65d541c7 : SUCCESS in 1h 38m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9b1ec99f2337470fad3cb9601f72e0bf : SUCCESS in 1h 45m 42s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d7a2828169434cdf87d2d6e8b48eff1b : SUCCESS in 1h 24m 25s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7614438fd589416b8b4dab6a666428a9 : SUCCESS in 57m 16s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":3},{"id":"383067ae63e63050204f1ec2575a8f0db4bceca6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-23 09:36:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b80d8cdb99659a92fed03c230d2fb1f58e7a8f04","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-23 11:29:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/7889548f05ef4c09af804039cdde3d34 : FAILURE in 26m 16s","accounts_in_message":[],"_revision_number":4},{"id":"0a7849d64dfd3d6bcd69d2cd58fc0fe9a27d95b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:59:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4e8f95d4781c42fd8129c7aef330a6f1 : SUCCESS in 45m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0dd300b5cd6f44528ee5a12604af1c7d : SUCCESS in 58m 20s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5a54dc96cdd944dca28895d0db896171 : SUCCESS in 24m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/59746f3c490e4724932362e3498969b7 : FAILURE in 36m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/161bdc7a02844b72ba6f7b0e9c2a2c91 : SUCCESS in 1h 20m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cde628d4a8404488b7f5d119ba8d468f : SUCCESS in 1h 09m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cc0ac125014143a9b5e074f1f504c7b7 : SUCCESS in 1h 19m 07s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/414ff7dd69114a079d0580d3da5ff05f : FAILURE in 56m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1b429a505163493db05fd356e8a2e9ce : SUCCESS in 2h 13m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bd19ec51eeba4c5c8aea8881926be6b2 : SUCCESS in 1h 16m 45s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/94ed8d0448ef4aec86aba5b66ece626c : SUCCESS in 47m 18s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/fa3f5c2595ca440891c725e62cd1a8e7 : FAILURE in 35m 04s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/fdea4d5588ea4bb8aaa55b87df6610ac : SUCCESS in 2h 15m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/41d9e50ecba54e168f63005f814e1d24 : SUCCESS in 2h 13m 22s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/88a697e4bd4e41758a4edd7809e7c0d9 : TIMED_OUT in 2h 03m 39s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/b49f3f2a6d1246148e50b4fb036773ac : SUCCESS in 1h 41m 40s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/17edf41c9de04280a626efcae9a735c9 : FAILURE in 1h 32m 38s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f311c2e574e94574a089df58909d8a65 : SUCCESS in 1h 32m 56s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/a63479ac9d0743dc889073fb549e8230 : SUCCESS in 1h 24m 00s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/00d35a1278e840eb93665da463ebf8a0 : SUCCESS in 43m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2042e6ca23e24f088aeceff958180818 : SUCCESS in 42m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fec553f62c324aacb5d3b3a9cedeb2dd : SUCCESS in 1h 16m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95de25ad9ee241ec923dbb9b2c08dd75 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1dc59177c0e473ead1dc7ad1d833bf6 : SUCCESS in 7m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/652ca7dcb9c54b48b3c1abd948dc2148 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04249d30f9fa4345a54a267903a6059e : SUCCESS in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6f7e53575f94a6b9a759eeb04f131ff : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d43fdc984d9490b9453edbc97a4ae04 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/777ec7d5bf9a48158e65f84c887deeaa : SUCCESS in 6m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c19bfbe992c45cbb4cf43aa6951405b : SUCCESS in 24m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32a068ff6c1e440fb567ff9b1a97f327 : SUCCESS in 28m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c337a2d061d43e5acbc2bcb5207faef : SUCCESS in 1h 02m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/16a9799af8684886b417a10952f072ab : SUCCESS in 1h 01m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec8312e57bc641ed95a1dfc15f4cc535 : SUCCESS in 1h 05m 35s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/785ba436fa1f48b1b4cc1402b62ad1c0 : SUCCESS in 1h 45m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/d3d49fa379ef4ea9aca89ee8aee929eb : SUCCESS in 1h 55m 39s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2a2b5a5dcf814a9d8db09a54d6b088f4 : SUCCESS in 1h 26m 53s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0493dd773a1b46ce8ecc4ae3a6c83cc2 : SUCCESS in 1h 06m 37s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"1e3901ec26f839e2bc497ef2e875a2867cea2523","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-23 17:06:39.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b884ebeb93051ac623181c4e5f388f6e385c24d1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-23 19:39:47.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/89934eacc29f44c89657e851b3849c26 : FAILURE in 27m 01s","accounts_in_message":[],"_revision_number":5},{"id":"530eb3356145adecc9a797ac12b33d8da3c61606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 19:50:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bbd1ebd09724420aada492b2993baf5f : SUCCESS in 40m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c05255c232e34623a10fb100ee3ad6b7 : SUCCESS in 58m 07s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c7fc8bb9376d498094a9389f3047af3e : SUCCESS in 21m 43s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/dcc5e3133bd0417abdea9a4b553fbdc1 : SUCCESS in 40m 46s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9257b4476b3743e6a322a9aef5c0d720 : SUCCESS in 1h 27m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/499bd2ba193e455eb21a61fe462c35bc : SUCCESS in 1h 04m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e346ca77311545b7ac0c9364c42708ed : SUCCESS in 1h 12m 12s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3638d112080b41939f2e79ecec7d0c26 : FAILURE in 47m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/88f0b7b164024638b53d4bb153e7f2a1 : SUCCESS in 1h 32m 42s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b2c51e63ab3549c3b9552652755758a8 : SUCCESS in 52m 45s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/b22f66a2fbbe490299fef6b30a5b721d : SUCCESS in 41m 01s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/4566f08ada1d453aab9c1939a45689a6 : SUCCESS in 30m 19s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/a55984cf6063493a90a49f83a0da2063 : SUCCESS in 2h 02m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8e2cc8484d354ccbb44ca934c9256feb : SUCCESS in 2h 20m 15s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/ffe663f38e6b4d1198c906937dd47e12 : TIMED_OUT in 2h 04m 02s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/490575dfe6ff4bc0b05ef359cba6f00d : SUCCESS in 1h 35m 23s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/33f5e8edc64b4db3aac42ed220ede513 : FAILURE in 1h 34m 27s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/6f365f0a02a84eccb8250a9d019e6f11 : SUCCESS in 1h 40m 54s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/23c61c38d8164ae3b49b712491ddc2d9 : SUCCESS in 1h 12m 21s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3ef3804493744f61a5939934debbee1d : SUCCESS in 25m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ba1edb704c9240968f0a62b3a618d893 : SUCCESS in 19m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2241ea8ebcee4678888a0a7d79da971a : SUCCESS in 1h 03m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09e46e8a69fc4980a6b5a11f2b37f2a7 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9083f98d4343e68f1266ffbbd317da : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94be880cd02f4e79be495ac9d04cb072 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeecd489a64e42f1b35650ca576916e9 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e55ae7c3ce624a5481446a6659e890aa : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d28ec00fedc4a9590456d0af548905b : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf21f67e213b430e88b6657849542fc0 : SUCCESS in 5m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/696ca7a96d874802b2c00f34d564f462 : SUCCESS in 24m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2203e903b1ae4c16a5757d8a76af266a : SUCCESS in 27m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fcdc612f60524edf9872e6365b7ca626 : SUCCESS in 53m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/12b960f0234f47669e3b42e4bac476e1 : SUCCESS in 58m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d507d132d6c04e79a57aa44ddc003ac1 : SUCCESS in 1h 04m 25s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/c26a3dbc21a74bf99ada52828c4072f3 : SUCCESS in 1h 24m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/7383c1dd343c465189d50a930ed41534 : SUCCESS in 1h 45m 03s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/071c39a4d49a44ccbedaf1d99a4694c5 : SUCCESS in 1h 19m 40s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/e7ec1f8f3bf041baac4806ed1888729d : SUCCESS in 56m 10s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":5},{"id":"07e0179af7ad76216df13de6641ac6bf2580a0bf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-23 21:35:11.000000000","message":"Patch Set 5:\n\nCan someone take a look at kolla-ansible-centos8-binary failure?","accounts_in_message":[],"_revision_number":5},{"id":"a8068b505ecc694ce9308c25d1dc47aa3fdc170a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-04-24 05:49:28.000000000","message":"Patch Set 5:\n\n\u003e Can someone take a look at kolla-ansible-centos8-binary failure?\n\nhttps://bugs.launchpad.net/nova/+bug/1856312","accounts_in_message":[],"_revision_number":5},{"id":"7ea0a58258a23f15319adcccd22629215cb5dcff","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-26 16:37:27.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"a39e7a2632fdc7e5ac7bc0a54786472856f20b71","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-26 17:05:12.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/2b19851546b84225a5324460b74827c5 : FAILURE in 26m 46s","accounts_in_message":[],"_revision_number":6},{"id":"0b489854403bd1ba89bdb0d5c5f56fd9db025611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 18:50:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6f621d5986c34aedb233515ed44c7277 : SUCCESS in 1h 41m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7852c074b8f64db3a4507b8c51446870 : SUCCESS in 1h 42m 54s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f924532f281444b4959324530be62cb0 : SUCCESS in 44m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3539d7f12f754e45bc276dcbf4b06f1e : SUCCESS in 49m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/290541365ead41199f8944079b847423 : SUCCESS in 1h 32m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f38f5017b78a42af9d9b833fdc6a728b : SUCCESS in 1h 25m 19s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5d8be70da86b4e1c99705eabbefe82c9 : SUCCESS in 1h 24m 14s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bcefc248020346d3bb313057102e1d01 : FAILURE in 51m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3f30d98b951b4dfabf39ab5a24017ca1 : SUCCESS in 1h 35m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/01cbf5ab6763406db9ef4e474207c9ec : SUCCESS in 57m 26s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/0226927be90542188ddee56b75c8a85d : SUCCESS in 39m 33s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/01dbc5d05fab43a199203930464bf581 : SUCCESS in 27m 39s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/00d364d9736647ce93a94524f73233e7 : FAILURE in 1h 19m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e18f8dfd9fe0403a921c4cd68a980552 : SUCCESS in 2h 11m 14s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/66852022830f44029abe124e03dabbd5 : SUCCESS in 1h 51m 55s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/596d71cbde7f406d903a405feef33fba : SUCCESS in 1h 54m 30s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/9eb89cbd2c074ecc86a5ca169a467407 : FAILURE in 1h 16m 28s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/b2ebccb422f14768be64e1a10b9eca92 : SUCCESS in 1h 33m 48s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/68bda693b8c04a03a566db7433787b00 : SUCCESS in 1h 29m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/09c835c4714f493f9448ee48e1e88989 : SUCCESS in 21m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cc950247cbf64301a2c65f197cc07176 : SUCCESS in 40m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e53169dfd8ec40148162650185706987 : SUCCESS in 1h 10m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/877984661e584c529bacad1dea01e716 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf36cd7e70364822acbac676f89a7ff5 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ec19a57233142eabb58d8216188327b : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4cf1cb11ecf49ca9ac728b0a70d84ed : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44594e38cd7244f4a186117853336d8f : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd38e334a7c44549c3d2c47024a9381 : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6151addcf624ce89101e5d772764422 : SUCCESS in 4m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/071de76ace9343c99669077b55792784 : SUCCESS in 45m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/263f45ce271343a09dfd78a615579530 : SUCCESS in 28m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c94516330af480391376862461285f2 : SUCCESS in 58m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d11b9edce32404eb35cb882d776b380 : SUCCESS in 51m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dac270cfafd84191a727a4690714972c : SUCCESS in 1h 08m 56s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/f2e7fefd03014a00a1a0b46da18000ed : SUCCESS in 1h 45m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/2ac934f758cc4acfa0596e7226fdf9da : FAILURE in 1h 57m 04s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/9c8d27f12dbb4b92b145d987a4477dc7 : SUCCESS in 1h 34m 16s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ad78bd13041d4f43b00975da63588ada : SUCCESS in 1h 06m 08s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":6},{"id":"28babaa2f2bcb2baf8881e55d9c1a8a7ec75be54","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 13:56:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4920a92bbffef436fff8867cd500e0ddd0c99a73","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-27 15:18:40.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bf53cfe05193b204944258391d5755203db79e25","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-27 15:21:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"669dc6f14764549d2733148d40789c787c56fd3b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-27 17:42:14.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/f9a044f6a8e2412d942983ab6fe4f22d : FAILURE in 26m 00s","accounts_in_message":[],"_revision_number":7},{"id":"f258b609e60823fd32d7487993e35f6cbe14f087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:48:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/06d7c9963dd3476392f3b201cf197686 : SUCCESS in 42m 46s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d433246a78d14e4690b15238cd423edd : FAILURE in 51m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3e13750620ea44efa23c8dc0a3a946fe : FAILURE in 23m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/10d8dcc60a8e4c89bbf9e6af5218abcb : SUCCESS in 42m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d04793f02286442680b3397e3ced986c : SUCCESS in 1h 19m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/28983b35d8734d9bb05025184dffe6d9 : SUCCESS in 1h 12m 11s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/afe389468b834d04b1197272f9b0ed31 : SUCCESS in 1h 19m 10s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6e3368fe99534edbaf9721a00afc49fd : FAILURE in 24m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3a731312033b4496a630e398e7219a47 : SUCCESS in 1h 34m 09s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/dc7d3f59b4e74aa4b9ffcf77db637563 : SUCCESS in 59m 06s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/7a19bb53bf3d404d898641a25e2434c5 : SUCCESS in 49m 38s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/14c372de0f7f47f5aef064310ea1c82b : SUCCESS in 33m 01s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5c3ec7e3d4ac455b87f7e3a983309d85 : SUCCESS in 2h 13m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/865dd426fb824ea596b2d5072a73851e : FAILURE in 1h 44m 14s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/d84c2aa5c3f04eeb803a525db4a6fd6e : TIMED_OUT in 2h 03m 24s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/c34fb80de232434f80f0ebcbed37e15f : TIMED_OUT in 2h 04m 46s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4e1914784038462b92c3f2383f9776e2 : FAILURE in 1h 48m 19s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/e4e8b125c68a42a2b4bf029d2801089b : SUCCESS in 1h 31m 38s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/e6b632e7d4594f53884e2678ee893d21 : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1be0bc51b83c446087c9163c3774cfcf : SUCCESS in 26m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e08954985009474c9ec87564ed622b8a : SUCCESS in 19m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/56e2bcf3bb4d4092b1c89cfe8ece071e : SUCCESS in 56m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c7a06108df848ae9542d23c4e35d765 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/136871bfda124faba4ab2f34fca0828b : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce69189a5cbf4a8b906cfe81fc20569f : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42cdc35f655f448db0d47ea0a3a661b5 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fd1d23e8af24267a3ed9d4661ffafaf : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0509fc6950614d2ba95b3c7fa7f270e4 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8af2e977e9a94318b4b805b92ce4ce87 : SUCCESS in 8m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71b42ad122ad4b69bb9b7ec772b0f335 : SUCCESS in 43m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eacd9ca7f89644189a16bba9e869fbfb : SUCCESS in 34m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0026252cfeb4ea69c36d94f72952585 : SUCCESS in 58m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6b5f1dfb9d74df483a4db0342d49b25 : SUCCESS in 55m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ad4c40445b70488ba85d3a679eafd5d4 : SUCCESS in 1h 33m 22s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/9f025bee8edd4fbfae1526d8060fa737 : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/659fc11f13694bcd85b451864bd707a8 : SUCCESS in 2h 05m 43s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/30161fb2e1744f0c8c4311e2a90c637d : SUCCESS in 1h 24m 35s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/cea7f101913f441a85430f127d08026b : SUCCESS in 1h 22m 19s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":7},{"id":"9a143c6340f5b54883e1414d12cda14ae9647b44","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 18:03:37.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"739891c9bf88708df04e9194af099844f92616c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 18:05:43.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5aae5f19e413600ab6a4115b38063ffa4589601b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-27 20:16:45.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/5669932ebd0c4a649b18dfa037cb7416 : FAILURE in 28m 32s","accounts_in_message":[],"_revision_number":8},{"id":"ed9dd3e49c5d5d06b49ab4c3c823b1d9abcc8064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 20:21:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a68b1b8a091344fdbb11caaa3bd4941f : SUCCESS in 36m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ae575046a7ce433d948aff58832aaaf4 : FAILURE in 47m 51s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8c9914ea8b70437090d532297f06c1d7 : SUCCESS in 55m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f82d1434cc814670a71dc803b23bf7a4 : SUCCESS in 36m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/52a51fedd2ce4fa59f3aec73a5059da6 : SUCCESS in 1h 27m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/738b0117069d4de79ed98f4d89a9c4dd : SUCCESS in 1h 12m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/654a7255286540a188b5441386092f6d : SUCCESS in 1h 24m 31s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/daf607316f754fe88c644f8b50e0b946 : FAILURE in 1h 26m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f52e0abfdf7d4125afb6ceb1993b21d1 : SUCCESS in 1h 36m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d2e62e7531a1406f90074840f60c07b9 : SUCCESS in 52m 21s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/72198e87c21349518f249f8a8c3db206 : SUCCESS in 45m 18s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/72189c6b306543f5b95c8718dd5e2dcf : SUCCESS in 31m 46s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/9beeaaebe63c44cfb5a04d541de7595f : SUCCESS in 2h 10m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/42d139aec8de445e9ea9beeda7faedb5 : FAILURE in 2h 00m 23s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/a0d82f10691c4f418d5ffa3814e358f9 : TIMED_OUT in 2h 05m 06s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/dd4bf9ff7d70402d90762bbd390b33e1 : SUCCESS in 1h 42m 33s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a1673f8ef7dd48ec831dfd05eeba4a0a : FAILURE in 1h 26m 44s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/93b203da11bb4de5a6aadab3ba18e77d : SUCCESS in 1h 37m 22s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/a179d41fde5c441696bf0b35ed96644b : SUCCESS in 1h 15m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cadd5330be804b96832c35d4e041db76 : SUCCESS in 1h 34m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2e3841e012ef4f869d52405d846ebf4b : SUCCESS in 1h 33m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d26c533296ba45658e8ce7f2ca93d0ab : SUCCESS in 1h 03m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cffd3c3c14aa49acb5e5d03678d432b4 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63b194666093407d9aaf3c33a6f9db48 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e6a49040c64494c99a3532c14132b3d : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d1d5d24e99d4f0fa1f0927dfdcfcdbe : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02aa275e61f04035929810dd9c56b469 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01a66128074419bb57a57569e265212 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d978996d5ed4a29ab8330a97033648f : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6bc48b5a1a574dff8821bfe3578ffc0d : SUCCESS in 30m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a6f0831f83343c1b6ac812225c5f4df : SUCCESS in 27m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ebe83ccb2824a61b86ce9f257f8f51b : SUCCESS in 1h 29m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d84ff821032542f8a96339df432b8903 : SUCCESS in 55m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/21eecc5e95eb4288b5f59f0727115998 : SUCCESS in 1h 08m 16s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/24a62b673d854cbe9f5682d6d79af493 : SUCCESS in 1h 28m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/384a058109e4490a8664812ed0c48a52 : SUCCESS in 1h 53m 09s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f82b3a2d813b4db8a73d0e41512e5fbb : SUCCESS in 1h 28m 37s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/38a600606918478d8959eb1e312352ed : SUCCESS in 1h 03m 07s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":8},{"id":"df168af76e2bf698bda5318e0ba7b3216ae5ea6b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-30 13:23:21.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"e28c22bdaa63fc0f6b1870396e69bcb95087b0f5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-30 13:52:16.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/94d613840a2b49b0b98528b7a859a6f0 : FAILURE in 28m 02s","accounts_in_message":[],"_revision_number":9},{"id":"f91a4cec667c7aec11618bb5fcadc05be5cfa70c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-30 15:27:48.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"53071dac6ff19a9655f5a5fbe0d6df1758ec3553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:31:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6c8c05e5b194415c93a9b5b886fc0524 : SUCCESS in 38m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/78f07f9ee8564c88a6e750be6bd48bb5 : SUCCESS in 50m 52s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a71253b62c8b4e73a68598d752a957ae : SUCCESS in 1h 17m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0af2c8706a944e9f8529ee8e5a7e170c : SUCCESS in 1h 19m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/248c608bb98146a3919acbecafb18793 : SUCCESS in 1h 16m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3df99522301740e7930c384c64e2d8c5 : SUCCESS in 1h 46m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5771b746c69940a28ad786f3e7174fc3 : SUCCESS in 1h 24m 41s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/51bc1af74412479ea7db20707d0e1692 : SUCCESS in 1h 50m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bd9a84090a9c4edd9327ec921106b921 : SUCCESS in 1h 42m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4d70a58726a44df180ded9a3e32c86ca : SUCCESS in 56m 08s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/b425d80fd8124a29ab4c51c12e2e5069 : SUCCESS in 59m 10s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/382a8d424e644e07b3b397945c10415a : SUCCESS in 31m 07s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/7e3363edae1d4c69bbbadf960a0cd0ed : FAILURE in 1h 34m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9380a598b1c7450d9ef4961b543b0af9 : FAILURE in 27m 46s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/fb0726a1376e4f5db316832c6ec84d83 : TIMED_OUT in 2h 05m 22s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a80bea135ece4361abc1606cc6814fbf : SUCCESS in 1h 46m 52s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/cad0c4fd06304677b976e7fd3a6a563d : FAILURE in 1h 31m 18s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f727dfd852964d14b4cd031e01c79864 : SUCCESS in 1h 26m 17s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/b371894a7e474eaab13c72953a913ebc : SUCCESS in 1h 16m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4ad6d50e681141bf878abfe6bd9fc238 : SUCCESS in 29m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f3b6f4397c2941599367d3309dd1a7d7 : SUCCESS in 18m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/125a92d61fca4aa481029ac2610dacb0 : SUCCESS in 58m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1783a4c0af23475599fd87cab9e4cd53 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3006f72020df4597b7b845db74435425 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e723998189454b2ba87234c8d479bdf8 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f72ad1ba961a44f5b24c5a2fac9976ff : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c5c1ebe6fb7490e9ec74603cc0efda3 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4cb219197c04e69af7e4684a4037cab : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c42697134e06482a896036d39b112ea6 : SUCCESS in 4m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83ed392b37cf4f20951d6445358a2d46 : SUCCESS in 23m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7679365fbdd4c68949fd67049a93804 : SUCCESS in 25m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1257f4a77f3f4031b3e15f5434ad2fb1 : SUCCESS in 55m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b1cd21b9cfc486eb1ce23ed3f23f054 : SUCCESS in 58m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee46d3e9fab547caaf87bcd100f3b2e6 : SUCCESS in 1h 11m 37s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/b7e0d2941a464e2f8699a29ce7124335 : SUCCESS in 1h 30m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/4794db7d6d3542beaba52d104f1fbd79 : SUCCESS in 1h 47m 33s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/52c47e64ed3d470a8c731fa933ec047f : SUCCESS in 1h 27m 08s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/be3b3bdb6d8545499b78dc033e9ab3c2 : SUCCESS in 1h 10m 47s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":9},{"id":"422d4e233ddb45221e5d4fdaf3fb24bd91bf585f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-30 15:57:51.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"05bec813029c578e3b6eb65beefe306b7b9ea6ff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-30 16:27:43.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/be6d21010706435c955e00025a3d1318 : FAILURE in 27m 57s","accounts_in_message":[],"_revision_number":10},{"id":"a2dce318ffeb7be31b07321b527f8a07b9a92bad","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 16:29:00.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f8c55256df15d70dc0403ee34026ed578c943f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 18:16:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/13bd11188dc04256b80635a275f3491c : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f017d3ce6d21436b85b6bcc6342d5656 : SUCCESS in 52m 10s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1372ef10fc2a44feb610e8eb2af1884e : SUCCESS in 25m 04s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d05d4c6c90504f9caf5d77e73977b6b4 : SUCCESS in 36m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3364a47af7b148b1b55bc98ec87cc886 : SUCCESS in 1h 43m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0c5152ed809b46d89d2b44efe90a19f5 : SUCCESS in 1h 08m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a96e54d18b124c52b55f8e21a7c7e936 : SUCCESS in 1h 15m 05s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/38a6622c7f474b4bb2a22433f7b78819 : SUCCESS in 1h 05m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8607c58bf32748cd93178551c55d8412 : SUCCESS in 1h 45m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/52b5ed107c1944c1908cbe2ac4051ac9 : SUCCESS in 1h 02m 36s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/0ee21753350e4857aaac892317cc963c : SUCCESS in 52m 35s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/cac49918d7314ac1963d94ad6eda0b53 : SUCCESS in 32m 34s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/99989255f11f4fc1aa55ebcfee19f91f : SUCCESS in 2h 15m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9828a3d5b95740ba8f715c55e99e2844 : FAILURE in 1h 34m 47s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/7ccbe0717c0147d5833e6a1e92b88909 : SUCCESS in 1h 52m 14s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a406ad689f2148c8aec6d921417cecae : SUCCESS in 1h 46m 39s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1238fd9c04f747639292ef88706928c6 : FAILURE in 1h 34m 01s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/83df030d26e241bd8db879a56d345b52 : SUCCESS in 1h 39m 41s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/0044d2da2fa94a16a969c832628d66dc : SUCCESS in 1h 17m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a9d79861ee194245a1b2c6bb2972a58c : SUCCESS in 29m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/542c5054805c478cb9724af293febfc1 : SUCCESS in 26m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8353dcc5a5fe4cdcbeb50e93284bcb2a : SUCCESS in 1h 03m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e16a9e03916048709e926d0f2f9cdc17 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4a72156b52479da2813c79930e01cc : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75e57baa25c4452188ae5d3fe9470c63 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f71bd7943184a24af999127f5686494 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8438714b5b1e41539e426f0212a7a0fe : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de1da9d4d6c5449a9bf2b60bc8423b09 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e91368db1aa4c53a8d89661eccdfd59 : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/133ec69cac2d4948bddbdfaa01fd51f5 : SUCCESS in 43m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bbc4bc43b9dc483f80898039cb14b68f : SUCCESS in 37m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ef1009d07434651bde261921c489af6 : SUCCESS in 59m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3529bd7139f0404aa634de60b2e7438e : SUCCESS in 58m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/064d89fc3aaf4052b699acb90041a283 : SUCCESS in 1h 24m 36s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e04f2d447e6c498391d55c6af229258b : SUCCESS in 1h 28m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/40418365d946450690ff128f3ecc398f : SUCCESS in 1h 58m 54s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/13f8b9e0f1244d75ab55f6ce908a37c4 : SUCCESS in 1h 31m 06s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/101aa5d0745145aa87d59ae76a09492f : SUCCESS in 1h 04m 33s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":10},{"id":"03858e21ba1a8ba92685f0c830605a447518acb8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-30 18:34:21.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)\n\nunsafe to merge w/o gpgcheck - does it not work?","accounts_in_message":[],"_revision_number":10},{"id":"0d072e9388c7d20bc7725f2e635fbbb2232d06fd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-30 19:40:58.000000000","message":"Patch Set 10:\n\n(3 comments)\n\n\u003e unsafe to merge w/o gpgcheck - does it not work?\n\nThis patch waits for \u0027centos-release-openstack-train\u0027 package to be released. In meantime you can review rest of this patch.\n\nOnce package gets to mirrors I will drop repo file and alter base/Dockerfile.j2 file.","accounts_in_message":[],"_revision_number":10},{"id":"9b90532aa8bde0a547bc300b8a7c85447aaf6cd9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-30 19:45:33.000000000","message":"Patch Set 10:\n\nOK, let\u0027s wait for package release and we are good. I don\u0027t care much about pv. Thanks, hrw.","accounts_in_message":[],"_revision_number":10},{"id":"7b9862c931d752b25ef43ab929583fc480a574c3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-05 07:04:40.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"756ca9ef76966cb91d877f99ae12dd9df6c50777","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-05 07:05:20.000000000","message":"Patch Set 11: Review-Priority+1","accounts_in_message":[],"_revision_number":11},{"id":"139f355166e8bc032b3d28b73ef95a9a2ca2b228","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-05 07:30:52.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"04528c573ae71d5713617cf58b132eeba12b57e0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-05 07:36:01.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/d5c39e3c36ce4b188fba80699fea19eb : FAILURE in 28m 25s","accounts_in_message":[],"_revision_number":11},{"id":"2238253132d2e51743f43e395879a6046e948fa1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-05 09:02:39.000000000","message":"Patch Set 11:\n\nvery red on ansible - I guess we need a patch on it","accounts_in_message":[],"_revision_number":11},{"id":"9e93ebcf2fcb923e93eb733ca767eba19f40fb57","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-05 09:16:01.000000000","message":"Patch Set 11:\n\nwe aalso need recheck as some jobs were just unable to connect to docker hub","accounts_in_message":[],"_revision_number":11},{"id":"e99103ad2a85a65f98ebe798d3beafeef0359eaf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-05 09:19:07.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"d4b927d267e317e406065725cf3c0e2bb67c964a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 09:42:38.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f4f16c5c58944cb6aaa14de1e2ac003c : SUCCESS in 45m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c34ff5b24a5c4c58a32c3e4f1bb582a7 : SUCCESS in 51m 23s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a60e45ab0d4b4255aa74e8b0f2a684d8 : SUCCESS in 26m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1827658d3c124fe99a04f24aecf3bb7a : SUCCESS in 1h 13m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ca9fd27eca594a769b307f5279c19316 : SUCCESS in 1h 28m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6303ca4d707d4a769033ae6dcccd7b7d : FAILURE in 15m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f5b5f77d9b674ccf9a383801ac499748 : SUCCESS in 1h 16m 07s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fcd52fc2f5cd48309ba85a2b229ccb09 : SUCCESS in 1h 11m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2f80ee842e424cff99cb2ce778b15f39 : SUCCESS in 1h 46m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/918b8ee035d24a9ba63772013f56d2cc : SUCCESS in 1h 04m 06s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/61ab8af7d3ed4aeebc19a85f6b846f59 : FAILURE in 17m 13s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/aaa03947da984b8a85dfa65ba02d5533 : SUCCESS in 36m 01s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/b33b274064f84c2e9913a4e72f33af30 : SUCCESS in 2h 01m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/345c4a18869543a798fd62495e7582a8 : SUCCESS in 2h 30m 47s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/6eee8d3e7c51461691282147674625d1 : TIMED_OUT in 2h 03m 45s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/7f0b3e6b58ca4c0e82b3322fccf3d8b8 : SUCCESS in 1h 56m 14s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6c50b3908002491aad4fc0e7ad6b53e9 : FAILURE in 1h 28m 34s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/1235124e0e5041bebe3ed4e28df89266 : SUCCESS in 1h 53m 17s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/9fd3e9b52e954477938eccbe3ddfb469 : SUCCESS in 1h 16m 39s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/580c11be4e254dde97726c7d19a6e711 : FAILURE in 8m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/47818e4274f44f859e97f225cbc83210 : FAILURE in 11m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d350d7415e5a4e5b8fe5f84406c38660 : FAILURE in 10m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eccd5a11555f474eb212f32460dc9e72 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b6691c486c496c9b9065b234761982 : SUCCESS in 7m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85044bd817004a94bcc6cc6c17b4b4a8 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ead7cf1b01804acf91d0e62fa870849b : SUCCESS in 6m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fc374146f7249fe829f0f65bffaba7e : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af61626bb8e416399484127b42a736e : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a3c10f965124193a64d675e20140a21 : SUCCESS in 7m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c23c68b5fc3b497fbc94142c25c4cf46 : FAILURE in 7m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2f6afa8dc42545ee8a90578cb1838bec : FAILURE in 11m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f05fdb337be9467298b31c3b8eb83756 : SUCCESS in 1h 17m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87389162aa6345c99f67be91b131bacf : FAILURE in 9m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ba4666b68734256998652244eccf4fc : FAILURE in 9m 11s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/dc949b820d8d4773a03b242a6f24add2 : FAILURE in 15m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f9e4989de32f48c3b39a2a8584642b6c : FAILURE in 15m 35s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/14e1f21520e143059ab0ec9f8ff459ba : FAILURE in 12m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/65ec169e79524464b6561bf0da191ca2 : FAILURE in 14m 00s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":11},{"id":"cccec9866f721341b3701bee3af090f30c498965","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-05 09:59:13.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"e2c7d73efaaea685e145fb41098114a91908f4f0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-05 10:12:43.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"ca32905404fc30209bacbede901be5cf81fcc663","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-05 10:12:48.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"2cd84e71aa889771af1a59d7221db31232761d6e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-05 11:39:53.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/c2d40e713dbf4b49942e9fedf5956ca8 : FAILURE in 26m 26s","accounts_in_message":[],"_revision_number":12},{"id":"c1e9ce6a2713a48382f43b7af63a31894d59bc48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 12:40:17.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e12c4865adc34c6388b47858557d6b3b : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/613d8507b1e34150b486c882b8356c66 : SUCCESS in 1h 17m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6e81190329d349dd916948303ca1cbb8 : SUCCESS in 1h 29m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/de4eef9b6bac40dcb753cf8152f6bdf2 : SUCCESS in 43m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b69a47e3020e48e5a231ac3874420058 : SUCCESS in 1h 33m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fe963d6459a24904a9585ca1afe2d898 : SUCCESS in 1h 21m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/942d0c36bea149d6af813f227c757b71 : SUCCESS in 1h 30m 40s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e7f45a5d6e584444af24896d793f8281 : TIMED_OUT in 2h 02m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ee6e50c711f044a096780aab90fe2ed1 : SUCCESS in 1h 35m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/99efff56645b46ccacc15026c53f51b6 : SUCCESS in 55m 20s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/ffd1a283675a4f30a051625940b45660 : SUCCESS in 46m 56s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/3b690618e2b14e21b7f6d0558725bed0 : SUCCESS in 39m 14s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/a50d732cc25f4dd2970db76521c9297b : FAILURE in 1h 28m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/74459085dbb749388640e2852c36646e : FAILURE in 2h 23m 04s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/c6f6e370d8ce4499b6e5d4a951f9d478 : TIMED_OUT in 2h 04m 27s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/3af2b870747742f5b159be8619f21425 : SUCCESS in 1h 39m 03s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ac3dde5bd44a42399b74d176141f071e : FAILURE in 1h 36m 28s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/79d6e4e54e1343fd9ed9264d3cc38a53 : SUCCESS in 1h 35m 09s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/9142f6ba2de94bf783adf0b6f2a55b94 : SUCCESS in 1h 32m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/67aa6ef4f0664b6bbfc2b603cbeea7b5 : SUCCESS in 24m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/389c4ed702f44bdebdf0c25b0fda50ba : SUCCESS in 21m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2412005936ea443395fb23ac9b1b8de5 : SUCCESS in 57m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/134ee3ef00bd42c9a63aba8bf7e76c88 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e8733545a274ceb8a52fa36890fd2b4 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03761f2d594944ef9841aa5fd5985ded : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4040337e130149eb85517b5526b4a09b : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0483b7056ee421a9950856e07fcface : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ec0068bf344193aa1772b8dc4bad4c : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eff062c273d24eb0b4caf57f8e8d008b : SUCCESS in 5m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2d889306f1024809a0c96b383fc5a67b : SUCCESS in 24m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d02fd6a8e4ef4a1999d05bce5c5edd18 : SUCCESS in 36m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb1e785600c84da4be7a5494cbbe7aec : SUCCESS in 51m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/220399d6d4734c06b01bd9f5134f1034 : SUCCESS in 1h 02m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/06f878729b68419aa16606abc12f695a : POST_FAILURE in 18m 30s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/9ffa987f7ffc4a9f9cc59abff8541c68 : FAILURE in 57m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/934433b1c7f04a2589ac2d85b05e14dc : SUCCESS in 1h 51m 45s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/26b274074cde49e3bb6ed28e6c9220e4 : SUCCESS in 1h 34m 19s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ad9a3f4c647d42ff8ad73d5984c4c156 : SUCCESS in 56m 51s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":12},{"id":"67d7a35aacfb2093e47dff740c0f5621d6fcb415","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-07 08:51:07.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e00165c2b432453234ac5347d3ca5e02b59848f7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-08 10:21:13.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"54ae99a9d40400e645d3218289077cf97357f3d2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-08 11:05:01.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/59db086a25c845af80f32a70d4cac839 : FAILURE in 27m 01s","accounts_in_message":[],"_revision_number":12},{"id":"66eea9e9433b34f8d5dbfd81a88fea0537397967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 13:02:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d8b4bf5b5dd04f7d92136dfc7a60d7b4 : SUCCESS in 50m 23s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7f5578a6585943ee89ae5fa05088f278 : SUCCESS in 52m 05s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f865977a0d944b33b3a2011d7836df00 : SUCCESS in 25m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ce8524c39d8448ddbf49ce65d42ef1d8 : SUCCESS in 39m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/330bb92883ac473fafd41de0affb24e5 : SUCCESS in 1h 23m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/83da8bb50d844a1a83ec76288935e4bf : SUCCESS in 1h 03m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7762a7971d414a9ab6d5025766ab9452 : SUCCESS in 1h 24m 24s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/aed3648c475e4fb1898f59e94de8053b : SUCCESS in 1h 07m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/480c29ccd92d4971b2875077c2d1253d : SUCCESS in 2h 14m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e87c1dcf12784e858d0e412fbf299c3a : SUCCESS in 55m 34s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/c167771ee4834538a0d8227fcb55e4f0 : SUCCESS in 46m 38s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/f8d6249f88154eceaed5fe3ad9e62405 : SUCCESS in 33m 13s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/ea773daa54584194b56e0e90caf02d6e : TIMED_OUT in 2h 35m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/2aab138d3f12477fb5e20982056e5536 : SUCCESS in 2h 19m 07s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/43c19307c9274924bc3d5068c97511ca : SUCCESS in 1h 53m 19s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/238ad7e10f8f492db906d76b83b7d656 : SUCCESS in 1h 49m 59s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f4375439d4c64b75b55687b5579aca0c : FAILURE in 1h 23m 07s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/d0d5d9fda8ec4dee97030da4435e3995 : SUCCESS in 1h 25m 37s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/24418a148e7b4b62a9808c7206f72043 : SUCCESS in 1h 26m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7bc3c09c8088474a929a56ae03879454 : SUCCESS in 23m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/88baa29d6ae4467d9d7981db15059508 : SUCCESS in 21m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f84221af45494edaa0deef24d75446f4 : SUCCESS in 59m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6f390d90a7e4216950fcf17b115c15e : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a3f8d33ff62427d8c23367ed370f824 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/848f31a95c5d4b999f9c37b2f27e0200 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81b49ca9259e4929a0323f91bd853975 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70f245597a754dca9e87e962f9105fe6 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e336410b502a497b81db1599cfeaf868 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c15deb200e5a4c1989c6f3527daa135d : SUCCESS in 5m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d8d7d858c6c402aaebe8e5958e44d85 : SUCCESS in 22m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/312965798d1e430991b68c210aae8638 : SUCCESS in 26m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/30ac288a21604d78ba5927824b0a2f0f : SUCCESS in 53m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1365177ee0be47429a561deed5c88ab8 : SUCCESS in 55m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a329dd112bce44f3b63e245c9eae0c2d : SUCCESS in 1h 09m 18s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/6b1464b7cf8a4f1a87af6310dcc6214f : TIMED_OUT in 2h 32m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/1fdc9174872f4d62b4c81eca83659282 : SUCCESS in 1h 55m 23s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/12171b75c20342b79218842655913ef5 : TIMED_OUT in 2h 04m 01s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/78bb0cec70264ae2a104809217d880f3 : TIMED_OUT in 2h 01m 19s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":12},{"id":"deb87aaede1cb3593ce255e88d14cd39bcb804d2","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-08 17:03:03.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"39ede064593fe17971aba16ced02ab131b1cdd38","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 09:37:18.000000000","message":"Patch Set 12:\n\nrecheck\n\nwhat\u0027s with all the timeouts?","accounts_in_message":[],"_revision_number":12},{"id":"e91b5e99a3e0a76e9c7eb374e363281856ca6096","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-11 10:07:12.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/64deea9dbacb48e0af88d801dcef27cb : FAILURE in 28m 58s","accounts_in_message":[],"_revision_number":12},{"id":"5a23f469b34fc57db1950bd101b7b7ec662feb78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 12:09:02.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/85731960a57645b8a01d98fdeaa09081 : SUCCESS in 42m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/477699d9e7d44ea595e095e62ecac402 : SUCCESS in 52m 21s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/38bba8072c9a4207a79fb4df0bc331f3 : SUCCESS in 26m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/586cea4a8b8945709c79e22505235f28 : SUCCESS in 37m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a128fd101fbd4739b5c7467b5731d0c2 : SUCCESS in 1h 20m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/74b87d0867da4512aa2759c0de077b02 : SUCCESS in 1h 15m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7d23b8e67aa0466eb521ba5a10e69849 : SUCCESS in 1h 31m 36s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/edefbd5739934774890e269de3fdeaed : SUCCESS in 1h 28m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/55350affa0be4c0a87c1b3f4ab8922d1 : SUCCESS in 1h 47m 50s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/df87b54f69504fbf995e589a06cdb473 : SUCCESS in 55m 22s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/5c25ecb24210482d808a597e0f3a0b9f : SUCCESS in 56m 03s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/66bb85eda3a140f183ca3351a35785a0 : SUCCESS in 30m 27s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5aabf8086459462dafeaec5808988e05 : SUCCESS in 2h 10m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/866a7c6905334e9eaa54e5ddaadd1a17 : SUCCESS in 2h 21m 00s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/8a07db5009874f5fb36604dc0cb7e36e : TIMED_OUT in 2h 04m 28s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/82643959e84848cb80fb6464193b164e : TIMED_OUT in 2h 05m 24s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/233621521df34d4ba3ff4e7033bcc4a1 : FAILURE in 1h 26m 16s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/3ba2ad9e218e44b9ad6c6ece9c107651 : SUCCESS in 1h 22m 39s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/cb5220a793b4419cbb0c74c2cc5790a1 : SUCCESS in 1h 21m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/376108fa0c174d54b6c0e5139d0eb5f3 : SUCCESS in 26m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/226e36bfed1e42fe9e1811da72830ade : SUCCESS in 21m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c2d4725de7f145b6ad503324a411fafb : SUCCESS in 1h 00m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be4fa92ae2f248f7b9c856c6176af396 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1638b0e9eebe470fa3559a354f48ab94 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2284569003144ebaa13e602d9079ffb : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/839c71e484aa48aea610b7ee0c922739 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/276248851f3b447dafeec067b422149a : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d7afa687dfe49f4ad9dc9d04a05675f : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/130342d6c27e405aa28a5799e36942fd : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/582e5c0c31dc4002a702c6eb07fd75a4 : SUCCESS in 24m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f3a5c501bef04467a984604646af4588 : SUCCESS in 31m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/851b683cb0e24042bcd53d2b84581c8a : SUCCESS in 1h 01m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e2b9b7c95384d99b5ee08b5b0507f84 : SUCCESS in 1h 02m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b996f8838fb44288981c61bfdbe19674 : SUCCESS in 1h 12m 59s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e6dbee3a79c74e99a9fa614c48ef7b1d : SUCCESS in 1h 37m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/0f5c084c0d8d45f28f2e8b30c934cdbf : SUCCESS in 2h 20m 42s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/e93e2e1135894697adc69a7a1acb7542 : SUCCESS in 1h 51m 16s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0e954ca1812f402c93d719bd3d02e253 : SUCCESS in 1h 08m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":12},{"id":"4c8d8fa2fa86ed7947e69ea381d42e72cd803408","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 12:56:20.000000000","message":"Topic changed from wip/train-use-centos-repos to bp/centos-rhel-8","accounts_in_message":[],"_revision_number":12},{"id":"5e9d25708476aa29c41287c5d970650ea8a7629d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 12:58:43.000000000","message":"Patch Set 12: Code-Review+2 Review-Priority+1\n\nshipit","accounts_in_message":[],"_revision_number":12},{"id":"5a572fab40b669e963d1f74eb2f321b78f9306a5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-11 13:38:28.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nnothing outrageously wrong, +2 W+1","accounts_in_message":[],"_revision_number":12},{"id":"9e1b7e4ce5d7acaf24ca9c104a0dc63b0b32317f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:38:42.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"503644c83363160fa37f65ce27e56f7486cd4dd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:31:32.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a2e3293bbb74424087fc58f2bae61718 : SUCCESS in 56m 51s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/feaba273ffaa4c369faebb25ca330144 : SUCCESS in 34m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/18c60b29f9254a57ac72f6a09b3a1839 : SUCCESS in 1h 25m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6fde4c5590f7459db96f95f0d7f5968a : SUCCESS in 1h 14m 52s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d47147a1975a441aa09ff570f5c39cad : TIMED_OUT in 2h 34m 23s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7e9e8fa3bddf40548215c431cab90048 : SUCCESS in 46m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38c6e65ad38a4edcb20929f0e63bb6f5 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79faf4ecc1674210b88453f4b092efde : SUCCESS in 8m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f8a665eaf7243039d633f1c0e470bdb : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39a08adb48ff458ba07ebf0de38d7003 : SUCCESS in 8m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24a07b32fda04ca6b59d75d32d40318e : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db92213d21a244d7986f716cbafb3fe1 : RETRY_LIMIT in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7ceb7f714ac4802aa4ee79d314214c9 : SUCCESS in 8m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f0009131ff24f109fbf16479c91ed8d : SUCCESS in 32m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ac5c3c655f54bd3a7f3c5be75201887 : SUCCESS in 58m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8eb75ef9367349b694c73b44cccd3dbc : SUCCESS in 1h 16m 22s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":12},{"id":"f52d023832b39b3cc71949801a058ebc73906e9e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-12 09:26:18.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"2779908191a71d6a5f656731b3a200bcf8cd1239","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-12 10:59:26.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/8be49e67dfde4fce93ea83eca3d51141 : FAILURE in 26m 57s","accounts_in_message":[],"_revision_number":12},{"id":"d0465bbff4995d7ec24754271c94fd0f7e994347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 11:37:35.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5563e193f00448e4bfbcea86a257085f : SUCCESS in 47m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/57629fc59fc540d5bc43f822cb580f46 : SUCCESS in 54m 17s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3bab672361f747aa812974ff379ea762 : SUCCESS in 24m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8b9bd820fee64084b7e60db71c04cba3 : SUCCESS in 29m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7390f43eeae143e2ad1f221143206590 : SUCCESS in 1h 33m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/61c9417c2ddf4ef4a22ddf40a5a5a15e : FAILURE in 28m 53s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4937f624ede64422a1ea4f78dbce8b35 : SUCCESS in 1h 10m 54s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fc34bf87be9c4be3b60d5d3b386a97e0 : FAILURE in 26m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/08f3768ac4c845958b3483c5ac1a62db : SUCCESS in 1h 39m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8a6603f543b5469daf84dac40a948582 : SUCCESS in 1h 01m 06s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/518d99f94ba949e885e8db0d5f9c6223 : SUCCESS in 47m 45s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/1360a2c7556e4bc9a3fe0f1914cccbff : SUCCESS in 35m 11s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/428ebf4b0a594c6e8eabbe89705b4da1 : FAILURE in 1h 53m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/421c764aba2c48f190ffaa10011b9ba4 : FAILURE in 1h 39m 55s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/b8641444480f46c593e0deca27889f5f : TIMED_OUT in 2h 03m 45s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/ba830651d1934b8bad5f8a2607503e0b : FAILURE in 33m 17s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1e86a17440894188b3bacfb2610434ad : FAILURE in 1h 17m 44s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/429d2b515b094d588fa50ecfbc8ca7eb : SUCCESS in 1h 41m 35s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/9f89deb6ec954421beec010e0a7f83e6 : FAILURE in 30m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/32f7616efef4464bbe2efaf65d57a3ce : SUCCESS in 25m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/999e6b4e51924aa8a119501a83d98770 : SUCCESS in 19m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad5289bee941409bb4ef886fe5e39569 : SUCCESS in 59m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb8ad6e5d0e0447f8276b04b052ac99c : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d944ceec44d4ed9b77f8cf6b624d59d : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17f9f91cab274af2959a6fb71988b84a : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/181759beb125430a91d8585f53cbfc97 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7835edfa2414ca283cdf6b0f35104c7 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf8e636026041a584a96867412c51eb : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6efec09f511d414281b13d525d98fc2a : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0afc619cd1454edaa9cf7ea336f83108 : FAILURE in 22m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c72af3147fe14e6f855edc85cf862256 : FAILURE in 51m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ea8fc990b444c3bbdc3f0f58d3b43d4 : FAILURE in 33m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b23689456284a868a88a098c48e0bda : FAILURE in 21m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8c9aca7d97eb41728b155531e8b1ae31 : FAILURE in 22m 31s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/bb70d69dd82c4edc9ea240e9ec825b59 : FAILURE in 24m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/bcd4e8b4480842659e7870ffa4007f26 : FAILURE in 30m 07s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/aef2d580c1d24604a59eedb749b63a75 : FAILURE in 21m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/22b6e394fd9a4d0f9d70a40a153445d5 : FAILURE in 19m 29s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":12},{"id":"30376bb68866d82aa445849f6c816256261edefa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-14 19:12:06.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"c7f02ef5fb82c7ecc24c0b0cb99d886559805048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 21:16:30.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/351528cc4c474de0a452b638520902e4 : SUCCESS in 1h 04m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b0c24796c5a44b95a0c8fdbcf4ed3c22 : SUCCESS in 51m 29s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/374d4d19cf9443b7be14268d5d9527ef : SUCCESS in 24m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ad7014c0627949c889bf5aa4b9c62b29 : SUCCESS in 29m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/adf099741fa24382998521bfde3a989f : SUCCESS in 1h 20m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fe46f3eef4ed438fbb091356368c539a : SUCCESS in 1h 21m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/862810fe2d994735a62fc3f4eebb81a8 : SUCCESS in 1h 21m 46s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f6b0825852194184a7c73f999c6da49c : SUCCESS in 1h 37m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/75976b6908244aca892ab3c808f8601b : SUCCESS in 1h 44m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/55a1c580ea334e90888cd9198e313536 : SUCCESS in 57m 41s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/0475e4c1432344b49f65c8a0ee96e5d4 : SUCCESS in 46m 40s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/7cf42f4dee524f10ac06fa0d8e81d6cb : SUCCESS in 30m 35s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/bfb3dc208511449d8a6199fcf1ab2a55 : FAILURE in 1h 22m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f02372895f4d455083c442ace0450074 : FAILURE in 1h 32m 50s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/1742b93a17cb40c4983db5ab51a187d1 : TIMED_OUT in 2h 02m 58s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/7dd1fc95d89e49219d7bd42e2d5ba23c : SUCCESS in 1h 46m 24s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ce8d57e92ab94bec9de963ae7985ebe1 : FAILURE in 1h 28m 57s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/193c923571eb46509fd1a66e215002eb : SUCCESS in 1h 22m 58s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/b05282d461ce45c5a76e9872aa0a011e : SUCCESS in 1h 17m 01s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7c5a50582ecf42a5b0d2846a8675f185 : SUCCESS in 28m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8a51cc7227934099ba9363fd9ee4841b : SUCCESS in 23m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a2d0d08768404ced83031ba5d2096395 : SUCCESS in 1h 21m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c6e57c6819b45e892fa6156515e9014 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2daf16a19b2444186cbd9b8fbcd30a2 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b37e8a5e57149668fd10940ae0e71ad : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe3a722c79214c58a1f31581175aff76 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c50e91dc7174723837a53611aeb912f : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b0f583856d4436f920155ee69e4cce8 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/857199addcf143a49852fbd0ca654edf : SUCCESS in 5m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/947d718ff66e4fbcb2c21479a359d870 : FAILURE in 27m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bfc1009b555143d598065a119c4676dd : FAILURE in 29m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/22f176b3a65149a7b60592c182cd69f7 : FAILURE in 24m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/98b4e7710ae74520b62d52f268890608 : FAILURE in 20m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/782fd2212bb542ab83130a81ec0cb519 : FAILURE in 22m 54s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/2ce19d6421e54d27af6950c0f5b2a83d : FAILURE in 21m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/5a05c30c30b84057a48b31c926e55d35 : FAILURE in 23m 52s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d781264d727544ecbedcf0662693e852 : FAILURE in 22m 08s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/ff193461ae644750b031f10ec1b68de6 : FAILURE in 21m 32s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":12},{"id":"b812f9375be08eede9db1b8c28f9f383f603b46c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-15 15:48:26.000000000","message":"Patch Set 12: -Code-Review\n\nUbuntu jobs look unhappy","accounts_in_message":[],"_revision_number":12},{"id":"3e39b874da834b2f68f62cef57a3c4364af812e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-16 08:31:50.000000000","message":"Patch Set 12: Code-Review-1 -Workflow\n\n(1 comment)\n\nrecheck\n\nthe comment is unrelated to ubu failure (which is likely due to b0rken mirrors) but really needs fixing","accounts_in_message":[],"_revision_number":12},{"id":"b5c1a5d38aa8efa82a56eede9090c97f30473e24","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-17 13:02:18.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"039e7ec9e231c1f79187066c28b0d06bf2fe7ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 15:12:21.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/aa6f44ff199c41aab1d6c8403d5afc9d : SUCCESS in 42m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/59d4835d912b4afbbef16dbe1e1b5521 : SUCCESS in 45m 00s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/778ab994d1f441f8812c46a618fb73a3 : SUCCESS in 26m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2650b0a2569045a39fc9ecb2310b7097 : SUCCESS in 33m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f44f248944bb484f9444531322b8789e : SUCCESS in 1h 36m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9121f526226b4e6cbc1c770fa5eaa24e : SUCCESS in 1h 06m 26s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/014b009d7df04528b9a71ce3cc75e80e : SUCCESS in 1h 18m 18s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/822d69b6f86946fe8c00570c4da9e68e : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2d10f85924344a519bd1053115ab7061 : SUCCESS in 1h 34m 13s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/652e4ce02a6a4ef58f8c4303080e3a86 : SUCCESS in 53m 09s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/dcd9882cc7df473d92cb7c92bfb1f585 : SUCCESS in 57m 09s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/8939a7383004407881d3a8ad5a805a1c : SUCCESS in 29m 34s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/b5a3c6d0c7e54af8b46e718d68948385 : FAILURE in 1h 36m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/14009b1357ba48fdafbaeb5f98107351 : FAILURE in 1h 41m 55s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/bc4867a45d0746d6a2682f6c6cd3f29c : TIMED_OUT in 2h 02m 44s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/aae82a89b4374a07817e56b15196b702 : SUCCESS in 1h 43m 20s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c47c78f680fc410fa9efc2c65762758a : FAILURE in 1h 35m 49s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/08e6dc27d4004569b4cb3e57a119e6d7 : SUCCESS in 1h 36m 45s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/c4391fad55444484828ca37e042f84fc : SUCCESS in 1h 13m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3b3af08b41934f50b87dc0aa10fc7c55 : SUCCESS in 25m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e3275733502246bd821f53df4faaa2c5 : SUCCESS in 21m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ece76f9a9d7343e8b1842917add40cbe : SUCCESS in 1h 00m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9851e4e455f241fcb4babdb635d3b2fb : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac208ea909f84534bfe5a18adc24c094 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64483269481e44359efa1e38b06a0ee3 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/747b791588bf4e688a46cd4917fa6e04 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d79743649ec44938b223aba607f96e5e : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04046bafa3104a2b8bc7d9bb72c20a43 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc283f81dfba4a0a936fa0a57dc38f42 : SUCCESS in 3m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f425259afd8e48d780a508290781d0f2 : SUCCESS in 23m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fac412568f264588a4bcf231112aa426 : SUCCESS in 36m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf6bc6026be24e4ba75487e92bddc104 : SUCCESS in 55m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9927c9d84c9744689cf56c4144cf93fa : SUCCESS in 55m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1cf5bc34b8704912bd55cc8d12126093 : SUCCESS in 1h 12m 34s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/74651d1b432e4e7e9cba218213fb9311 : SUCCESS in 1h 30m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/080c0fc338734889a0bdb80022fd85ca : SUCCESS in 1h 45m 05s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2c1c01c13e94428a85ffae8d25df19d1 : SUCCESS in 1h 19m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/4a9707a40df943d0a43ac56b1276984b : SUCCESS in 58m 40s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":13},{"id":"be5df35a4063a60dbd081fb46978169668f671c1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-18 06:51:10.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"1bb09f4c3ada22af528a89674e9ec5f398a16a9d","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-05-18 08:05:05.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"19b0df660b5de8d1641f92be79fce2986e467159","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-18 08:56:33.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"b6e9a3867b019b313ee3e0ad0d9ce92aee23fe3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 08:56:46.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"f3c6efa5507c2ba23f3685b845c43dd903a1449f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 10:38:48.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c44a1dee065b4e53b9b232f57681b677 : SUCCESS in 56m 30s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/71c7df0be0af48689fd44ee7ea1f9112 : SUCCESS in 36m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7f5820508f0b45a0b9402751e7643eef : SUCCESS in 1h 19m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cc3c0ca146ef476284ded8d5a4ef71ee : SUCCESS in 1h 18m 29s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3ff44b70f598414dbbb36869ce46970e : SUCCESS in 1h 36m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/23c0949ef8d146fda02c4fce4fd284e1 : SUCCESS in 47m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78e6175454344269b1e227e00b330af1 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50aef547ad104c7b9caea0743a59af7e : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7789dc1e429147a9808d6fd246f2b190 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a08fd0e33b2d44a2bf8ab2686f02ebb3 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c248b9891ce4f32937467388555059f : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bed0a018b364d3aa5b6b44e2bd4f134 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6fc85176e6d4b5382294d83683dd5dd : SUCCESS in 3m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7cf51e2fe6c14609ad4944bb7d87fce2 : SUCCESS in 26m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b0491250f514e4d8c539ed3c8eb284b : SUCCESS in 1h 00m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c8984f25d83345e990dd519e4c36e7e8 : SUCCESS in 1h 12m 13s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":13},{"id":"8737c1e9c9c1bb5e03141543a6962ec2255bbb3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 10:38:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"6dc8626bde94aaea080d9143914792228e740edb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 10:40:57.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a00f38bdbc4448a38eeb4a725d009418 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4eba85d6bc94752a94b4d9051239506 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"bd21ac14d73e01c97a780ec8c12fb99e0b8c4b10","revisions":{"2e10aa2f644ea8ed76f398ae17367af116f606f6":{"kind":"REWORK","_number":1,"created":"2020-04-20 19:02:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/1"}}},"commit":{"parents":[{"commit":"0cae8e2be35156265fa9ccf5f60f8c5b4da2bb00","subject":"Merge \"Fix that the cyborg database init failed.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cae8e2be35156265fa9ccf5f60f8c5b4da2bb00"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 19:01:31.000000000","tz":120},"subject":"[WIP] switch to official CentOS 8 repositories","message":"[WIP] switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\n\u0027centos-release-*\u0027 packages are not prepared yet.\n\nAlso three packages are missing:\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n- python3-networking-vmware-nsx\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e10aa2f644ea8ed76f398ae17367af116f606f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e10aa2f644ea8ed76f398ae17367af116f606f6"}]},"branch":"refs/heads/stable/train"},"d942cac9e27b4b670f9fd14cdaaa5abd553a7849":{"kind":"REWORK","_number":2,"created":"2020-04-21 10:24:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/2"}}},"commit":{"parents":[{"commit":"0cae8e2be35156265fa9ccf5f60f8c5b4da2bb00","subject":"Merge \"Fix that the cyborg database init failed.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cae8e2be35156265fa9ccf5f60f8c5b4da2bb00"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-21 10:24:11.000000000","tz":120},"subject":"[WIP] switch to official CentOS 8 repositories","message":"[WIP] switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nSeveral \u0027centos-release-*\u0027 packages are not prepared yet. EPEL is\ndisabled like it is in Ussuri.\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\nAlso some packages are missing:\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n- python3-networking-vmware-nsx\n- python3-requests-kerberos\n\nRDO team work on adding/checking them.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d942cac9e27b4b670f9fd14cdaaa5abd553a7849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d942cac9e27b4b670f9fd14cdaaa5abd553a7849"}]},"branch":"refs/heads/stable/train"},"1d9d7c0163cda13efee2ca755d49769dbaa541d3":{"kind":"REWORK","_number":3,"created":"2020-04-22 12:10:53.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/3"}}},"commit":{"parents":[{"commit":"0cae8e2be35156265fa9ccf5f60f8c5b4da2bb00","subject":"Merge \"Fix that the cyborg database init failed.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cae8e2be35156265fa9ccf5f60f8c5b4da2bb00"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-22 12:10:23.000000000","tz":120},"subject":"[WIP] switch to official CentOS 8 repositories","message":"[WIP] switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nSeveral \u0027centos-release-*\u0027 packages are not prepared yet. EPEL is\ndisabled like it is in Ussuri.\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d9d7c0163cda13efee2ca755d49769dbaa541d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d9d7c0163cda13efee2ca755d49769dbaa541d3"}]},"branch":"refs/heads/stable/train"},"d41c68f8a0e9524881e0c43507c5018e82dcd014":{"kind":"REWORK","_number":4,"created":"2020-04-23 09:36:18.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/4"}}},"commit":{"parents":[{"commit":"0cae8e2be35156265fa9ccf5f60f8c5b4da2bb00","subject":"Merge \"Fix that the cyborg database init failed.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cae8e2be35156265fa9ccf5f60f8c5b4da2bb00"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-23 09:35:38.000000000","tz":0},"subject":"[WIP] switch to official CentOS 8 repositories","message":"[WIP] switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nSeveral \u0027centos-release-*\u0027 packages are not prepared yet. EPEL is\ndisabled like it is in Ussuri.\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nAArch64 is not supported until CentOS 8.2 release.\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d41c68f8a0e9524881e0c43507c5018e82dcd014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d41c68f8a0e9524881e0c43507c5018e82dcd014"}]},"branch":"refs/heads/stable/train"},"6f49d8e5cc07cf8235a894efee2774489d1665db":{"kind":"REWORK","_number":5,"created":"2020-04-23 17:06:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/5"}}},"commit":{"parents":[{"commit":"0cae8e2be35156265fa9ccf5f60f8c5b4da2bb00","subject":"Merge \"Fix that the cyborg database init failed.\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cae8e2be35156265fa9ccf5f60f8c5b4da2bb00"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-23 17:06:31.000000000","tz":120},"subject":"[WIP] switch to official CentOS 8 repositories","message":"[WIP] switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nSeveral \u0027centos-release-*\u0027 packages are not prepared yet. EPEL is\ndisabled like it is in Ussuri.\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nAArch64 is not supported until CentOS 8.2 release.\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f49d8e5cc07cf8235a894efee2774489d1665db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f49d8e5cc07cf8235a894efee2774489d1665db"}]},"branch":"refs/heads/stable/train"},"91c37f0ac8082bad13633b6fa6c6528814a5893a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-04-26 16:37:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/6"}}},"commit":{"parents":[{"commit":"98860981c7f3c45b3d8e8a0845555b88ad1b7d86","subject":"Add nfs dependencies to glance container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98860981c7f3c45b3d8e8a0845555b88ad1b7d86"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-26 16:37:27.000000000","tz":0},"subject":"[WIP] switch to official CentOS 8 repositories","message":"[WIP] switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nSeveral \u0027centos-release-*\u0027 packages are not prepared yet. EPEL is\ndisabled like it is in Ussuri.\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nAArch64 is not supported until CentOS 8.2 release.\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91c37f0ac8082bad13633b6fa6c6528814a5893a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91c37f0ac8082bad13633b6fa6c6528814a5893a"}]},"branch":"refs/heads/stable/train"},"59110424b0ef62d4260647af37b1ddebe5c84d39":{"kind":"REWORK","_number":7,"created":"2020-04-27 15:21:44.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/7"}}},"commit":{"parents":[{"commit":"98860981c7f3c45b3d8e8a0845555b88ad1b7d86","subject":"Add nfs dependencies to glance container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98860981c7f3c45b3d8e8a0845555b88ad1b7d86"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-27 15:20:51.000000000","tz":120},"subject":"switch to official CentOS 8 repositories","message":"switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nSeveral \u0027centos-release-*\u0027 packages are not prepared yet. EPEL is\ndisabled like it is in Ussuri (RDO packages may conflict with it).\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nAArch64 is not supported until CentOS 8.2 release.\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59110424b0ef62d4260647af37b1ddebe5c84d39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59110424b0ef62d4260647af37b1ddebe5c84d39"}]},"branch":"refs/heads/stable/train"},"544a1223587d3be5dd1b9d1e4453f6880d7acb95":{"kind":"REWORK","_number":8,"created":"2020-04-27 18:03:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/721329/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/8"}}},"commit":{"parents":[{"commit":"98860981c7f3c45b3d8e8a0845555b88ad1b7d86","subject":"Add nfs dependencies to glance container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98860981c7f3c45b3d8e8a0845555b88ad1b7d86"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-27 18:01:01.000000000","tz":0},"subject":"switch to official CentOS 8 repositories","message":"switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nSeveral \u0027centos-release-*\u0027 packages are not prepared yet. EPEL is\ndisabled like it is in Ussuri (RDO packages may conflict with it).\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nAArch64 is not supported until CentOS 8.2 release.\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/544a1223587d3be5dd1b9d1e4453f6880d7acb95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/544a1223587d3be5dd1b9d1e4453f6880d7acb95"}]},"branch":"refs/heads/stable/train"},"af803d830b4c5cb236d2b453c90a2bc61c018cc4":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-04-30 13:23:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/9"}}},"commit":{"parents":[{"commit":"dc560d132931401edbda8e81f684512e10e14163","subject":"drop tripleo-common-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc560d132931401edbda8e81f684512e10e14163"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-30 13:23:21.000000000","tz":0},"subject":"switch to official CentOS 8 repositories","message":"switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nSeveral \u0027centos-release-*\u0027 packages are not prepared yet. EPEL is\ndisabled like it is in Ussuri (RDO packages may conflict with it).\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nAArch64 is not supported until CentOS 8.2 release.\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af803d830b4c5cb236d2b453c90a2bc61c018cc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af803d830b4c5cb236d2b453c90a2bc61c018cc4"}]},"branch":"refs/heads/stable/train"},"d9c567be7e78d8b11eeecae05336cc0d578b069b":{"kind":"REWORK","_number":10,"created":"2020-04-30 15:57:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/10"}}},"commit":{"parents":[{"commit":"dc560d132931401edbda8e81f684512e10e14163","subject":"drop tripleo-common-tempest-plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc560d132931401edbda8e81f684512e10e14163"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-30 15:57:41.000000000","tz":120},"subject":"switch to official CentOS 8 repositories","message":"switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nSeveral \u0027centos-release-*\u0027 packages are not prepared yet. EPEL is\ndisabled like it is in Ussuri (RDO packages may conflict with it).\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nAArch64 is not supported until CentOS 8.2 release.\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9c567be7e78d8b11eeecae05336cc0d578b069b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9c567be7e78d8b11eeecae05336cc0d578b069b"}]},"branch":"refs/heads/stable/train"},"73930aad753ef806152ef668e0c0b5d08bb4ab0c":{"kind":"REWORK","_number":11,"created":"2020-05-05 07:04:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/11"}}},"commit":{"parents":[{"commit":"4c8863b0a877a00a810dab55be818163aeb83c61","subject":"Bump versions for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c8863b0a877a00a810dab55be818163aeb83c61"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-05 07:02:42.000000000","tz":120},"subject":"switch to official CentOS 8 repositories","message":"switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nEPEL is disabled like it is in Ussuri (RDO packages may conflict with\nit).\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nAArch64 is not supported until CentOS 8.2 release.\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73930aad753ef806152ef668e0c0b5d08bb4ab0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/73930aad753ef806152ef668e0c0b5d08bb4ab0c"}]},"branch":"refs/heads/stable/train"},"9316c1f643161533037b9788613733300869bfad":{"kind":"REWORK","_number":12,"created":"2020-05-05 10:12:43.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/12"}}},"commit":{"parents":[{"commit":"4c8863b0a877a00a810dab55be818163aeb83c61","subject":"Bump versions for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c8863b0a877a00a810dab55be818163aeb83c61"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-05 10:12:40.000000000","tz":0},"subject":"switch to official CentOS 8 repositories","message":"switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nEPEL is disabled like it is in Ussuri (RDO packages may conflict with\nit).\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nAArch64 is not supported until CentOS 8.2 release.\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9316c1f643161533037b9788613733300869bfad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9316c1f643161533037b9788613733300869bfad"}]},"branch":"refs/heads/stable/train"},"bd21ac14d73e01c97a780ec8c12fb99e0b8c4b10":{"kind":"REWORK","_number":13,"created":"2020-05-17 13:02:18.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/721329/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/721329/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/721329/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/721329/13"}}},"commit":{"parents":[{"commit":"4c8863b0a877a00a810dab55be818163aeb83c61","subject":"Bump versions for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c8863b0a877a00a810dab55be818163aeb83c61"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-20 13:13:31.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-17 13:02:14.000000000","tz":0},"subject":"switch to official CentOS 8 repositories","message":"switch to official CentOS 8 repositories\n\nRDO builds of stable/train for CentOS 8 went into phase when we do not\nneed to use delorean builds.\n\nEPEL is disabled like it is in Ussuri (RDO packages may conflict with\nit).\n\nErlang for RabbitMQ is used from CentOS 8 \u0027rabbitmq-38\u0027 repo while\nRabbitMQ is still 3.7.* from same place as usual.\n\n\u0027python3-pika\u0027 is dropped from openstack-base as it is EPEL only package\nand is not used by any image (dropped in Ussuri in\nhttps://review.opendev.org/#/c/710003/ patch)\n\nAArch64 is not supported until CentOS 8.2 release.\n\nChange-Id: I9d3ed54f4d10ef7ad2d52ed3aeca15f3968cd163\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd21ac14d73e01c97a780ec8c12fb99e0b8c4b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd21ac14d73e01c97a780ec8c12fb99e0b8c4b10"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
