)]}'
{"id":"openstack%2Fkolla~721764","triplet_id":"openstack%2Fkolla~stable%2Fqueens~I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6","project":"openstack/kolla","branch":"stable/queens","topic":"bug/1814552-stable/queens","hashtags":[],"change_id":"I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6","subject":"nova-libvirt: add UEFI packages to support UEFI instances","status":"ABANDONED","created":"2020-04-22 01:46:37.000000000","updated":"2020-05-27 10:42:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f014d87479f8a9092b980aec1616adb3b6d33a2a","_number":721764,"virtual_id_number":721764,"owner":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"date":"2020-04-22 08:46:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-22 02:42:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"date":"2020-04-22 01:46:37.000000000","_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 01:46:37.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2020-04-22 02:42:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-22 08:46:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"1a06f554a700e534f14ccb3e69d5d739732568f0","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2020-04-22 01:46:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"e4a043fe9b224d85a8e8a032200dc0271c38f77a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 02:42:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/73632c9cd0dc4a918e6001bad53323d7 : FAILURE in 33m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f8c744c332ea4b318c4e51d7239558b0 : FAILURE in 46m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5ad3d4d5d80549aca5021af7ae1b1825 : RETRY_LIMIT in 2m 13s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f2013254b3d74565a6f76280f8b4432a : FAILURE in 41m 36s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/dfd56cc532014daa8e84303bc124994d : FAILURE in 41m 15s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c4495e1008b54a78a8e81ff1f1fe0aeb : POST_FAILURE in 51m 47s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/33bdfa46fb564dbca708c57a127f8e73 : FAILURE in 43m 47s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3e3047ddefe94f058839038effabeb50 : FAILURE in 44m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c6cb830cef487fbad7f14126b54401 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0608658b89bc4785bd2a338a7b97f26d : SUCCESS in 4m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/df1c34688d1d4a9cb3469303904cf5bf : SUCCESS in 4m 12s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d320c7450c55403994bbdd5801fa293f : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7601e2663da4ebd9cd5405f7ca94e6c : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/665d61332f0941ba9224cab955b23282 : FAILURE in 5m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eab93e3d3dad481883081a0a98fbc1b8 : FAILURE in 5m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de51db8d144b40cebb41bb54b250a350 : POST_FAILURE in 2m 45s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5aff3513b6dd47ff9343a81c4b8028ef : POST_FAILURE in 3m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e60095f626ec76b93ce5949dd7b42aa084ae698","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-22 08:46:44.000000000","message":"Patch Set 1:\n\nyou know the drill, fix CI or give it up :-)","accounts_in_message":[],"_revision_number":1},{"id":"102bd8cbc4bccefc2c0a5bc7bf9fb60e3194665e","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2020-05-27 10:42:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b85d3b34dec9a631e39929b2bb9182cf292a2837","revisions":{"b85d3b34dec9a631e39929b2bb9182cf292a2837":{"kind":"REWORK","_number":1,"created":"2020-04-22 01:46:37.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/64/721764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/721764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/721764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/721764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/721764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/721764/1"}}},"commit":{"parents":[{"commit":"53a83b46d0b7d0de07e28fcfbdc50356d7468bff","subject":"Merge \"Add xfsprogs to nova-compute\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53a83b46d0b7d0de07e28fcfbdc50356d7468bff"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-09-16 06:46:41.000000000","tz":480},"committer":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2020-04-22 01:46:37.000000000","tz":0},"subject":"nova-libvirt: add UEFI packages to support UEFI instances","message":"nova-libvirt: add UEFI packages to support UEFI instances\n\nFix inability to run UEFI-based images/instances by installing UEFI\npackages also in nova-libvirt image which is not based on nova-base.\n\nAdapted for Stein and below (e.g. Rocky) by removing\nany C8 support.\nAdapted for Rocky due to later base_distro refactorings.\n\nCloses-Bug: #1814552\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\nCo-authored-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nChange-Id: I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6\n(cherry picked from commit 15b68c15c1dd0c2df1cc49b550f2fa03a4c65e8e)\n(cherry picked from commit ffaf585d0f05ccd9123459bb0cdf345af5b09573)\n(cherry picked from commit 757fe57563ed54e53c14c8cfae780b71fb49b7ba)\n(cherry picked from commit f39fc9b0a77c2c7d75afedb217bb0f8d6e7eed19)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b85d3b34dec9a631e39929b2bb9182cf292a2837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b85d3b34dec9a631e39929b2bb9182cf292a2837"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
