)]}'
{"id":"openstack%2Fkolla~721999","triplet_id":"openstack%2Fkolla~master~I266cdfa09efcba1c54efaa3c16ffe556808b3673","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I266cdfa09efcba1c54efaa3c16ffe556808b3673","subject":"aarch64/centos8: no fluentd, rabbitmq, kolla-toolbox","status":"MERGED","created":"2020-04-22 15:18:41.000000000","updated":"2020-04-27 15:11:06.000000000","submitted":"2020-04-27 15:08:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721999-1588000084600-71cafbf4","meta_rev_id":"42dac3474c2d642a997d3d0243587d692d6d373a","_number":721999,"virtual_id_number":721999,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 15:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-27 09:14:27.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-27 09:31:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-27 13:44:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-27 13:44:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-04-27 08:48:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 09:14:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-27 09:31:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-27 13:44:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-27 15:08:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2306287fd73e35d0d25d10c2245ae8fecec711f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 15:18:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a44bb4029932acbb5c14e3e5daf4b03231ecee8f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-22 15:54:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"18e0f56e72ff0b5e7685563e8ba0a6a11a83e90b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-22 17:01:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/a2ba314b2743402b84109f10b413cd32 : SUCCESS in 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cca4dd5fadd61c5ad3921badbab79e72c26bcf22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 17:48:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0a73402d3d08467b83da073113813247 : SUCCESS in 26m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/36edb783b69d4254acfa2c82c5560b14 : SUCCESS in 29m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6c7e7949654b4fa3aa392b970d5b34c8 : SUCCESS in 1h 51m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c88eb22a97454abca4e3ebcef0273d2e : SUCCESS in 1h 07m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/83745d75a55841f9a1dcd279656100a5 : SUCCESS in 1h 30m 24s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/4ba0dada07f74a79be58522a286eca28 : SUCCESS in 52m 03s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7d7eb7987436423fab37da5166c2c75d : SUCCESS in 21m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5cab7f7d04d141fb87acfc61e14ee5d3 : SUCCESS in 18m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cfa40ff8aa0042aab496417450aea172 : SUCCESS in 1h 05m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c411f520d0e4785b3a564947fabbfb6 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63b0f52c60924969b882da4984599357 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb3ce21f6f8344989d283ccfc3711120 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54b6ef9f973f439ba7e50ec242145e89 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1815616523a24f9599f9fd209d6637c2 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99ddec5799c94b468ab0c21a84c9c1b0 : SUCCESS in 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70ba287d41f8428ba23c39a20b5b79c0 : SUCCESS in 7m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/046a43eb17f840b0b95263f659fc8931 : SUCCESS in 23m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4ee46fcc98948ffb310808b07e602ef : SUCCESS in 1h 08m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f09e15dc3fdf41f1980c13c58383e1d4 : SUCCESS in 1h 03m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f52816d73bcc40eaaf1ff00c9e30ff46 : SUCCESS in 1h 26m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/49da8e7392084b6d84721efcb3bfb07c : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":2},{"id":"fc26abdfcfdbe3c9ec264f6cf19a3ba55d004ade","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 18:02:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/92dad72124b3488ab88d69ab434d48bc : FAILURE in 1h 06m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b250a5673b0402d9ba97dd0dfa13a94 : SUCCESS in 1h 02m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c97e236174e843aa8dbec0739e75e5fb : SUCCESS in 1h 56m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d9bf0a9e9e54e3f95672ba8cf97d9a5 : SUCCESS in 1h 04m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d92244ecf4ec3fd1e2ec29cfea4519e856e6638","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 09:19:55.000000000","message":"Patch Set 2: Code-Review-1\n\nAlso fluentd failing","accounts_in_message":[],"_revision_number":2},{"id":"352907e36ed779b884314b3bc6f4b5f95d4e2a5b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-27 08:47:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"744c5bde3673b6cf6e24ab8e56708becde5230e6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-27 08:48:16.000000000","message":"Patch Set 3: Review-Priority+1\n\nFixes C8/AArch64 CI","accounts_in_message":[],"_revision_number":3},{"id":"f05fe4c768e5fec7511d13c1e6402d810b00bece","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-27 09:14:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/bb4d3714cc924b1f94c5a8c19b857f40 : SUCCESS in 25m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"858afbe12fb155fad94081804721c5e424fd7bd9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 09:31:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3828c885e1d20687f773dfbff7aa6c09a4b78d71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:52:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/98bc2d5b7fae4f17ae45321a6b278583 : SUCCESS in 55m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b19ccee8e039481ea3bef86dc7ac7774 : RETRY_LIMIT in 4m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0fe98241b89142239fac21518673447a : FAILURE in 4m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e046f447ff084d69971b3555b06addb5 : RETRY_LIMIT in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a36e1402a3f9c1aa19d7c960b0ea14bb4ce64b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 10:16:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/aace583cf4b443879d2c147e95bb32b8 : SUCCESS in 33m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/246e8d3ad4164134a9b65a7bef17e440 : SUCCESS in 39m 09s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e13cf2297bed41199f8471c9ce956c19 : SUCCESS in 1h 17m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7e51bca6a204426993db39abe391b37e : SUCCESS in 1h 08m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/7ed98cce509b4e438fc58ff050cd2eb8 : SUCCESS in 1h 15m 37s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ace464352faf4e96a193f9b7f3e71383 : SUCCESS in 45m 40s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7ed3b0062fa04919b0dadeb9e69519e6 : SUCCESS in 21m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4f539ca8d526482da48c3922ac3288ec : SUCCESS in 18m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7a4fd48121e74b9098e92a0498682751 : SUCCESS in 1h 00m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e9c80be9af74ddf817b026f818bde0f : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c8285f7e5d44681b8f88f543bbf1aa6 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b94e9425f35043c0b1b5fc01cf6f5943 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4facaa5dfb44d9a844698cbcc49543c : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/251ace7f36a146a89ebaeb6145caa6e0 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/535aec46e6784151ba786d9601a15e6b : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed046a92b071457982948518768ab6ec : SUCCESS in 6m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f20cc15d7ea94a92b9f5c22fb3b9cdd3 : SUCCESS in 39m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a26faa9a653143ecbdef72dc9aa6af19 : SUCCESS in 41m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07c3bd404eac4f66922c1824637d804b : SUCCESS in 1h 16m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1207557180e54f66b4c2bec07df17317 : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1716321f7cad40c5b34bd70639346979 : SUCCESS in 1h 26m 58s","accounts_in_message":[],"_revision_number":3},{"id":"1bf70308264d4bc5dcdc50d67e06823e8aec8146","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-27 13:44:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6ad06cee2c569a532659b127350d04de7a89c915","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 13:44:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"112077f863f18b74a455d9b38ede1fd0cc456e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:08:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a42d537a7a7d48f4be5bff69054c3c24 : SUCCESS in 29m 38s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/84ca3b25950a44da929373e37ba01dd1 : SUCCESS in 1h 15m 50s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/311afaa2a55c44a891cb86449fd059a0 : SUCCESS in 1h 18m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/317c9fad3bf74dd5a59804e3a2eaaf90 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7dce2c4d01c45f08d3adbc17853479a : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5b7884d68fb4430ae9efe241f6b27f5 : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46f30fa2406847f6b46677dc1080f073 : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e6190f727b94cc8ade22288bf4abfa7 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76fc049e20524025b9e56d11d5927b9e : SUCCESS in 8m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7918d3b05aa54310869fe197aadcdc08 : SUCCESS in 35m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a1789ff1928440f8fd754dd32d918d3 : SUCCESS in 1h 10m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/580a6735bf9d447d94a09993be826244 : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":3},{"id":"50bbff07497f089b70f0134ef9da2df1267cc543","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:08:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"42dac3474c2d642a997d3d0243587d692d6d373a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:11:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28cfb7c99d3b4224a5c4c7ec2827fc78 : SUCCESS in 1m 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d3076b640794d2982cc899d85be4d98 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"39a6e9a9a463170f37be24a0fe2f624583b2755e","revisions":{"7dd2d9f18760653f1ec5f38e4601991bfa6077b9":{"kind":"REWORK","_number":1,"created":"2020-04-22 15:18:41.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/721999/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/721999/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/721999/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/721999/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/721999/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/721999/1"}}},"commit":{"parents":[{"commit":"d318e5c1a193914ee90eab1c9a762d61d1b07acb","subject":"Merge \"Zun: add CNI loopback binary to the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d318e5c1a193914ee90eab1c9a762d61d1b07acb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-22 15:16:08.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-22 15:17:59.000000000","tz":120},"subject":"No kolla-toolbox \u0026 rabbitmq for aarch64/centos8","message":"No kolla-toolbox \u0026 rabbitmq for aarch64/centos8\n\nAArch64 lacks Erlang in CentOS 8 - we have to wait for 8.2 release to\nget it. So no RabbitMQ for us.\n\nChange-Id: I266cdfa09efcba1c54efaa3c16ffe556808b3673\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7dd2d9f18760653f1ec5f38e4601991bfa6077b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7dd2d9f18760653f1ec5f38e4601991bfa6077b9"}]},"branch":"refs/heads/master"},"b7f52d95361df13bd094bca95f00bdab9cb17280":{"kind":"REWORK","_number":2,"created":"2020-04-22 15:54:49.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/721999/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/721999/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/721999/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/721999/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/721999/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/721999/2"}}},"commit":{"parents":[{"commit":"d318e5c1a193914ee90eab1c9a762d61d1b07acb","subject":"Merge \"Zun: add CNI loopback binary to the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d318e5c1a193914ee90eab1c9a762d61d1b07acb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-22 15:16:08.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-22 15:54:41.000000000","tz":120},"subject":"No kolla-toolbox \u0026 rabbitmq for aarch64/centos8","message":"No kolla-toolbox \u0026 rabbitmq for aarch64/centos8\n\nAArch64 lacks Erlang in CentOS 8 - we have to wait for 8.2 release to\nget it. So no RabbitMQ for us.\n\nChange-Id: I266cdfa09efcba1c54efaa3c16ffe556808b3673\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7f52d95361df13bd094bca95f00bdab9cb17280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7f52d95361df13bd094bca95f00bdab9cb17280"}]},"branch":"refs/heads/master"},"39a6e9a9a463170f37be24a0fe2f624583b2755e":{"kind":"REWORK","_number":3,"created":"2020-04-27 08:47:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/721999/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/721999/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/721999/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/721999/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/721999/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/721999/3"}}},"commit":{"parents":[{"commit":"63a55048912c2132d00e8b6d0b79e6d0d4835500","subject":"Merge \"Add mariadb_clustercheck image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63a55048912c2132d00e8b6d0b79e6d0d4835500"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-22 15:16:08.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-27 08:47:23.000000000","tz":120},"subject":"aarch64/centos8: no fluentd, rabbitmq, kolla-toolbox","message":"aarch64/centos8: no fluentd, rabbitmq, kolla-toolbox\n\nAArch64 lacks Erlang in CentOS 8 - we have to wait for 8.2 release to\nget it. So no RabbitMQ for us.\n\nAnd no Fluentd.\n\nChange-Id: I266cdfa09efcba1c54efaa3c16ffe556808b3673\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a6e9a9a463170f37be24a0fe2f624583b2755e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39a6e9a9a463170f37be24a0fe2f624583b2755e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
