)]}'
{"id":"openstack%2Fkolla~722102","triplet_id":"openstack%2Fkolla~master~I92c81c77e6a16570a108cde8031f7977930fb02a","project":"openstack/kolla","branch":"master","topic":"bug/1874298","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-10 09:41:40.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-10-11 08:42:20.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-10-11 08:42:20.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-10-11 08:42:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I92c81c77e6a16570a108cde8031f7977930fb02a","subject":"Fix writable rootwrap/privsep config","status":"MERGED","created":"2020-04-22 17:49:02.000000000","updated":"2022-10-11 08:43:33.000000000","submitted":"2022-10-11 08:42:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722102-bug/1874298","meta_rev_id":"d3595a4530777a464671b54063f053f95e1c3f57","_number":722102,"virtual_id_number":722102,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-11 08:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-10-10 15:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-10-11 07:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-10-11 07:27:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 21:13:27.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 01:32:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-23 20:07:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-04 16:04:11.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-10-10 07:32:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-10-10 14:52:15.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-10-11 07:27:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e392224699216222733e6055e4d62aa7c93243c8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-22 17:49:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaf00767dce82f6ccc883ba76370c9c0011d934a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-22 17:49:49.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"7a0091f6155d1d571ce06caa9ce8f6cc06c6fba7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-22 18:33:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/38f39341f4cb4012a0f8db72cc610301 : SUCCESS in 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d156ee0602b962441b7c7d1b7d06f3cf3891df0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 19:33:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/efdf0b1f97aa4e1aad562ac828907b39 : FAILURE in 48m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/34a2bb66829342da851bc7df7decbe97 : SUCCESS in 48m 24s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e9235f5685146d58206827e5fce9668 : SUCCESS in 1h 38m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/64e018c3e73c4b6b83b77e2792abe542 : SUCCESS in 47m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c68b5c8cf3d5e26f1473ac8884e17638755814c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 20:33:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c47e66af06ad4369b97066324dde3971 : SUCCESS in 26m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/10920a0bbea848b8946d080b7c951c65 : SUCCESS in 35m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/58f534892cda4c23bacc7cdd9f06c255 : SUCCESS in 1h 15m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4c79937e395a4ee8a8997aa4769ceccd : SUCCESS in 1h 11m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0364d79625b04f4a8a7f0da701ebf574 : SUCCESS in 1h 28m 30s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/656d138791b54a58a93c6f1d4f75daa3 : SUCCESS in 45m 44s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/4effb8db6af14d8f92641bda14d63c4e : SUCCESS in 2h 35m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/91f49145c32f49319b43d0cf78f6be74 : SUCCESS in 2h 40m 24s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/e804397d3619470c943a7b36425adf78 : SUCCESS in 1h 52m 32s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d1eaf78de64648ddab97dbc4ae3a6620 : SUCCESS in 1h 51m 47s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/afccb2711db543c19b0102253cf2aec5 : SUCCESS in 1h 47m 10s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e26fe847448b45a0b03c03dd3a664eb3 : SUCCESS in 22m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a874c2742b254b318e3749a70c14e3d6 : SUCCESS in 17m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/275d327839c4416aba2a67c8a4d3bcda : SUCCESS in 1h 03m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7d637c6542640cbb4e3f99754054785 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fb4846328d3415d9d4df401ca8dffeb : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4e5ee0415614659a0fb011012ae1192 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44323aab4ffb458daa68a1f88e77bb0a : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10df8055f0a24a10b7d75857681149d6 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e325b680a3d4140b9ef8cb553c03e04 : SUCCESS in 7m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bb54a0b89c0d48e5a390a1ec7692ab45 : FAILURE in 22m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f093bdc0796440658a5305ecd3aa3c57 : SUCCESS in 29m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/02005c9e10f24409b0cb99ebf214e85b : SUCCESS in 1h 04m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1aead304c3e448ddbd4db8d1e3c15b63 : FAILURE in 18m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0451bc0230eb463b8d231aeef142e5d6 : SUCCESS in 1h 19m 48s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/a8717d1170a04778aa899809a817dde2 : SUCCESS in 1h 32m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/98297dd924ee4096a476d6bdf588af42 : SUCCESS in 1h 47m 46s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/94290358a6e84ef5954c73f76d05f50a : SUCCESS in 1h 33m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a4d95c4b66cc43c3b01e5a99b5554eb7 : SUCCESS in 1h 06m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6b66d61f9744f72ed4bf2dd42aa15c24f4c5d17","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-22 20:41:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02161793c61e254decb1ccae2883b6b3b6d48abc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-22 21:13:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b6b7c60e2dd647d7ae8fcf3c76b331e9 : SUCCESS in 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"66ce78c26535899f1fa522a3763bce8c70ecbf27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 22:28:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/a867861cf5d24f45b89e3f92076c8cb4 : FAILURE in 50m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a85d6c304c94c36bd1256ea1f78606a : SUCCESS in 54m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a62383a4ffc41969b8a9fb63f96fdec : SUCCESS in 1h 46m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/de9a6dc08f984ff690d43b55e97b5837 : SUCCESS in 51m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e4ae175136284aa882f6ffaa34839d809d652e23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 01:32:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/629474035a1d47edb5a6388415383bfa : SUCCESS in 29m 11s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/949b9fa91f174d8d9c5c3209b77e1efe : SUCCESS in 29m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b8b75d1ef1de4be89ef67b5aa5411960 : SUCCESS in 1h 16m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/96e4b66cb98443c5992f284ebfa515a3 : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d50563be997d43068124cc1c83f9e8fe : SUCCESS in 1h 21m 40s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/126b1a8524c2423ebe36770ca3c11292 : SUCCESS in 51m 00s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8b0b73c4f1fd4e94b4d65c12d3f76930 : SUCCESS in 1h 51m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/13ac90a5dae544068faa9970ef96f232 : SUCCESS in 2h 23m 06s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/3c9701346fab4cda86025950795d93ff : SUCCESS in 1h 49m 21s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/634b4d7e955c460390a25fe71ac47398 : SUCCESS in 1h 50m 40s (non-voting)\n- kolla-ansible-centos8-binary-ironic https://zuul.opendev.org/t/openstack/build/de82b0ad337240968616e48e9abd189e : SUCCESS in 1h 18m 12s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/e353ed6031fa43c7ab6d2a14fee70827 : SUCCESS in 1h 18m 14s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/b9b5049a2fab41f59e7d8e3f5ca9f333 : SUCCESS in 1h 50m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4184aa02e04a459db8ef4fbfb4fa94cf : SUCCESS in 19m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/473a2810292a4a3cb1250c468e0d4dd8 : SUCCESS in 18m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4d9697e98b6241a5ad8d13ba0efbe83d : SUCCESS in 59m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c33f577435f47e9a21952ae0d7f5be0 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e477571579e041c9882901938e290a9a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ebe80d8a0ae43a89017f9de5a509a00 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87d49d20a4af4aaa97ee81b484244085 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/601abb59774c4f2ea8be5f0fe6877761 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af69b31560eb406a91265b261a2336c9 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/353649f7fa8a476790fb3530f9006cab : SUCCESS in 32m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/246b521fe0314d0fa2d65463c74300df : SUCCESS in 28m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/193d8d1f38844791938b749f5d174e76 : SUCCESS in 1h 06m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/72b64a9666fd4ea8ad0d98428ad64f1f : SUCCESS in 59m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/70bf6af0b99e46aeb7d419fce584e2cb : SUCCESS in 1h 10m 18s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/9ddd99e832ba4fba939989a505bdd230 : SUCCESS in 1h 29m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/5e9821e50a8b452bb8b7b68434d8c637 : SUCCESS in 1h 51m 34s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f0f2ca4fdc4443abb2fee7d2ca115918 : SUCCESS in 1h 36m 33s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5bdf88255aee492a87bb3ee98bcce903 : SUCCESS in 1h 06m 00s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/c218c8ada7ef4204a29d8d276b8dd066 : SUCCESS in 1h 45m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"438f6072b0584109634f2ab058b42699deaff849","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 20:07:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0052e20b0df8dfc8c91fab5d0427ac55c8c30b18","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-24 07:15:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e9a51337d3b3eb5cea8838c2cd2711f86f85027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-27 18:28:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e7c562e8cc31c96753f9ea789cc7acbbf51150ca","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-27 18:28:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"67ec8f4531819dc78c647518895f4ab49241e9f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 19:45:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/95ed9f2639f341ecb8d6e155f0b60126 : SUCCESS in 37m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b5c50cdf08a64d4c8628944b8457b1af : SUCCESS in 48m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/26067553ad4e4b339eaf81f4a920b88c : SUCCESS in 1h 10m 58s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e15aed9f42ec470aa8e15eb3777839e0 : SUCCESS in 57m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/78522aaa475443e1a9c94486beb965b0 : SUCCESS in 1h 14m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d3780755ea2b40fab089f15a4fd6f5da : SUCCESS in 27m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1a49ec3b3bc0429ea4a6e90b5542b0cb : SUCCESS in 20m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1f02b69410664c2d9722334a5641fd15 : SUCCESS in 1h 05m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6b6c33562d84e54952aa3d4b761f962 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05e97712a7044ec18ec3ad55b7ac146c : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0ee5449d10c405cb40901bf1d36fed5 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/261c5572320f483f8e4bab9ce187d5ef : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01dc82b0640b4cdd89340d2f51e37dc4 : SUCCESS in 6m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f7c1d3c679e414ca5882a8864220010 : SUCCESS in 18m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0644ddd18f4c49ff9842e40b43450177 : SUCCESS in 24m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/955b1edce4a64f8092ecbb5fe1b7816f : SUCCESS in 51m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1e275db5b9a46cb99188403e3dfd5e3 : SUCCESS in 47m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e3487d1522f9406b846dc5457656f6d2 : SUCCESS in 1h 06m 59s","accounts_in_message":[],"_revision_number":3},{"id":"d0ee0a739e03204107ecf81f3b44e5261b2b5f32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 20:08:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d8978d6c6e949218b31f7451d236b28 : SUCCESS in 1h 25m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03ef544690304385b07a4e2792d44d7b : SUCCESS in 1h 08m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/655111f75cac45d2818c714757d91266 : SUCCESS in 1h 36m 51s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/559a185d1bfc4ae195d7896b2c5e0e8c : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f8d5175cf904a7197bd3d280880aa8e : SUCCESS in 6m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/567c3414eb954e5594a76533ff3d2a84 : SUCCESS in 53m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"37e3bdc8d9f9971a07e99543195513cc6a305652","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-04 16:04:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bdae271b2dd3b0252d53fe16c19e6b5f5be6ea40","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-04 19:17:21.000000000","message":"Patch Set 3: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"41bd484cc03ceceb1901f7a020c639fe47a933f5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-29 18:14:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bc2bf085985e1b582a2d39b854c310e974438342","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-29 18:14:22.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"1a65b5aac9b0c5121171609c77ce31714f5d4dc1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-29 18:14:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"93a463d2c144d0186fcbd6e758578a9eecc8f6c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 19:22:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42bea3a06dca4259a0f5706bc797b879\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/94ad96e6147b41589b15531e5b7174ba : SUCCESS in 17m 18s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5245f5a3f1cb4fcab26247b2da71259a : SUCCESS in 52m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d7a7a1f8a6554a38997404a4a2079ec7 : SUCCESS in 1h 03m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/168341621a024efd8a43e063570a0a45 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66208697e3af40dfbbde593e3104d9ac : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74313ff63299493b9e491e0f55084a1e : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f214dc11469644acaa205fde3a2f6c0c : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31784de686274f26b6ba8e3498682331 : SUCCESS in 6m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/204e61737131410ca1ad998e6e145c14 : SUCCESS in 28m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6465045579424a159f9478b5ab245347 : SUCCESS in 57m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f4d477f91bd5429aac4dd237d8113307 : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":4},{"id":"2099eea02380e5852ae52cbaeece8ef69cff071c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 19:55:07.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81c35b006cb24f9a81c62ecb21053326\n\n- kolla-build-no-infra-wheels-centos8s https://zuul.opendev.org/t/openstack/build/d469d30cc57e4e15ae52f59e35fb248a : FAILURE in 15m 34s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/352ede03479d47708e1bb9f9096614f5 : FAILURE in 13m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4b21222fce424ade965339397c4f9eaf : FAILURE in 21m 26s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/bb885d30ccf749dbb190e553e72e7e5c : FAILURE in 13m 25s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/7f2d36425f7e4ae088693a8cc1fa8287 : FAILURE in 12m 45s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/38b6d6774de7444892c2938487bf2d39 : FAILURE in 12m 07s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/5374ed1158714b80a733a6877de75aba : FAILURE in 13m 34s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/48da98c503d645648cb057c2811d8290 : FAILURE in 13m 45s (non-voting)\n- kolla-ansible-centos8s-source-octavia https://zuul.opendev.org/t/openstack/build/2a24973dd96048f5bec92848a308e0a1 : FAILURE in 11m 59s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/85bdbca30b6c4e549c8501211b5cce06 : FAILURE in 13m 37s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/0c254611946f47f48f3f02aaa5da44f1 : FAILURE in 11m 44s (non-voting)\n- kolla-ansible-centos8s-source-kvm https://zuul.opendev.org/t/openstack/build/61f1c9e438564479b15d538b606268f2 : FAILURE in 16m 18s (non-voting)\n- kolla-build-no-infra-wheels-debian https://zuul.opendev.org/t/openstack/build/ff14570999b44c879496c546d790d7dc : SUCCESS in 22m 34s\n- kolla-ansible-debian-source-ironic https://zuul.opendev.org/t/openstack/build/a36351a6b8434ee6910bde5accd358e3 : SUCCESS in 1h 09m 08s (non-voting)\n- kolla-build-no-infra-wheels-ubuntu https://zuul.opendev.org/t/openstack/build/4e6a6f49ca50456aad33ed38c1ea345b : SUCCESS in 30m 21s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/e1d127e7d2074487b60ff1c1737d6232 : SUCCESS in 1h 23m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6c43e68337e14aa68483c10795b4b628 : FAILURE in 1h 38m 57s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2b9a31877dbe4645b1d6428b75794a92 : SUCCESS in 1h 26m 31s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/900b74324de04705837ec35dd92f17e1 : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/30afa8cb848c48729e54dc726f6458b3 : FAILURE in 21m 33s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/33f5a90a175b439d8cbe144d0d58f022 : SUCCESS in 51m 48s (non-voting)\n- kolla-ansible-ubuntu-source-octavia https://zuul.opendev.org/t/openstack/build/5a6f1e0041d94c1caf427b4f592e6cf8 : SUCCESS in 59m 13s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/83ae56152a794410b4d8878e12c4c212 : SUCCESS in 1h 11m 41s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/c9e6f2285a43426c92cdba0e5c91cbbb : FAILURE in 22m 32s (non-voting)\n- kolla-ansible-ubuntu-source-kvm https://zuul.opendev.org/t/openstack/build/d0e77f6dcc6244ce8d2625bea1305ab8 : SUCCESS in 43m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5d0eb1529f1e9edc2c3369cc77e4d399d5b50487","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 21:56:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af84273851c54ea9853d642dd7c8841f\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1072df76539f4204972bd4bf63e14578 : SUCCESS in 2h 42m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ff2b162cb8c45f4abaddf634ae1da6d : SUCCESS in 2h 41m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28ad283604e64ae3a4ff0247a640fd1a : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d96000cbc434fafb4b1cfebba714c1d : SUCCESS in 7m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8cbe04b6a4b44a7289191d5490b014bf : SUCCESS in 1h 54m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0c49e811268dc0b1a7cd27a9dd9302d378cb2800","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-10 07:07:38.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"7601be0576023a52f0b86f07eee3e75de2aa23dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-10 07:15:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f4761069cda587e6025f4648ab312ea03000e092","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-10 07:32:11.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fe95d2a736c590b391476d9086237b6ee9f8fa67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-10 07:43:38.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"a3d5fc7b5a877ca2da367081ba0cc07387c7b81b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-10 07:43:50.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"71d4fe2cfb7eb0287b307e19cafc96299a9bd2a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 08:59:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68cfd718a45a46ffa0bc4b2d75abd78a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9ed1ace74e98490185abe9ee406f98e3 : SUCCESS in 37m 21s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/eba10fb69d424583a24b96bedd737ea9 : SUCCESS in 1h 13m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a37d1c0580e9430eb09de9ba2f1bc195 : SUCCESS in 17m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/264f786c170d45b3833a26bfdc22bfb6 : SUCCESS in 56m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/da1ce48a746d4157bd16909a7a5d22c5 : SUCCESS in 1h 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5860e3de18154533a8c45b43df826659 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a57094df62ab473bb57cb8e95b4e69d4 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55e0825f894847c292f57207d6e547e9 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f49aed3c80ad44b0a093c5a53e78d76e : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23bf066de60046ceaf3530c7a9164364 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8673e81cb554f0fa3a2ca04904540f1 : SUCCESS in 3m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cbb5ede4f1c84b0aa69ef754b1ba3cb7 : SUCCESS in 19m 07s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/ca8a68be08d34c6899d48953198eb9eb : SUCCESS in 1h 07m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ba0894d6386e48fbb0874ac0a9ffdd53 : SUCCESS in 17m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/749ab0cd02c44ca5893ad7e6e844bbea : SUCCESS in 49m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a1cb9c5f4eda4cbe8f7c3bdfd069cad1 : SUCCESS in 46m 15s","accounts_in_message":[],"_revision_number":6},{"id":"d7127f011a11766954523f3fc678bcef093bd0f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 09:28:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b875654594d47c391f7254f07c9230f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/81cce344b8a14e46906c44e8c9819ccd : SUCCESS in 1h 41m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b78ccbd871214324b482d2fed8c4edc1 : SUCCESS in 1h 36m 15s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5898d385706d4c5a8b6284335e8936ab : SUCCESS in 1h 14m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/685b5571df8b459abc8383ccec9925e9 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac13233d57744128a2cf5c78242236c0 : SUCCESS in 3m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4fd7b792117b4b8bb4bdc2cfb0d9b318 : SUCCESS in 28m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7bc969592c564ded9a89b2e404532290 : SUCCESS in 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"717e7758b95c0be63c2685fded8331b2912e448c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-10 09:41:40.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"c4eb1aba73641a97117692ca25bc8e1550475397","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-10-10 14:52:15.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"51346a65766cb2b6d2bec504e0edc5f350dd0da3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-10 15:06:05.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"bc659336074817262ce13e7ecbba8c652200fe32","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-10 15:06:27.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f6d0ed7d6b0f9bb9522ac01316e81afe7190a8d8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-10 15:35:59.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"143f2cf58e2e06f1da02b455298b2496d234e8b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 16:29:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37afa5b163bc4a1f833707042a2ff085\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1201edaac45143968fbb1c3dc9c2d0a0 : SUCCESS in 26m 02s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/875435af3e8046ad9b11a20653f2ac77 : SUCCESS in 1h 07m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/862b9eb1c90c41f9b4385ebb11afa5c3 : SUCCESS in 21m 31s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9d3cc608a5d45c3b2c114398190b91a : SUCCESS in 1h 07m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/464c1442a3a14c24a37eaae74945d937 : SUCCESS in 1h 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a63f4f647d4a21a1f39fc7a56596a7 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ae163c09ed64069877f4714391f0876 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2be6cec1a90c4909b8c917e5998df0ad : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b03287bb2ba485eafbad2597fffeeab : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77cc0811885a47fd8f758f6878d6853b : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e285dea0d6ca4c4f8f633c84ab7480bb : SUCCESS in 3m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ed5d1dde5ef94a5d9138ba03c4724103 : SUCCESS in 22m 38s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/f6948771c1c74ae7aa07fd655cb5c73b : SUCCESS in 1h 05m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4c487b5748944327bcd8afd361a0f15a : SUCCESS in 18m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c0c3e67ceaa6403d956dcf564e7bf1f6 : SUCCESS in 56m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/98bd6a8f3589483484a080578428224c : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":7},{"id":"3266173338913eaecd88fa163df80e3a3c5a76a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 16:43:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0f41580428847799f02968f461bcd76\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5bc7cdd9dab046c78f05c9e138a810e4 : SUCCESS in 39m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec6b7306b9d64bbebb7b2a04c5c6ab2f : SUCCESS in 38m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3fab11b34c404966b01bb3687a7b9ba9 : SUCCESS in 1h 28m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf66e5b819244edcbb875bcb00044b10 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0b1028f474241aaad7969a0a5f59201 : SUCCESS in 3m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0396fb52a6f14d7a978f92f78583c134 : SUCCESS in 38m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/36b4063e76b74ba9803da17c07e81fd1 : SUCCESS in 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"073fbf66a430a4f1606aa486b9dfc99995c5dab5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-10-11 07:27:23.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"3111db8322ff4384296b7189669a0248327fb60d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 07:27:47.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"34d3029468f09f9fae24e5fd3c606867afe95aff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 08:42:20.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c47ec033814dfdbebae3704f717310\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c993936986554f5fb379bfbc58650f6d : SUCCESS in 19m 41s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/3945ebf5f1504825903c3844c604e8cb : SUCCESS in 1h 04m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/70ecfcb8478948d3aa25fadf2028620f : SUCCESS in 19m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b735dd4740574fdeaeef7ebf2a3475d8 : SUCCESS in 59m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8b71f1c90d7842efaa54cc4821a68ffe : SUCCESS in 1h 06m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091c5acc21cc463a8fd04e42383b8698 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b81feb8a54c841198d8cee5f4897e3f3 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a40982953ec4e0bbdfe5f24eaea07e7 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eae0063156e4dbcb266518241139bdb : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ac0cbad2e99464b81bb70d9962f8256 : SUCCESS in 4m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/750dd9e1b36244e6a5b8eefb3fc5c91f : SUCCESS in 20m 55s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/3a7844c34e6d4e01be43a76de2318412 : SUCCESS in 1h 13m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/db204558a4db4db3ad76074dad4282fc : SUCCESS in 26m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3b8b8383eb443389bb66e6e7359db24 : SUCCESS in 49m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fa0b4ff1986645009542b098e7f5f3e7 : SUCCESS in 1h 08m 49s","accounts_in_message":[],"_revision_number":7},{"id":"55d0f199d0bba55936e46f7a3fd8162e86312186","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 08:42:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"d3595a4530777a464671b54063f053f95e1c3f57","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 08:43:33.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91a47673a9fc42008e81f8fa52eb0f64\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79de65f143814fba810aa764a9600cd9 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0ed406bd54b405da6f388897f61571f : SUCCESS in 50s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2daf4331a648cc2df6982c1a6ec47a705e038255","revisions":{"832ec23a9265551d800ea99381670c073bdcd9e3":{"kind":"REWORK","_number":1,"created":"2020-04-22 17:49:02.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/722102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/722102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/722102/1"}}},"commit":{"parents":[{"commit":"d318e5c1a193914ee90eab1c9a762d61d1b07acb","subject":"Merge \"Zun: add CNI loopback binary to the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d318e5c1a193914ee90eab1c9a762d61d1b07acb"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-22 17:48:44.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-22 17:48:44.000000000","tz":120},"subject":"[DNM] test something","message":"[DNM] test something\n\nChange-Id: I92c81c77e6a16570a108cde8031f7977930fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/832ec23a9265551d800ea99381670c073bdcd9e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/832ec23a9265551d800ea99381670c073bdcd9e3"}]},"branch":"refs/heads/master"},"969f6f764beec3315279904efe8a2e9b607d4178":{"kind":"REWORK","_number":2,"created":"2020-04-22 20:41:51.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/722102/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/722102/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/722102/2"}}},"commit":{"parents":[{"commit":"d318e5c1a193914ee90eab1c9a762d61d1b07acb","subject":"Merge \"Zun: add CNI loopback binary to the image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d318e5c1a193914ee90eab1c9a762d61d1b07acb"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-22 17:48:44.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-22 19:51:42.000000000","tz":120},"subject":"[DNM] test something","message":"[DNM] test something\n\nChange-Id: I92c81c77e6a16570a108cde8031f7977930fb02a\nDepends-On: https://review.opendev.org/722127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/969f6f764beec3315279904efe8a2e9b607d4178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/969f6f764beec3315279904efe8a2e9b607d4178"}]},"branch":"refs/heads/master"},"9ba343a3423d8652fa05006f34d5b58e1bd745e8":{"kind":"REWORK","_number":3,"created":"2021-10-27 18:28:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/722102/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/722102/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/722102/3"}}},"commit":{"parents":[{"commit":"1766ce8314bb263c7671eb61eeb34bc5f554d643","subject":"Merge \"cinder-volume/ubuntu: add lsscsi and nvme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1766ce8314bb263c7671eb61eeb34bc5f554d643"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-22 17:48:44.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-27 18:28:16.000000000","tz":0},"subject":"[DNM] test something","message":"[DNM] test something\n\nChange-Id: I92c81c77e6a16570a108cde8031f7977930fb02a\nDepends-On: https://review.opendev.org/722127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ba343a3423d8652fa05006f34d5b58e1bd745e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ba343a3423d8652fa05006f34d5b58e1bd745e8"}]},"branch":"refs/heads/master"},"4415116361690855556bab0d4f72c29421563066":{"kind":"REWORK","_number":4,"created":"2022-08-29 18:14:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/722102/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/722102/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/722102/4"}}},"commit":{"parents":[{"commit":"91c2cc0d8d4caf8fa6567b95d8f7e7b4fc4dd443","subject":"Merge \"[docs] Fix OVS link\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91c2cc0d8d4caf8fa6567b95d8f7e7b4fc4dd443"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-29 18:13:34.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-29 18:13:34.000000000","tz":0},"subject":"[WIP] Fix","message":"[WIP] Fix\n\nChange-Id: I92c81c77e6a16570a108cde8031f7977930fb02a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4415116361690855556bab0d4f72c29421563066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4415116361690855556bab0d4f72c29421563066"}]},"branch":"refs/heads/master"},"94b1418979a269abe537d9562713825b7800b0d9":{"kind":"REWORK","_number":5,"created":"2022-10-10 07:15:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/722102/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/722102/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/722102/5"}}},"commit":{"parents":[{"commit":"b8a352647d57d35463e551a0d360c19fb2ad34c6","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8a352647d57d35463e551a0d360c19fb2ad34c6"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-29 18:13:34.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-10-10 07:14:27.000000000","tz":0},"subject":"Fix writable rootwrap/privsep config","message":"Fix writable rootwrap/privsep config\n\nMore details in the attached reno.\n\nChange-Id: I92c81c77e6a16570a108cde8031f7977930fb02a\nCloses-Bug: #1874298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94b1418979a269abe537d9562713825b7800b0d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94b1418979a269abe537d9562713825b7800b0d9"}]},"branch":"refs/heads/master"},"aab9eb91aa1f01b4ef6c1d53217a6f071dd91a53":{"kind":"REWORK","_number":6,"created":"2022-10-10 07:43:38.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/722102/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/722102/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/722102/6"}}},"commit":{"parents":[{"commit":"b8a352647d57d35463e551a0d360c19fb2ad34c6","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8a352647d57d35463e551a0d360c19fb2ad34c6"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-29 18:13:34.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-10-10 07:43:38.000000000","tz":0},"subject":"Fix writable rootwrap/privsep config","message":"Fix writable rootwrap/privsep config\n\nMore details in the attached reno.\n\nChange-Id: I92c81c77e6a16570a108cde8031f7977930fb02a\nCloses-Bug: #1874298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aab9eb91aa1f01b4ef6c1d53217a6f071dd91a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aab9eb91aa1f01b4ef6c1d53217a6f071dd91a53"}]},"branch":"refs/heads/master"},"2daf4331a648cc2df6982c1a6ec47a705e038255":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2022-10-10 15:06:05.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/02/722102/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/722102/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/722102/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/722102/7"}}},"commit":{"parents":[{"commit":"b8a352647d57d35463e551a0d360c19fb2ad34c6","subject":"Fix Swift deployment issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8a352647d57d35463e551a0d360c19fb2ad34c6"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-29 18:13:34.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-10-10 15:06:05.000000000","tz":0},"subject":"Fix writable rootwrap/privsep config","message":"Fix writable rootwrap/privsep config\n\nFixes a hypothetical security issue related to privilege escalation via\nrootwrap/privsep. A potential vulnerable service could previously allow\nwrites to its rootwrap/privsep config and thus allow for more commands\nto be run with root privileges via rootwrap/privsep. For a succesful\nattack, this would also require the service to allow to run arbitrary\ncommands via rootwrap/privsep. Thus far, no such vulnerabilities have\nbeen reported and thus this fix is simply strengthening the container\nimages against such an issue in the future.\n\nChange-Id: I92c81c77e6a16570a108cde8031f7977930fb02a\nCloses-Bug: #1874298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2daf4331a648cc2df6982c1a6ec47a705e038255"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
