)]}'
{"id":"openstack%2Fkolla~722417","triplet_id":"openstack%2Fkolla~stable%2Fstein~I3aac1e6a7f7c5f839b036f15713af949fba206cc","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1868574","hashtags":[],"change_id":"I3aac1e6a7f7c5f839b036f15713af949fba206cc","subject":"Add nfs dependencies to glance container","status":"MERGED","created":"2020-04-23 18:23:03.000000000","updated":"2020-04-23 21:03:50.000000000","submitted":"2020-04-23 21:00:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722417-1587675622544-795c03e7","meta_rev_id":"4a94b14e037000615bb5228abd3e94bdb29aee23","_number":722417,"virtual_id_number":722417,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 21:00:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-23 19:24:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-23 18:39:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-23 19:54:08.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-23 18:23:03.000000000","_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 18:23:03.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2020-04-23 19:54:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-23 21:00:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3afe00e88e8b42e6c58cae0ba88a49c641d451fd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-23 18:23:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f413c67013c435cfcdbe428d9f3b0dfb6f7a4333","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-23 18:39:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45670cb1ce5198b55a17cc9f204ab4695928dae7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 19:24:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9b0e7c1956888e5dc010348f6656a242bc94ec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 19:52:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0146da28d49d4e9085fc3694ede2a0e3 : SUCCESS in 51m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5d727125662e4555a3f92c2fbfb99f40 : SUCCESS in 42m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b487211590f847e280db445d7b104598 : SUCCESS in 1h 14m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/106b78e6deed47179b0602ebf92f38d1 : SUCCESS in 1h 04m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fb59a3321ca64b4787c231d1c990efdb : SUCCESS in 1h 27m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/abb8c9136a324e3486364a379cdf4da3 : SUCCESS in 59m 32s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/00822cff14a74feab437c75ba1c7a397 : FAILURE in 43m 22s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/48d9d51c44874115aba6dfdfc878f2f4 : SUCCESS in 46m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c01fea2a278d4ceeb6fe6f59e16dd602 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5bd64605d494530b8d81f9ad310c702 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a02154a9b3a9442caa497b3bbe1ef38d : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c609f16eabcf44aaa2be1f9d8312dc02 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e884ddfd922b4352870703fbc6ee222f : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b2057f5ee3c4c37b91b7a9565e841a0 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2707871ccb8845149a16961d0bbba603 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/19a9d54c2d3540889666dcc789348356 : SUCCESS in 39m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/22cc0b22911849ebb3809719c6179740 : SUCCESS in 26m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/720227c0133749d4b5604c89ad2de628 : SUCCESS in 59m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0ccfe8b6d1d24acfa2ed991d3318c95e : FAILURE in 33m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/329e79ca3d5c47fe8015e21530f039c2 : SUCCESS in 1h 04m 45s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"f18d6e6ba120d5d2965aadf3ba8c8ff2318da9d7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 19:54:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3706478e2bf224a7381ccb727da05a5d7c1439ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 19:54:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c418a763e7a126eba90c1438875771e28ded63b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:00:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9d30a00bcf9380c2823a13affdcc06f3ecbc118f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:00:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a923cbbdbb374655aa5cbe2b9937ed52 : SUCCESS in 46m 29s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/170c528218ee4c7487a9a9ab6e01f15e : SUCCESS in 1h 00m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28a9847935c34797ba8e6ae4a2c6630b : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f556d9170aa4b1a9db765a647eb882d : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ed7dce6dd1b451ab8399c63a8ff2850 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59748339db8e437ca294b996345fefff : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b5e6f0d7ad24a51a43e47db22673644 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9090b6d49ab641db90144dbb0009b6d3 : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e26dceae90f43db98cf7ff3bdc58f23 : SUCCESS in 9m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b71f0c5521240eeb569a5e41be5f600 : SUCCESS in 38m 11s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"f5ae274a06266d993342727c599826635f1d7144","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:03:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3445376524264a9d85a17b1251524424 : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7019830aef364cacb8937bd2d8aecf16 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e22fb9796e0592d288310b23bd7245885f235cef","revisions":{"e22fb9796e0592d288310b23bd7245885f235cef":{"kind":"REWORK","_number":1,"created":"2020-04-23 18:23:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/17/722417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/722417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/722417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/722417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/722417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/722417/1"}}},"commit":{"parents":[{"commit":"95e421483e0ff7d7112b7ab981a32e24bf287c6f","subject":"Merge \"Add vitrage-persistor image\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/95e421483e0ff7d7112b7ab981a32e24bf287c6f"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2020-03-23 07:33:03.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-23 18:22:52.000000000","tz":120},"subject":"Add nfs dependencies to glance container","message":"Add nfs dependencies to glance container\n\nGlance API image lacked nfs programs/utils to be able to mount\nNFS exports.\nThis affected Glance usage with Cinder backend when NFS was used\nas Cinder backend.\n\nCloses-Bug: #1868574\n\nChange-Id: I3aac1e6a7f7c5f839b036f15713af949fba206cc\n(cherry picked from commit cac9d46aa6dbac6631bdcae67db125a88e1c3234)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e22fb9796e0592d288310b23bd7245885f235cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e22fb9796e0592d288310b23bd7245885f235cef"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
