)]}'
{"id":"openstack%2Fkolla~723238","triplet_id":"openstack%2Fkolla~stable%2Fstein~I5fb0fc4e6a8bf7ab07abc9c78f0673e3e764b4d1","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1872283","hashtags":[],"change_id":"I5fb0fc4e6a8bf7ab07abc9c78f0673e3e764b4d1","subject":"[rally] Fixes Ubuntu binary","status":"MERGED","created":"2020-04-26 15:13:21.000000000","updated":"2020-04-27 11:18:27.000000000","submitted":"2020-04-27 11:16:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723238-1587986193726-602cea05","meta_rev_id":"2359bc2a06e049a73408040f8b7b0d274d79e596","_number":723238,"virtual_id_number":723238,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 11:16:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-27 10:20:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-26 18:19:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-27 10:20:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 18:19:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-27 10:20:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-27 11:16:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16c257f34d19a94fb2ef5aae9a5dd77ee0ed54af","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-26 15:13:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fca423bd452812005c187b99e73dbebc3a268644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 18:10:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b607f9594b6b40819db76c2881a7f31a : SUCCESS in 1h 11m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/594814318d3c4733be457d3a73ae3576 : SUCCESS in 2h 55m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5462398eb8604445a49b0056488a8c3b : SUCCESS in 1h 35m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/090dfa8b0aa74277945d1ceb7c450937 : TIMED_OUT in 2h 01m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d92e33c7538f4b9cba3bcf816b073e2f : SUCCESS in 1h 36m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fab3fd888475439bba102c304848d3ce : TIMED_OUT in 2h 04m 23s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ad8094a7675549c0a953f4bf2410d16b : FAILURE in 1h 03m 52s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/ec20907c350a455cb44596c24321074e : SUCCESS in 2h 50m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b63580194b1491fac7d34137fd85adc : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d95e5e8aa98f4500958dc26a458056d5 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/01c9356fa23349f081d2f31116e19a7a : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63370a31cad54ef8b90bbb60f13095cd : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9da9d9b1a9844e9fb39bd0b3e1c35012 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea7e3a921ec4430b8b4c7a2799d5c1e : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85cbbb6ae4b14fbf90a45a88096591f3 : SUCCESS in 5m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fd703d2335e04d04b7eea02025154889 : SUCCESS in 2h 41m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/71506ad840194ce9b05ffca05e72bffc : SUCCESS in 54m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f028dc0d3fb46eabaf413b7beb53d10 : TIMED_OUT in 2h 02m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c17727fe6b742f28738670f94152f3b : FAILURE in 57m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/de328653b85842e797932bb2f00f964b : SUCCESS in 2h 29m 05s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"43616d1482082f49d6879fd004fc9d11049bb6ca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-26 18:19:00.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\nlooks like ubuntu has \"biggur problooms\"","accounts_in_message":[],"_revision_number":1},{"id":"a7d9d2a72ff728966a4ba546827f0699ad22bb79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 20:21:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4f6ae87d8d0646a395c03d2c5592cad0 : SUCCESS in 46m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1f1e1682adb845f79180312c1b39f3d2 : SUCCESS in 46m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ee8685badb1f4d12b32e502a4bff9eb7 : SUCCESS in 1h 25m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8d014dd348cd4c57bab825588ddc9033 : SUCCESS in 1h 03m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/359884e8c11f4deab5c57e5fa89c7ac2 : SUCCESS in 1h 25m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0466c40e9a724432a4f15853381c04d4 : SUCCESS in 51m 01s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/2f70376ccd3b461b89abdcac211dc3f8 : FAILURE in 47m 48s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/cf2b8a738fc54957aef0a7a6cbc2ee4a : SUCCESS in 1h 11m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0de378f2d87340a58939301146fccb8a : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572f7b04b5ed45898d9a3ed8de9b9c05 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/528121b42571470985fd41d6f0586525 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90c55f0f0b9045bdafdf2f830854a763 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88ae9cb123e24f57b2d771f37bff2900 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec2a4f659fe4c759b33124ca71e0c8a : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7949dca310004ad6a92d5e4a9f94256d : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d264544c30744c07a910cb72aed43d8a : SUCCESS in 25m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bdcc247da0984599908ec45f1f47f72e : SUCCESS in 25m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/319c1997926a4f66853dc496103fe606 : SUCCESS in 59m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/822fc4c22ce742238fcc92fd6c3fd930 : TIMED_OUT in 2h 01m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1524ff247d014db597291d003d06b3cc : SUCCESS in 1h 02m 44s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"d5486cb7a10018250f2ca14496d84eee6c9c4db6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 10:20:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46d7109ee65c31f607eece2a24979286bd63c5bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 10:20:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"071c7519e952eb0513d5f0de431301b52093df2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:16:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"92c4d5285f3d9b8a0f0ca2d651b70924f6190199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:16:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/07aa531554ce406db822a401e5ddc6aa : SUCCESS in 49m 15s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/915cb7e69e4a4a838229b1d2c430cd1f : SUCCESS in 51m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28fc380b5220463daafe590dc16f7531 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b27aad140ef4f29a3b1e6e2799a7a9b : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78347ddddfe548d5b83680d8948a70ae : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d328dcd8bff04f34a7775a0da8daef04 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b75a986bd4034bcf97816d1f8f6d5035 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0839eda780042a18f097c89fdeb856c : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd233f1dd8d6465cb8eabb7a24a8b067 : SUCCESS in 4m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c8a2bf226d547ecb8926d9ac638a379 : SUCCESS in 24m 44s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"7dcde9f5ecf606199ec0af549cd0efdecb6e9e98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:18:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec518ed941f455499a878078f84de2a : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74b83058e58540d3b9f7fb0a20c49c76 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebfc0a7bf8da8ea27e5a7527859c201dacf7c5e5","revisions":{"ebfc0a7bf8da8ea27e5a7527859c201dacf7c5e5":{"kind":"REWORK","_number":1,"created":"2020-04-26 15:13:21.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/38/723238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/723238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/723238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/723238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/723238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/723238/1"}}},"commit":{"parents":[{"commit":"e22fb9796e0592d288310b23bd7245885f235cef","subject":"Add nfs dependencies to glance container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e22fb9796e0592d288310b23bd7245885f235cef"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-04-26 14:45:02.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-04-26 15:09:35.000000000","tz":180},"subject":"[rally] Fixes Ubuntu binary","message":"[rally] Fixes Ubuntu binary\n\nThis is a follow up for [1].\n\n\"rally-manage db\" command was deprecated since 0.10.0 but Ubuntu ships\n0.9.1 for Bionic.\n\n[1] https://bugs.launchpad.net/kolla/+bug/1856693\n\nCloses-Bug: #1872283\n\nChange-Id: I5fb0fc4e6a8bf7ab07abc9c78f0673e3e764b4d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebfc0a7bf8da8ea27e5a7527859c201dacf7c5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebfc0a7bf8da8ea27e5a7527859c201dacf7c5e5"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
