)]}'
{"id":"openstack%2Fkolla~723254","triplet_id":"openstack%2Fkolla~master~I051081e547074bbe13b55143d73baef5651e18fb","project":"openstack/kolla","branch":"master","topic":"prometheus_exporter_updates","hashtags":[],"change_id":"I051081e547074bbe13b55143d73baef5651e18fb","subject":"[prometheus] Bump versions","status":"MERGED","created":"2020-04-26 19:50:12.000000000","updated":"2020-04-27 11:16:26.000000000","submitted":"2020-04-27 11:14:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723254-1587986058633-276f548f","meta_rev_id":"d12c304f680055cb1e3589bb371a3ee11ce9d18f","_number":723254,"virtual_id_number":723254,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-04-27 07:18:12.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 11:14:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-26 20:21:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-27 09:52:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-27 06:56:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-27 09:52:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 20:21:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-27 06:56:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-27 09:52:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-27 11:14:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5e04ff86e540d9972bfa1ab40191ecc5a69c206","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-26 19:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fd971f9eb06d9118a4a68d7549ced3282e1f19c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-26 20:21:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/fcd7f2d74aed48fe9f281818c2cfe993 : SUCCESS in 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1062c4701041dd8ec89cec8cb3d2ff28c14b37c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 20:37:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/450f41dac8e04781aa2cac407ada6aa4 : FAILURE in 46m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/76af97473fc9405a8f785ee1854b8413 : RETRY_LIMIT in 1m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/70685fe31b18415994b921a31d73cbd8 : FAILURE in 3m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/aa4147410f1d4796a1c057698ec728b1 : RETRY_LIMIT in 2m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"730531e94151024f9d8ccecfbd52373dfd464951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 21:14:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/84368c1274b84d56a347cb88638b1674 : SUCCESS in 25m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6370c0636184436ba426cba6aacd5193 : SUCCESS in 28m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/501aa339a0574cb68cfa323247aa8506 : SUCCESS in 1h 07m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2e26df0a34864526b9e28d46747e3355 : SUCCESS in 1h 12m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/edbae06361e746c4abfc11029ced086f : SUCCESS in 1h 19m 28s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/511dc5f3457146518e28856c9d77fdbe : SUCCESS in 40m 58s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0789e90bfa5943158697a05b8e34ad40 : SUCCESS in 22m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/628aa9debc4e43388777ed73d999ac85 : SUCCESS in 19m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8f6f25b7da6b41ec84b65d5ea0d351e4 : SUCCESS in 1h 01m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f65a2585833c430bb893b0b4fb612279 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d77d96c71e14420488e4bcebc7740b1c : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33cb03523b09471b83327e90387f8300 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fbf3c9c9fd94682930437ea5a79fbc7 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f31d0822a7784cff974e19c80b922004 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c363a12e487d4e1cb605abebba1b3018 : SUCCESS in 7m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/50bcd59779554fb1928c68ed1776da95 : SUCCESS in 20m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0240b01803847a8a737785673aa827f : SUCCESS in 24m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e61d6130f154fb286f0de53fd872443 : SUCCESS in 1h 02m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce3456037e2f40a6be675c77c7b91c2b : SUCCESS in 1h 02m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8910f37b0fd6438680b6df72f7a92ae4 : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1fd2a26b007ab70dbdd0c8cb7f40428aab6219fc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-27 06:56:58.000000000","message":"Patch Set 1: Code-Review+2\n\nI trust they work when deployed (could use some testing!).","accounts_in_message":[],"_revision_number":1},{"id":"f552c66fa83f726fa905f6e2178c5e590a95ffb4","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-27 07:18:12.000000000","message":"Patch Set 1:\n\n\u003e I trust they work when deployed (could use some testing!).\n\nI use the latest releases for a while for all exporters except mtail and blackbox. My deployments are Ubuntu binary so I would appreciate more testing :)","accounts_in_message":[],"_revision_number":1},{"id":"fb039b337cdef4340baad29393d1381b4e61d4b7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 09:52:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"871edb172cc30f2a7783969a6d0c9ed082bc9e29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 09:52:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a0685a0c13b3bf50714c17c097e7eecd6af8ea2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:14:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bfb28dc5c567ab6488e020e5f5df46a09de6800e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:14:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d0bdfe67c923407fb1e0daf8b320c90d : SUCCESS in 29m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9d54c8b43d3847488dff5f0d9d048488 : SUCCESS in 1h 13m 32s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d93223889b904fd1b0bca35f57e69ade : SUCCESS in 1h 17m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8302c915dbc042fdb2671bb1c8039233 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9facb08637d459288334469a677dc70 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e2a2e36cc5d4fc98f4e1b55dc824487 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6dcf1eedf79b4658a9a2be3a5c3c9288 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6553a67b74344ebbe46676270944ba1 : SUCCESS in 5m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2fcba5fdd8a6433b8893963993c6660f : SUCCESS in 26m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c781ae89eef54c56a8d7311c7b2d5451 : SUCCESS in 1h 00m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/028b0370abb642fa809854654cecf279 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9c36d89d2436672b75d51ca92c24297eb8abde22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:16:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e18836b01fb4fdd861688a20a74cb8e : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"657e6ce523c4057a77b95ff664cd1b9b626e7cba","revisions":{"657e6ce523c4057a77b95ff664cd1b9b626e7cba":{"kind":"REWORK","_number":1,"created":"2020-04-26 19:50:12.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/54/723254/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/723254/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/723254/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/723254/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/723254/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/723254/1"}}},"commit":{"parents":[{"commit":"63a55048912c2132d00e8b6d0b79e6d0d4835500","subject":"Merge \"Add mariadb_clustercheck image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63a55048912c2132d00e8b6d0b79e6d0d4835500"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-04-26 19:47:46.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-04-26 19:47:46.000000000","tz":180},"subject":"[prometheus] Bump versions","message":"[prometheus] Bump versions\n\nThere are quite new versions for the stuff below.\n\n    - node-exporter\n    - mysqld-exporter\n    - mtail\n    - memcached-exporter\n    - elasticsearch-exporter\n    - blackbox-exporter\n    - alertmanager\n\nChange-Id: I051081e547074bbe13b55143d73baef5651e18fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657e6ce523c4057a77b95ff664cd1b9b626e7cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657e6ce523c4057a77b95ff664cd1b9b626e7cba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
