)]}'
{"id":"openstack%2Fkolla~723887","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I9cab91b82941ea125b16d19e68722ce41a1f11f2","project":"openstack/kolla","branch":"stable/train","topic":"die-py2","hashtags":[],"change_id":"I9cab91b82941ea125b16d19e68722ce41a1f11f2","subject":"drop tripleo-common-tempest-plugin","status":"MERGED","created":"2020-04-28 12:52:40.000000000","updated":"2020-04-28 17:23:35.000000000","submitted":"2020-04-28 17:21:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723887-1588094466197-2ccb4115","meta_rev_id":"4ef5da13f3b954b8708701651a4d4694f1d0cb91","_number":723887,"virtual_id_number":723887,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-28 17:21:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-28 13:33:34.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-28 13:08:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-04-28 13:08:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-04-28 13:00:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 13:08:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-28 13:33:34.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-28 17:21:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afeedabe4997614b698cabd7efb38ce32b11c4ae","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-28 12:52:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6922043a786a8dfdad69873b2991da76b7aeaae7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-28 13:00:53.000000000","message":"Patch Set 1: Review-Priority+2\n\nbreaks all Py2 builds","accounts_in_message":[],"_revision_number":1},{"id":"a367ab38272c5d4b57dde22f8b27ee1df6df7096","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-28 13:08:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nnot crying at all","accounts_in_message":[],"_revision_number":1},{"id":"b34366093ba290ddaf98be691909f747acef0cb5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-28 13:33:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-train https://review.rdoproject.org/zuul/build/3c4829a236444ae7b0a875934a0a3b4f : FAILURE in 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4fb50e13bb686dc660986982569689bae62cc68c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 14:59:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3e38f4daaf51434c9cf8e8bcf99435fc : SUCCESS in 1h 15m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e8f1cfd0293c40009fed6bced72fd4b5 : SUCCESS in 2h 03m 46s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b9b0c15aecd3427486fb7ac95ac54668 : SUCCESS in 1h 26m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/865dd76862574e9bad73324a7909811e : SUCCESS in 41m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bb1a2ac93fd9435cbb874b1e1e59d733 : SUCCESS in 1h 27m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/687e556c5e374cf99cfe24e633fcc52e : SUCCESS in 1h 10m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5ffeb2c1b36f4860bed00cc8d635cbe7 : SUCCESS in 1h 24m 09s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/46f5b2ad0b4d45cbaf5c90615e0e6ee6 : FAILURE in 48m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/86f93dcb235e410f97b8d10092993751 : SUCCESS in 1h 44m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/41d9de4df043428698313c0ffef513a7 : SUCCESS in 57m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/953809735427419ea5f3dd0db819201e : SUCCESS in 29m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/60d2dc7dbb7b413585d9e1d56f12ee2d : SUCCESS in 17m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/03536629336f487ebd9036fd161033e3 : SUCCESS in 58m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa07dab33f0342559602ea3fbfd542dd : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57c6e33aa3547e682250b774c34b8c2 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7472e1c58e9546f5a6c6acd56a4ffb0c : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8374dfec59244cb2a57dab992d8f20d9 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b7442c4f15e4c679cf3301153dd8bb5 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24d84db48584c89b7f72cbef18f0065 : SUCCESS in 8m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d92096c12fac444b80ad197abebc2289 : SUCCESS in 25m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a5860d16a5884d2b8fe7b1afb2c21b3a : SUCCESS in 30m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d4414fd27144f77bd4f6fef40318407 : SUCCESS in 58m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c72ad89615040bb9ccc55b59453e12d : SUCCESS in 54m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e3b3856841d41cc82ccb780258a8683 : SUCCESS in 1h 05m 50s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"4d3c36a859cd0aa2a4f2e94bf10883b7e9179b62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:04:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a386f05317c15425d3607b39b41fa1c528d9b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:21:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ae56883e8efd437f9575db12c472f765 : SUCCESS in 1h 36m 46s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f205479754454b38bf22f25629d69bfc : SUCCESS in 1h 27m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f5b3f000717d4d6bb40af3db51d1338d : SUCCESS in 1h 35m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5be1fd4620cb4f688b32436868e5b000 : SUCCESS in 1h 15m 53s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cf61e4f021094a4c887f4c5f2dfceafb : SUCCESS in 1h 19m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6d9718ff559a411e904f4a11cf8ba753 : SUCCESS in 1h 06m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fb273823edc45c0adadefbbd61257d2 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f1eb5f36b0348c19dd109e395be7bbd : SUCCESS in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46fd98daf8d3487c8d9c6a0230bc9cd8 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13b998f59bbb45589bc69e0646ff8ebb : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18bfc36d60664a3a94440426d3e7551d : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7545b16c309f49a1af17343558ad505f : SUCCESS in 9m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b906eef102b4d3888da5f94055b530d : SUCCESS in 38m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/63345b87caf9462db44c1c322e786a9a : SUCCESS in 1h 09m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e033cb0e2f7048bea1e45efe00ccab76 : SUCCESS in 1h 22m 54s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"4e32ac83b634d86e5ff1b52096bca22bb62b1f92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:21:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c2446aa59d8dea24412404857a7a1454e9cbdcf1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:23:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acaf5d2409764066a384b2a5ffcc9cc6 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dc560d132931401edbda8e81f684512e10e14163","revisions":{"dc560d132931401edbda8e81f684512e10e14163":{"kind":"REWORK","_number":1,"created":"2020-04-28 12:52:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/723887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/723887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/723887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/723887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/723887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/723887/1"}}},"commit":{"parents":[{"commit":"f5a071318fea8d6a87e7d50e94311ed1070c85e2","subject":"Do not try to build skipped/unbuildable images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5a071318fea8d6a87e7d50e94311ed1070c85e2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-28 12:49:35.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-28 12:52:11.000000000","tz":120},"subject":"drop tripleo-common-tempest-plugin","message":"drop tripleo-common-tempest-plugin\n\nThis is not needed plugin without any branches or releases. And wants Python 3\nwhile we have Py2 for CentOS 7.\n\nChange-Id: I9cab91b82941ea125b16d19e68722ce41a1f11f2\n14:41 \u003c hrw\u003e is  tripleo-common-tempest-plugin your stuff?\n14:43 \u003c@EmilienM\u003e yes and it does nothing. I thought we removed that?\n14:43 \u003c@EmilienM\u003e apparently it\u0027s still here https://github.com/openstack/tripleo-common-tempest-plugin\n14:43 \u003c@EmilienM\u003e we should burn it, weshay|ruck ^\n14:44 \u003c hrw\u003e EmilienM: also fine to drop from train?\n14:44 \u003c@EmilienM\u003e burn it everywhere you can\n14:44 \u003c hrw\u003e THANKS\n14:44 \u003c@EmilienM\u003e it has never been useful\n14:44 \u003c@EmilienM\u003e and people who started it left\n14:48 \u003c@weshay|ruck\u003e EmilienM, yup\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc560d132931401edbda8e81f684512e10e14163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc560d132931401edbda8e81f684512e10e14163"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
