)]}'
{"id":"openstack%2Fkolla~723888","triplet_id":"openstack%2Fkolla~master~I9cab91b82941ea125b16d19e68722ce41a1f11f2","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I9cab91b82941ea125b16d19e68722ce41a1f11f2","subject":"drop tripleo-common-tempest-plugin","status":"MERGED","created":"2020-04-28 12:53:52.000000000","updated":"2020-04-28 20:44:05.000000000","submitted":"2020-04-28 20:40:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723888-1588106418030-81340386","meta_rev_id":"925c5ec27c34cdcc4b8cd5485e8fda8374b8b6de","_number":723888,"virtual_id_number":723888,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-28 19:01:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-28 20:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-28 20:08:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-28 13:39:49.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-28 13:07:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-04-28 14:56:33.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 13:07:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-28 14:56:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-28 20:08:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-28 20:40:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"958faa6d41976a7bc2f922f568157ed8572cd3da","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-28 12:53:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66b4d70c1fc139588eec38ba815435d9db207be0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-28 13:07:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9b25843a028ae7368a48a912c798bf8874bf7ec8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 13:17:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e3b4201974945768370b95adf447694 : RETRY_LIMIT in 2m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6def04d9890d45d48f0b70f774bf1dd9 : RETRY_LIMIT in 2m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/85cccf74fd524146886693cc5738002a : FAILURE in 3m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e32e1dc982644808af592487572114b3 : RETRY_LIMIT in 2m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"469311f9a986d1aa36233dabb74a83c07e9671fc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-28 13:33:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/212d5c53b6bb4f43831639c26d0a18b3 : SUCCESS in 32m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8993ed4a8b7e274945c3ebb88228e1e989233c37","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-28 13:39:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4512ffe090ba6fee7cdf1b472d2527324cdb9530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 14:18:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e95597f3ae314d9aa4565fe192348fd8 : SUCCESS in 31m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4555574fc6f84f03b64c8e71af50133c : SUCCESS in 41m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/425805cadf084483b0b1de45eb617bcd : SUCCESS in 1h 23m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/03b65cd3c02e400fa021419203fbc9c5 : SUCCESS in 1h 18m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/73d46365f1f643d7bf9323b4f781937d : SUCCESS in 1h 23m 47s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2ef6307645b84bce882c1d2e789d2613 : SUCCESS in 49m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4c1460aae3f24e0abe239757bb87790b : SUCCESS in 32m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a2ad5ceff878434bbf1db773366e3768 : SUCCESS in 28m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eaed4191471749c8b0f0bc2b0dff4963 : SUCCESS in 1h 07m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/704ad38d6ff54d3e8ddb8f579a79f348 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b5e3a459fd42ff847610b89559d31f : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0581e2fd62384efa920a811662050b78 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/617353606fa54872a77618dcdcc05eda : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b09802573a043fa97b311daa046ba97 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d737d8102984e6496b690e363cef72d : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef0f45fdf65546e2b86b1c1ab0904c31 : SUCCESS in 23m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca4f66af7cc24fd2a0d0164126564db5 : SUCCESS in 23m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8dbfceeccbc484db2288239fa258dcd : SUCCESS in 1h 06m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/26cbbf206dd5422aa17a52cdaeee0eec : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/73599baae84d4c15a2115439ab7e5aef : SUCCESS in 1h 13m 29s","accounts_in_message":[],"_revision_number":1},{"id":"efb162f9be9edc43b9f03a6fa02dbd0b68882b17","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-28 14:56:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a06750c137dd87504eab44a4323ec52bf9e85175","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 14:59:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88941c0641e3029481f066066054078158129920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:07:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7f481a04517e4ccea50d1277c3757917 : SUCCESS in 50m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/90dee39f02ec4a29bdb724919bd5c21d : SUCCESS in 1h 25m 21s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2a3e7222d0dc4543b4c212fe2b5b21fa : SUCCESS in 2h 00m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c7feddc41594b38a1c56f31c1298c03 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3bbc4be7e9e47a69f41e7844af7fcd8 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc1a26773be446299873109aee89146f : SUCCESS in 8m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7719f168d62d4e19b82fb2886aa09dcd : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2adbab36e784edfaa06a97c12426440 : SUCCESS in 8m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c226042a21fc420a84660441a4cd6fc2 : FAILURE in 33m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b05292ff65d347ad806d2a143d2a51dd : SUCCESS in 1h 14m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/66dde5e142c440768530ed66c5bf9ecb : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fd6b1d38f06c420c62d24a2f6b4c7c023f2f37d3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-28 17:20:35.000000000","message":"Patch Set 1:\n\nrecheck\n\nubuntu grafana mirror issue","accounts_in_message":[],"_revision_number":1},{"id":"995b5980046f02ba508a99fa8b1a31ddeb729174","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:59:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/db69a525afa048879abd28fe9b448f53 : RETRY_LIMIT in 3m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4f70669c7fe4587878331d578f3bf9b : RETRY_LIMIT in 1m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/64409e257cd241a8a821023b9e2a4ce3 : RETRY_LIMIT in 1m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0ba3c470ba94d028bb97055581cfe01 : RETRY_LIMIT in 3m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d7cecdda5569d7667a7cc26a6e6d1e19a6c400a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-28 19:01:22.000000000","message":"Patch Set 1:\n\nrecheck\n\nmirror not in sync","accounts_in_message":[],"_revision_number":1},{"id":"693ae08cf1acff1fe4d5ea5efa03f662aeb0c10e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 19:07:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/acf9ada6fbcb4a20af9b0634d6d23045 : SUCCESS in 24m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/042e0b40d269412ea87d0585621d6d20 : SUCCESS in 40m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b2df0b99ca8242349f5805b5d44abb07 : SUCCESS in 1h 29m 55s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b3232149b0d349fe8b0894242159054a : SUCCESS in 1h 17m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9596b9ad8a0c479c86c729e5b7f4919e : SUCCESS in 1h 34m 58s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ef524c645fa04c0c8b3fb18aac2bc43e : SUCCESS in 47m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4bc454df2b7b459fba2f4468e4f0666c : SUCCESS in 1h 08m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8a94fe11d4e746c3b7122fb7f97d5a99 : SUCCESS in 22m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aa3eecef011245438c4c5836c2ba3468 : SUCCESS in 1h 08m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d009ac404e9a4274917028bf66ee8a0a : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7edc38bf06e14d25a3ed645eaf055e92 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcc136b4de7741b880b133c0c3325520 : SUCCESS in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/868d6fcb616f41f08b10a08c5a156b63 : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52c58d640ee24198a77dff1e8b7661ca : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2936e6a25b594c04840479702a9a8579 : SUCCESS in 7m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87daa51347dd482e8adda3443ea3d50f : SUCCESS in 38m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/07540a45cfea4fd492cddf41a8d6168f : SUCCESS in 48m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4c398ffe7094190b9cec156a7dd2f10 : SUCCESS in 1h 16m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6f1d0357f0b144ecaee142ed04e68218 : SUCCESS in 1h 01m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/13252dbdf30f4cac911aa675f130aab5 : SUCCESS in 1h 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"86c2f460f9ec284b5ab6a240adcc60e599f40a2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 19:07:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49404ad89f1e4053695ca570935fca17c672298d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 19:35:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c02e6f7632a4138ae7e0fc2f0ac144f : RETRY_LIMIT in 2m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f2d66e48b5144f08809dff72f398e2aa : RETRY_LIMIT in 2m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b94ca571fef4a20a098a726a1372dd8 : RETRY_LIMIT in 4m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/57a586716cc24f8d83f12b63851934ef : RETRY_LIMIT in 2m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3b9812c1e7ea4a70d911507535f6ef5c4dba8ac","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-28 20:08:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/61bab2da491746d99d7ab9b8ddca4009 : SUCCESS in 28m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02e5366d9f59a9a224b503e17f1438c4fa4de6c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 20:40:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8fc9177032654ec7b5fe0e36be08c927 : SUCCESS in 27m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0c061d1e13824796bc35378a3e51d897 : SUCCESS in 1h 29m 57s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/574292aa3f3848d483fd7ac13ba207ce : SUCCESS in 1h 18m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f87e5e6b88d04a23a1f44ee81d9b9d5d : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeb6396e53f0420eb6f335d2c6527265 : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55a16b0590e8492faceb4bfc449d6b6f : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51f112c918bf47cda748dae0dac43839 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/881f6d668b2140cd874b2087e2d288eb : SUCCESS in 6m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/464eb69f12f54b499465863eb5ab61d4 : SUCCESS in 37m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/02c747c1d2c64de384da35fd4d6b10c4 : SUCCESS in 1h 09m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3d735604c7ed4680a0fc09f16e93fe73 : SUCCESS in 1h 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ef630472a94461da99256bf435bfd27daf31fc9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 20:40:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"925c5ec27c34cdcc4b8cd5485e8fda8374b8b6de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 20:44:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdbc1405c18248cebe019bd4b606719a : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65d7212f8400b54aba324327e869a344128a30cd","revisions":{"65d7212f8400b54aba324327e869a344128a30cd":{"kind":"REWORK","_number":1,"created":"2020-04-28 12:53:52.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/88/723888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/723888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/723888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/723888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/723888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/723888/1"}}},"commit":{"parents":[{"commit":"54a644409115a1db22dfb08d8f087f550ddef55e","subject":"Merge \"aarch64/centos8: no fluentd, rabbitmq, kolla-toolbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54a644409115a1db22dfb08d8f087f550ddef55e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-28 12:53:25.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-28 12:53:25.000000000","tz":120},"subject":"drop tripleo-common-tempest-plugin","message":"drop tripleo-common-tempest-plugin\n\nThis is not needed plugin without any branches or releases. And wants Python 3\nwhile we have Py2 for CentOS 7.\n\nChange-Id: I9cab91b82941ea125b16d19e68722ce41a1f11f2\n14:41 \u003c hrw\u003e is  tripleo-common-tempest-plugin your stuff?\n14:43 \u003c@EmilienM\u003e yes and it does nothing. I thought we removed that?\n14:43 \u003c@EmilienM\u003e apparently it\u0027s still here https://github.com/openstack/tripleo-common-tempest-plugin\n14:43 \u003c@EmilienM\u003e we should burn it, weshay|ruck ^\n14:44 \u003c hrw\u003e EmilienM: also fine to drop from train?\n14:44 \u003c@EmilienM\u003e burn it everywhere you can\n14:44 \u003c hrw\u003e THANKS\n14:44 \u003c@EmilienM\u003e it has never been useful\n14:44 \u003c@EmilienM\u003e and people who started it left\n14:48 \u003c@weshay|ruck\u003e EmilienM, yup\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65d7212f8400b54aba324327e869a344128a30cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65d7212f8400b54aba324327e869a344128a30cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
