)]}'
{"id":"openstack%2Fkolla~723891","triplet_id":"openstack%2Fkolla~stable%2Fstein~I9cab91b82941ea125b16d19e68722ce41a1f11f2","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I9cab91b82941ea125b16d19e68722ce41a1f11f2","subject":"drop tripleo-common-tempest-plugin","status":"MERGED","created":"2020-04-28 13:00:15.000000000","updated":"2020-04-30 15:43:51.000000000","submitted":"2020-04-30 15:41:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723891-1588261314173-56a7dbf5","meta_rev_id":"11ea5772fbecc6465f3d39b06bd61e77b9c1aeda","_number":723891,"virtual_id_number":723891,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-04-29 11:27:29.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 15:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-04-28 13:08:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-04-28 13:08:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"date":"2020-04-28 19:00:51.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-04-28 13:01:31.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 13:08:27.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-04-28 19:00:51.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-29 11:27:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-30 15:41:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"673550ffe30ed458110db8f8b48e3de4830003b5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-28 13:00:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9424f66170a397f86a9645e09aa34c1e9c2f2c5f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-28 13:01:31.000000000","message":"Patch Set 1: Review-Priority+2\n\nbreaks all Py2 builds","accounts_in_message":[],"_revision_number":1},{"id":"3dbf729f257033510e6f8a2698b48537f08bdcfa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-28 13:08:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nnot crying at all","accounts_in_message":[],"_revision_number":1},{"id":"2a4482b27b7b003adde1ddb9c0618e9a3e59d7a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 14:52:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d644e4b8300c43a2a1164af0513db37a : SUCCESS in 1h 25m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4e3437b9e5ea4137855220b6783223ab : SUCCESS in 1h 44m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/def3e01c98e54cb89eafdeb20aad2470 : SUCCESS in 1h 47m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0903874f39994765afa4c71eff77ee26 : SUCCESS in 1h 50m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/516379651ceb44de9891cda3e1b277b0 : SUCCESS in 1h 41m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8510394b9f954b0899573ae27b0bf297 : SUCCESS in 1h 37m 00s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bbefac51a87849f88dcbbdc02042c13e : FAILURE in 52m 31s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/eb04d462a70145899a1f029b6a14d72e : SUCCESS in 50m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d1f796e6f874741a68530c6d539354a : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93925332f8340c39a9b3e0c0165ea14 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f2fafcc96414f5095a1e523de524126 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01ec0d0130014a948f6604d47e2292ba : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58b91527b03e4e42a2416da8e1ca0daa : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb15744b91347eca604fd163ece954d : SUCCESS in 4m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dd4922872d4d4b49986b35544638e4e2 : SUCCESS in 28m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d99c48523a8f4c81a316213c44744a38 : SUCCESS in 30m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b65611fafb741488cd52bb4d3e7a4e4 : SUCCESS in 1h 02m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c44deb11e3bb4376bdfe6ab17fcf99be : FAILURE in 35m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee40b74a652b4c41a5978ed3963a0c29 : SUCCESS in 1h 07m 12s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"286c1e77981bc780410ee0556b983a92a888aed0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 14:55:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4c2eb8994c1b12836f7a5ba3e396efa15c3813f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:24:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d0bb790039904782a0943f0d8ed0ff57 : FAILURE in 2h 26m 50s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8967585eb9774d1b9f7859fc5500697f : SUCCESS in 1h 06m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc84ba1a31e247b0994b4f9abf56d371 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01841a6f84484f07b5f6c1a8b3434abe : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c724afadf17499bbdc844655b340a61 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c8d3d0e0ddf4687b22271011c20841f : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef7256f3a4c845c3bd12c3568d439bc6 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dfe3657938848d886fbf820f5729ae8 : SUCCESS in 3m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9d5145f318a048269b0d3c8485aa0c11 : FAILURE in 5m 31s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"a5df6fc1d6897c72aec70bc6d0f945a664a889aa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-28 19:00:38.000000000","message":"Patch Set 1:\n\nmonasca-thresh ;(","accounts_in_message":[],"_revision_number":1},{"id":"8145a973ece9f7e8d4fec3447e82850c1352283c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-29 11:27:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1f5a45b9a4323b1d3847bc7c2ecc6dedf680d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:10:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/08c5dcac6fa345c08b3c2879d0ae4d9f : SUCCESS in 1h 04m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a68e72349b05433da9d4adc4ff69c9d7 : SUCCESS in 1h 35m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/44dabdecb61a4520932fa7b2d6f8a657 : SUCCESS in 1h 18m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/34cf28954b0142d5b926b2b72e88da22 : SUCCESS in 1h 24m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5e418eb77f5844ae8c041c4b4e77caa4 : SUCCESS in 1h 38m 16s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3949b6a5931849f59ba3463294ed7790 : SUCCESS in 45m 39s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/84062b8f530f4ea9adf425525d6dd85d : FAILURE in 47m 49s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a57ed15ca7de402993b5ae9efccac396 : SUCCESS in 55m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e85d357652e34f40a1cb071bc13f4310 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1bb96228a2244c7b3757a4a59a229e2 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/135f511ee494490088c1e41cc99682ed : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be1c857a3cbe411a98efd78854021ffa : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31e106085c60402c8eb03740834743a4 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/043add1725c547509e865c69e22ba766 : SUCCESS in 4m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/783f7bb0b6af4eb18e44aa60c374affe : SUCCESS in 24m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0c92187c4cb4c9a8bd3a7456901af0d : SUCCESS in 33m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/772bf374db584d1da17812c67581747f : SUCCESS in 1h 01m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4f6107052eb42bd81b21b076a353fe4 : FAILURE in 36m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec1e5ed0dd88442397d66d131a446651 : SUCCESS in 58m 17s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"9abf1a69188a43eeeff23f47fa4dc82f3c798fb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:10:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6477e16181c69841e173a7587a889188154d1ff8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-30 13:16:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5ed6c06057bc650ab2056f5c819fd9c413cf174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 14:45:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/472715fe49bf4dc5b10134967c6c4cc8 : SUCCESS in 35m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/437901d80a4b4603aa85f4bf9b84516e : SUCCESS in 55m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c02db55b05e44ad69cc21ae0f5a80268 : SUCCESS in 1h 28m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d688614959ef40b69f3b9f3c4ff2270b : SUCCESS in 1h 05m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d0465e8e5a5e41bf9ba8f04ebccb4add : SUCCESS in 1h 13m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4088457bda414ec1aa662823ff052f25 : SUCCESS in 1h 04m 24s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/c4e5dae361694031b9e4d9483474036d : FAILURE in 56m 30s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/52575f61b7d54912a5a2a3b438f2553e : SUCCESS in 48m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1580f8d431d846b296ab4430a5b4eb92 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96ea8ff2bec4d5a967743d44f2f9c46 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5bef8c2de0e349f792e7ec2e7c674516 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07c94debebca43fdab4223852f6e6bd8 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4061714b122945e8883d9de53494a7ea : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9baf8d68274b467daf46e1853e94e2f5 : SUCCESS in 3m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a7a557657cbf41e0b20de89b02f6a2be : SUCCESS in 36m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dcd692dbea054cfb9560c6b2a7848829 : SUCCESS in 32m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5e484afeed544c9e92eb435bbc415a53 : SUCCESS in 52m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2399e5420e444730afdd8d9fb2e9aad7 : FAILURE in 32m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/55211efff277422091bcc7b7a455d548 : SUCCESS in 59m 50s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"8e992fd88c035b00eb73eb90757de129544d5509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 14:46:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"552c646e9060f804c6aed2c437e4bf272a462ba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:41:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c1fe4c79e48e4e74b72416f2aa55d34f : SUCCESS in 44m 17s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/dc56d44d88224f86908e67cdfa2178b7 : SUCCESS in 54m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd373c6a4df94400bce9571a63b7f204 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71875560452c4d259feb71e4344d1294 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db3e0fe1b09c43879bb9a9a440590ade : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7604ff0dbda04d79b4a3c98310644680 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cf06b3c3dd24c7999721ad183c18b40 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44c8f796769442888e18b3956fab475 : SUCCESS in 3m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b97a1160918c40709953ba61a0c7ba75 : SUCCESS in 39m 37s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"f82267142391c44b510946b7e43d2f538484804c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:41:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"11ea5772fbecc6465f3d39b06bd61e77b9c1aeda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:43:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc59f332d7624f12b51c92ee2f6866ca : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"627beae3b8ddd1de3ddc34637f3e29efcdd53681","revisions":{"627beae3b8ddd1de3ddc34637f3e29efcdd53681":{"kind":"REWORK","_number":1,"created":"2020-04-28 13:00:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/91/723891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/723891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/723891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/723891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/723891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/723891/1"}}},"commit":{"parents":[{"commit":"ebfc0a7bf8da8ea27e5a7527859c201dacf7c5e5","subject":"[rally] Fixes Ubuntu binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebfc0a7bf8da8ea27e5a7527859c201dacf7c5e5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-28 12:59:45.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-04-28 12:59:45.000000000","tz":120},"subject":"drop tripleo-common-tempest-plugin","message":"drop tripleo-common-tempest-plugin\n\nThis is not needed plugin without any branches or releases. And wants Python 3\nwhile we have Py2 for CentOS 7.\n\nChange-Id: I9cab91b82941ea125b16d19e68722ce41a1f11f2\n14:41 \u003c hrw\u003e is  tripleo-common-tempest-plugin your stuff?\n14:43 \u003c@EmilienM\u003e yes and it does nothing. I thought we removed that?\n14:43 \u003c@EmilienM\u003e apparently it\u0027s still here https://github.com/openstack/tripleo-common-tempest-plugin\n14:43 \u003c@EmilienM\u003e we should burn it, weshay|ruck ^\n14:44 \u003c hrw\u003e EmilienM: also fine to drop from train?\n14:44 \u003c@EmilienM\u003e burn it everywhere you can\n14:44 \u003c hrw\u003e THANKS\n14:44 \u003c@EmilienM\u003e it has never been useful\n14:44 \u003c@EmilienM\u003e and people who started it left\n14:48 \u003c@weshay|ruck\u003e EmilienM, yup\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/627beae3b8ddd1de3ddc34637f3e29efcdd53681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/627beae3b8ddd1de3ddc34637f3e29efcdd53681"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
